)]}'
{"id":"openstack%2Fkayobe~813265","triplet_id":"openstack%2Fkayobe~master~I16d721539e5956f0d86503dd5b01ea608b129873","project":"openstack/kayobe","branch":"master","topic":"bifrost-netmask-filter","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-01-03 12:00:41.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-12 07:50:20.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I16d721539e5956f0d86503dd5b01ea608b129873","subject":"Use net_mask filter instead of ansible\u0027s ipaddr","status":"MERGED","created":"2021-10-08 22:47:00.000000000","updated":"2022-01-03 12:02:37.000000000","submitted":"2022-01-03 12:00:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813265-bifrost-netmask-filter","meta_rev_id":"655be04050e7110cc02ca38cdb949b171de8a018","_number":813265,"virtual_id_number":813265,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 12:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-03 10:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-09 14:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":2,"date":"2021-10-12 07:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-01-03 10:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-09 00:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 13:48:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-09 14:07:31.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2022-01-03 10:03:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7f03c7d106fc2e96c5de84cd914f1af919071c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-08 22:47:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feb8b5569501e5f157200ac6b3486b8af6561782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 00:52:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5347f4cbac84acb9816be324a261c2b : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295c7134f68d412a9b6598b55f3b1695 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f63c32d3bc1e42e89498c6872c70cc8d : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9272ed9687b4c39a0f3a2de47284f0b : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d97e217a8c564043a3fd9b05b76beeca : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1da90e6de043459f77ddabd4866d2b : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef1560e667994ddea497a7a396d33163 : SUCCESS in 7m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a44c1ed7a784cfa994bcec81850200f : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3332589be3a415e862ed02208b6952e : SUCCESS in 32m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fee5b1f8b8ce4d3c81d2e53a9c631f2a : FAILURE in 2h 02m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a413491eb94d41a38fe24297ef96808a : SUCCESS in 46m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9e01c6ac10bf4ebc9fbcf61a2bb0494f : TIMED_OUT in 2h 02m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dd3550c64b5f4d1b8df52bcdeb90b525 : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af5b3b6062b44f5c8ddb3f2764d68e3b : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a656f495e84642608dd29cb4c14659e7 : FAILURE in 1h 37m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afb45d1601204df8be7edb2a83dac7c0 : FAILURE in 1h 16m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/37c8644e353d49fb97e6686a2705763a : SUCCESS in 39m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f0028e0444457fbcbeddc64a228d4d : SUCCESS in 29m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/055d413e56e74975ba30c5c2a4ba7629 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a098baba38e49a090ac805ea6214deb : SUCCESS in 24m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0be65ef144594284acd1d75833c3bb51 : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c8085d260454e36b5f94c4838418745 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8b9374b43d1980f1b013cff0862530b8916df778","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 00:53:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33a59a3d4d16f0b1aec2abd633ec21769f840ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 02:57:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1daa2c07f924664adfbfa6faf753fb2 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f9ebe9cadc4d6ab28972deb96ffd9d : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2899cbc89aa04692bc6cdb24cd0b49fd : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df1406d12c54483a53e91b7078be801 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4647886689e4caba0ea908a6948a446 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904ce365ae324393afca5cb09e733a5a : SUCCESS in 10m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dabbd887b8740bfa10edd5d02325b16 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/660b18e65d4a4511b59e1be071a03bf2 : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eca2e576a4b04e6e95efe16f5d36f665 : SUCCESS in 28m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/15fc74a37c314f0db971ac9fc8d8551e : FAILURE in 2h 01m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0929fda6af1d4712a4ebb97b9ff3ef45 : SUCCESS in 43m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/017fadda7b1e40e8af8468e78312d6b9 : FAILURE in 1h 59m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e4a43418d0cc47e98df5775bbb231a98 : SUCCESS in 22m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee71dfffbc5c4cbeb37347e772cb67ae : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/83a65f60709645fb9eea0f99dd2f3bc4 : FAILURE in 1h 44m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cda1c1df2b35401a9b03df1dd069ae7e : FAILURE in 1h 22m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cbf707c584d34ecb9514b347750798a8 : SUCCESS in 39m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a40f782780ec48a2a057ca06ea468a89 : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aea010ba388a40399ff0740e2779f2fb : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6625d277d5424fa13aa3a6338997ea : SUCCESS in 32m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b285e271c8974fc7811c494f684358f5 : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0611a506cd1487883fbbaa162e0bb2a : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"aef57275bee69081a9d74e3d12f97149f2a8bc28","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 07:25:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c36e38f59fff1341b88d850dd7af23b8eac0ef72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 09:04:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f46766844c74417811b7024eb707893 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e49c8e53a5c431eaf0ddb8a19d9f41c : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b620acb05b47aa86deee0b02bd6714 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbfc31d798d54df28c96666c4c22fb45 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ab4af864a6747cdb416dbaee9814b4c : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e4649550a645aea8ea8e75abab6d90 : SUCCESS in 9m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a0ad59fad704a0baca9da08d8286982 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea42ab81a66743a7800216c5dec1951b : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/53af04711d5146fc9984ea8a27f521f1 : SUCCESS in 33m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c2fcc28377934aaa8791d06907532e1a : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dc8b8fe769b42c2b1380b5356d584c0 : SUCCESS in 47m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d4e6a79102ac4c78b485714c8c26a181 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/63ff8f6888c04e5da60ed072558173e1 : SUCCESS in 19m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f695ea86ccb4040b9f57a4585223f0b : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b6bb3329c3b468ab837e473f66defd7 : SUCCESS in 1h 37m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95c50a82ef4c4784b4ec3e53c01ae8bd : SUCCESS in 1h 23m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/236153379546442aa436df43d4137359 : SUCCESS in 37m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62dc51a8b1584f5ab986d76d9028226a : SUCCESS in 30m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fe16d2a15994bda92668aa32a8df9f4 : SUCCESS in 32m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45a8f450c4034e2c8d29a7861fede53c : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/732a3e8af1bb49b98b0146157ef292bb : SUCCESS in 23m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbc5c0530e2f4e1183374502cca31207 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"922f88ed79d2560ecd607d323eb97d57c8c5bbe8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-12 07:50:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72706aa9010cf5d09f8692bbf0aae0b03dbe67bd","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-12-09 14:07:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c3b145afd2db4489fc4ac2df37a4f8ffcc6679fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-03 10:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"abb735bddc3e493c5d5af0350bb475febd45e5c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 10:03:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e2d8fb153c440a59b21a16c42bccff851097502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 12:00:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f7b5bc72b644bab4e1dc8f333aaa93 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06911e31d9e84d87a90647ef19e2bf71 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4b6f67a80dc4669814c5e7288f6bec1 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9315ee886d54870856c84b13d3a0ea7 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad200918de814ee9ad03033af8779979 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7615c412482a401ba7762d6f4d27ce01 : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/801b25f18e58410385e91c4018d4f8ed : SUCCESS in 30m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c11da8829c2475a8223572b8dd3cd34 : SUCCESS in 1h 12m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2114756c09884e7f9d8aa83d4cdeac27 : SUCCESS in 51m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3fb6940ef5ff4333a1d368c935ee1b40 : SUCCESS in 1h 11m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1d98bc8ca9414475b698e10356c08c72 : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e493645eeaf40d0a1914834a9f1b593 : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e2742f766db43d58fb6b370031df759 : SUCCESS in 1h 55m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a22f9ab95ae456684b7afc289ba6a94 : SUCCESS in 1h 17m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d0785feba654b07ac0087cf0560f6af : SUCCESS in 49m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ba074397ff149548f596623ccfd8a00 : SUCCESS in 38m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c387d722eb3417f89c5c7d3cefe12c0 : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ca730d33d7547b6b75344af128c6ee8 : SUCCESS in 23m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/afec60ed7c174cb991f6d8a59c714ff2 : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86828487d93a4d929912c332a5b080bc : SUCCESS in 10m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba7995bb0df34fa997833a68ca6c34a0 : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c984d191a7b94563afbc9250104b2825 : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2d3f52bf4cf393fd71663a46cfa9e9e7f67b6f03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 12:00:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"655be04050e7110cc02ca38cdb949b171de8a018","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 12:02:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99f2c4df7ef48dc9e5818b77d7ff5f1 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0751a316f4ef093221ae3bab2d6514689cd980de","revisions":{"0751a316f4ef093221ae3bab2d6514689cd980de":{"kind":"REWORK","_number":1,"created":"2021-10-08 22:47:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/65/813265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/813265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/813265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/813265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/813265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/813265/1"}}},"commit":{"parents":[{"commit":"580485790a5a8568f054cd47bbcb87bbbd063be8","subject":"Merge \"Import merge_configs and merge_yaml from Kolla Ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580485790a5a8568f054cd47bbcb87bbbd063be8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 22:39:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 22:45:41.000000000","tz":180},"subject":"Use net_mask filter instead of ansible\u0027s ipaddr","message":"Use net_mask filter instead of ansible\u0027s ipaddr\n\nFor a long time [1] we have net_mask filter, so no need to use ugly\ncombination of the net_cidr filter together with ansible\u0027s ipaddr\nfilter.\n\n1. https://opendev.org/openstack/kayobe/commit/dc45036c92cb28fd4e9f2d9c84a03438bf7ad26f\n\nTrivialFix\n\nChange-Id: I16d721539e5956f0d86503dd5b01ea608b129873\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0751a316f4ef093221ae3bab2d6514689cd980de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0751a316f4ef093221ae3bab2d6514689cd980de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
