)]}'
{"id":"openstack%2Fkayobe~813266","triplet_id":"openstack%2Fkayobe~master~I9488a72db588e31289907668f1997596a8ccdec6","project":"openstack/kayobe","branch":"master","topic":"inspector-dhcp-range-netmask","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-11-25 19:33:17.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-18 08:18:51.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-13 07:42:26.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I9488a72db588e31289907668f1997596a8ccdec6","subject":"Add support for Ironic inspection through DHCP-relay","status":"MERGED","created":"2021-10-09 00:08:24.000000000","updated":"2021-11-25 23:27:42.000000000","submitted":"2021-11-25 23:25:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813266-inspector-dhcp-range-netmask","meta_rev_id":"263afda582161264963cdb30bd895ac5d74eb6c9","_number":813266,"virtual_id_number":813266,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 23:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-25 17:12:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-10-18 08:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-11-25 17:12:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-09 01:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 08:04:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-12 19:52:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-15 09:51:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2021-11-25 17:12:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9127a6df386f19c2f887cda4e94db47c1ca5af39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 00:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fe2655a18bd2cb8682101dcf894ffa014850b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 01:09:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe649eb1921c4a349b6a3a36f99a71ba : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a60da29e20640c5ad6c347ad5201686 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61d29cd42a0e43b0bdc9ad3e6745e22f : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa3a3fbd5787401598a1cfa7cc94730f : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e41e6c5e8b541488daf9fb70583028b : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51989efeeec24ab4b6a7cd477752b087 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9997244cfb7f454d964c89cc08586b35 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7eec556a18bd4ebaa373c91aecb5fd19 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/51a16b3343c04bec88bdaa57e5003b61 : FAILURE in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a27fdab4c6944f5e80fc2981ef1856a1 : SUCCESS in 31m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e7c0f89edd9f46ffbe7a88be5bae98eb : FAILURE in 18m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b724fbecb1db49169e8f612ed951a274 : POST_FAILURE in 10m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f0db2096cb274f049dc786e2f94760ff : FAILURE in 11m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1d046dd3ac804f78baba572f3292d6e6 : FAILURE in 17m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdb0dd5612b346179849e6ae8e3f019d : FAILURE in 12m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a75a3f88b5df428caa828d241acfd8ad : FAILURE in 59m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05dc05541e3348e1b00d29aecee250aa : FAILURE in 52m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0dd24164a5df4d6fae1ba9831940715d : FAILURE in 18m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cf322bfa3fc41ddbeff8b58d63b18f6 : FAILURE in 10m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5ef955c2228c4c9c9763ed945154686f : FAILURE in 24m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d45c671998c34eecbf0131dfd1726238 : FAILURE in 21m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb08f8c716f74c008604a1ed9afb1708 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/effff84c88e743398aaae9d147115abf : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5078e1367e7f409df354d79222a29f4d1b0811c8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 07:26:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9638ee84b5ece8ea4bed154cd8f8a4aa68dff41","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 07:51:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ddcb3de4aa33933e87d6f2d38e2973fcc3690c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-09 07:59:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d635bbe7a40fc2505fa46205b2177da5eb78fc73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-09 09:46:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/923c03be0b3d42bbae2a2e58130388d2 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643d72490c184f048ae86e359c333b13 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6939c2b0ebc94459ae13532ae723cdcd : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4698a8445e3b4e2fbbf696ba4c46e951 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/431f7bd291f74a2e9ad8401ffba82eb0 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b712d9ee5894105a4c0c9ea9bbc3144 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53adbb993d0a4d24bf93c99a60cd9f92 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/089b227f53964d0aa4e8b908a7f2d25a : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c59fce858d74360ae771a57e86c0495 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/651f19af8d7743d08677880d0a788a52 : SUCCESS in 27m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/33c4ada422cd46139a98abcf205cfb7d : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/086a3047857c43cf80b30e1ecb5a2a8f : SUCCESS in 46m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fa876109a2be417a92a9d19ec1c25369 : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d645a575f9c04843bd5be0e13bbdad67 : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5679410a1b245978a8eb1d0ade28640 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/528fcbde594d4158924d56df8c9a1dc3 : SUCCESS in 1h 44m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97a682b87b6c48b39c2d15a06934a710 : SUCCESS in 1h 07m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ede6708fb8444b2fae596adefcc144ab : SUCCESS in 31m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aacb4df91a1b415297199f4b22b46f3e : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c84747ac9b6448e892ef2735d88ac82a : SUCCESS in 25m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a94a2e385f45eeb5bbb347d6e79f51 : SUCCESS in 24m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b4ffe08e1bb54f8a9a70ddcb56dc0bca : SUCCESS in 12m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d14c9992c2bb4eb0a1d5193b21b8eb84 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":2},{"id":"cdf2005aa265ebeea5cf51bf554dbee5f65ab49f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-12 08:04:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe539df7415c5f5d7c74dc6e2e2726530132ab2b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-12 08:05:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94261c11f5ab10be688126cb97433733f2b14fb6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 08:37:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89b2cc1154b551eab7b3355a867740f22d97cd2d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 08:38:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8fb52bc52593143063b7206bc83a03ff102f621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 09:03:29.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"79e5e96c3bc41402a2abe5801051f1df5c166a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:05:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a97a0bd3be24a0a9582f4130738687f : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006b4e2e84ae40dbbbf2da268da1036b : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8bca8354ba843f3bbf17136c0d48504 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcdd40f0269440e9a247d9c6264a540f : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6df90523571540f492951a7d1e1e8c41 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5011c7b95fbe422cb17e7dd48c6b8e82 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f6de4d9a0845d8ae703216eccb6a2a : SUCCESS in 5m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f61333949ea4b85a8d7d0bff8c0a568 : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8f3ef6c1be34b3ab3a9f13795c681d0 : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47eff7d5dba140ddb9a38d761857e030 : SUCCESS in 32m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c33d4786757446a7956648ee758c606f : SUCCESS in 1h 24m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f482d8ffe184ad49a909e456a633388 : SUCCESS in 49m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bc17a661af5249b39469789bc62ddc0c : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ca6cb782d6e64829963ceed741e0374d : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92d39bfb6988478ebfa05e2a1d5f5727 : SUCCESS in 17m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b7acd75298e452aae5c6b7ee2420716 : SUCCESS in 1h 58m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1850c19462de4118a31cf37cf3519259 : SUCCESS in 1h 23m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7a8dedefccda46d89581595e1d31b94c : SUCCESS in 37m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a220b978aef947d4901e8bbabd3c3bf2 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2775a50b27b94817babdf65a75ead750 : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0da598cc84f844f4a9a1843fa5b93e2b : SUCCESS in 26m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a2988caa7774fae959891c51018c5b9 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbd74d170b8b4dc3afa1835bf84e662e : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":3},{"id":"26facbf678a47bd3a65c6d457c50635b4b930baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 19:51:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b021e3be17c49a7418f077a506248eeab0ec04fe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-12 19:54:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f9fc02ac8f866ada5e6c2e582309e2f0558546b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 21:55:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/937e115157a74e58b442301f1fcce94b : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5cf49292fb94049b989c48a28c75441 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ad9dad7acd46a9aaa8421edc13e590 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25fb582d2f3f40dea82014631bdbe01c : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e60c90ab7b4f44a581b0513bbacd53bb : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4af4d73b0b4a4b95f2ce51d20b8dd2 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c93135264b0d40fdaad1c22688af7e94 : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26dc035a2634419e89f65226894edc3e : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c944d6e99ca74522a33051eca750298b : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/097f3dbbf3a445d6aec7ffcd667c7919 : SUCCESS in 26m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8432a9e5de6d4cc4875235c1c24c4ba5 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e7aa942722440d6870cfe5830df48db : SUCCESS in 52m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/521f9f34d7664e63b2809065f5cd7b96 : SUCCESS in 1h 17m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2da73092217e4e46ac91c12ada0b9c20 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d56f8c122b84862b7630943cdd89d0a : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dfb7784abf43465c9b594fdfa966799d : SUCCESS in 2h 01m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d098b393c0b4de2af52cce5ddd03737 : SUCCESS in 1h 27m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9b2bf697763f45e5b4065f927e09c8e5 : SUCCESS in 39m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7378eba415764e248b0e5dcad22bfd29 : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbf7df1377e745f1838fdfda452dbcfb : SUCCESS in 34m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6a0acc36f6b4d71aec530d897d2390c : SUCCESS in 25m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a65a9ceeed9149f7b2c6f99ce0dd3498 : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed503269be5841d7b3db407dd411ac86 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e72ffeabb66b0e2ff215a34f43c311329e2e81ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-13 07:42:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"025fc399c606ccfb4bb35c068db8de1257b02d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-13 08:53:05.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"faa6ac2ce4e60c9e5c265800979261ed025da83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:50:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9210251733b419e998df91440113ab4 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329919a929af4f86b28baa277f9cdd9d : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/518c6bf482644082b3896808a18eee81 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c512b58f6759435c9a2ca41f9ed0ce14 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a73bf9eab40b41cab89a4595f7df4df0 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202d6ed4aaf44ea3a4da053975a5d9de : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2634194bbc1548e4b7dd4f1228867407 : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b0268007ce54f508b39979893019ecf : POST_FAILURE in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a156b2c80eaf4f3f82a2845e1572bd04 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f31100823cd47ef9b8608a7819c937f : SUCCESS in 31m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f4f677c9efd64cc9b0eab467ad4db622 : SUCCESS in 1h 02m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1919411ab58548758ae8df355cd71bce : SUCCESS in 48m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/20cafc00607445708c9b5a8b802effe7 : SUCCESS in 1h 12m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3fa4f9f22ddf418e8570312a0f343807 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc2559bb71c4373ae8d930fa636897c : POST_FAILURE in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/83610240e3b7421aa74766adbe312c68 : SUCCESS in 1h 56m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b21180c561947c8beec457c93ddbe05 : SUCCESS in 1h 11m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4065b49877248e6a22814d85fa680cd : SUCCESS in 31m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55a891494b794e3ca26a9d8df123c247 : SUCCESS in 32m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/023a3e13eaf64e2cb336402172285cb7 : SUCCESS in 30m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f67aad79b3ba4f8fb57da53007d38eaa : SUCCESS in 20m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/49a6e6f16c5f46a481f600ed5c321e7e : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c40a3987166475d8f6e6e45772ed9cb : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":5},{"id":"099765806fc88013092a7a52362b10c955b95e3f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-13 12:39:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c81f53738373e88e6414d496c95d81962ae1665c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:43:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/486c0f1f37814269b5177218ba56677e : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f8fa480ecf4c03a4a7a7282979f7ce : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbf356546bd24aa69c8ded27b2495faf : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c089d5565844f6b0c0a9ccc87f165e : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a68e51909c54462599af7d2431f2101a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18b8c1be9bb41dbb27031825eb30f76 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5e721f3c081423c8eb24910e2681cb2 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a204669f7b2c4b24b216fff1d5535056 : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9e63330256a468799c1eb4364c149d1 : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19c736a6341044aebe3f0e033fc0caa9 : SUCCESS in 31m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ee3fdcd993704a64aac4bed82f6241b9 : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8ee278a12446a9acf5493661638fcb : SUCCESS in 46m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/808482979a7746d3bb8c223d4dd2c80b : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/074d9f81171646dcbea57a65624fb36f : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7020ee68e0a34ad4a0bf39d871da0b78 : SUCCESS in 17m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/92ce282097514a7f8036b41c6a7748b7 : SUCCESS in 1h 57m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9770d5787ee14b11b1fead9ae506c77b : SUCCESS in 1h 17m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/058f7335ddb1416483ff566cd3a5e43b : SUCCESS in 36m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7ea39fae37a4960b92b070e73ff00f6 : SUCCESS in 33m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/022c63c486464928b43491e018fd6d39 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76b9dd3fc68a44ec9923abcfdc874d25 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81dbf80a2fe24eac93587b1c0eff1403 : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd6094da14004c629192d60ca44b01d0 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":5},{"id":"4bcf8b4f973791d250b82f135b22cf5d74e3a0ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 09:26:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"701021aed5fc4c4d7616888283f84f3a242731bb","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-15 09:51:15.000000000","message":"Removed reviewer Radosław Piliszek.","accounts_in_message":[],"_revision_number":6},{"id":"cfe443024cf4200ca9b24de4886ffffac9b573cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 11:26:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff7cdd0a59e5479cb050c54a4bc4d0f6 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36116d94aa649a1a0ed6f4c4e4da0bc : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b613eef405094b71b7bc481dbf46510c : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a159c35c064385aeae012ef90590be : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b66102f1b7664328af539a7ead6c7d5e : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411d5524230e4b89bbbfdf61cddfe837 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab4e26d836c475ca27edca33b499091 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ab231622b814e7eb062ff5eb3b97e58 : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f210120870ff44aeac9b639390a584a4 : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d96a9815d1143f88563be9d0f070bcf : SUCCESS in 30m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f8a8a228b2f34d369151cf7a1c7995b0 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0bf3cac4f524341adc33f91ea143ee0 : SUCCESS in 43m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/42f9d3474a574dca8f32cad539b7756d : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d660a72664094a44ba2d6c0e3cf403df : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2d8d0c0a4764eebab41e33888abc39c : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d6e09ef43de4bbb883ccb6f3ae443ea : SUCCESS in 1h 58m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd531f8385e40ecb0c586c3cb271f26 : SUCCESS in 1h 31m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fe9771053d0f4e4c811eecc4d3af8a48 : SUCCESS in 38m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a953d7a19e34f9892256e9dfa10bd7d : SUCCESS in 33m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/69ece0821ffd47f2a5641cce9796bb2f : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407a31dc7fb244e3893af091145d13b7 : SUCCESS in 23m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c210efc8b9e04b919e6d0d356d4e7f49 : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43ef6effe0d04b82b7ab1f83ca6381e5 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":6},{"id":"4060d8ba3f7ffbd7e821d6c23327a4ecb32100cc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-18 08:18:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"accff8aa4a1455d92615d54671b5f83461b48533","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:12:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2740b444e70b569295af40321b053e7db015fcca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 17:14:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b5f16e4c3f42cf5ceff963976b1598c64b92c5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:23:02.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af6cd6691a7f4489a3804bc52bbabe91 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4b25c1ed054e81b3bea3a49e00f469 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67ec537ee46749c997301201e8eef20e : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201a96f00edc4b2e899ed25955c2b753 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bf32b96e77345ffbe4b1d16b12490ce : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a8e36e5bbf74acaa603754052c8ab4c : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a176cf02537744e983324cbaeefe07d3 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d82c02344eff4ee49b12914b2d4ff5c5 : SUCCESS in 24m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef1ab1fc460b460882bb6126beb7b33f : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70eb2f63018e4d4aad22af2fc19fb62e : FAILURE in 41m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/614ea571295841ceb0660a2036d1fcdb : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4f78b078f3c14996a3df9139b2502edf : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a57226f7f8a4e6f9185300cdc281aaf : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bd441af4ea541599e8de7867101fcd0 : SUCCESS in 1h 50m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/155d0c01b01f41f697d3f7c670e20a63 : SUCCESS in 1h 18m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cfe0974ebd15424ebd32aafe3d10ce81 : SUCCESS in 32m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6759626643049f5a2664652d6fffa2f : SUCCESS in 39m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c5eb803e2204f049db2c728ced79d29 : SUCCESS in 30m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fed189b07d644d30a5b72bdceb278fcb : SUCCESS in 24m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b9da858d0a3841feaf006046a8272eef : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd16109d1e3474d839df26eeb553e0f : SUCCESS in 13m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c789d068149549138bdb06c04cc44357 : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/292c27e5f0a4496aa7da27cd103619a1 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":6},{"id":"73a9bdb64f703a2f4641007461edf90bda9efa2f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-25 19:33:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b09c51fea308dfe0241a7772d933b7ca38a9e9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 21:23:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/857a6d28fdd340f9ae0a30d876b22fc4 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336dec53c0f7472b9032067dd5d15913 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6e79eae1b4344bb9ec1ed8be5a1ccc9 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e0c9558d18483a8d0cac0b0d347b64 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e307ff55c3438387e9d1275bfb9ded : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09837e2ec134b0587fe094d4dad8516 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37ddd564b46e48d8909d53830afbe771 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e946bf98026742aa90afb9af29dead0b : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3e0e202ac9046d1bb00ca112e7f8bee : SUCCESS in 32m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a7c6c042a216451b909063071edbbd8e : SUCCESS in 55m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f9498837fa74990bfd437e6d210db71 : SUCCESS in 47m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7426cfb81ac84fbdb759ebd1c567edb5 : SUCCESS in 1h 17m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7aea31f9907a4e9eb74cf87009cd34ec : SUCCESS in 22m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99917f60cd0645588bed71399800db46 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/486c18440353456fba85f94410b4bcd1 : SUCCESS in 1h 48m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1b28fe1e2324e28bb73a765b3cdf59f : SUCCESS in 56m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0e959045a4324478a7ab872f09e1305a : SUCCESS in 36m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aead11ed90cf48438bebaeeae8214d53 : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/21b72b5a44ab43508aa3208811e35be5 : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba1d69fe7c54acabeeda83503a0dd55 : SUCCESS in 18m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb1ab61d28b4493783e4aa8fb55a80f4 : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a5d706d6a9468cb598318a55b58aaf : SUCCESS in 10m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f164222f8d514bd6b33a13b319d53b18 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6398f8e75854495e88452ceb527d5bbb : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":6},{"id":"bfbf817ffa8cfac3140c0132929411ee6cf49ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 21:24:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"20b7b22143ff917115fc1f3a057c7be21d6794b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 23:25:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ce563b1b7644f894c5474452787b52 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/792d2ad280b8488595cab9feeeb9329f : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7502a4cc976b48d6822eb8fbc9d85893 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4e455b4883460186560d2ff78a628d : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf12b98c0d8b41bbbcc0c0545dc40581 : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a68450ebfa3473fb238d277baaa1682 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20931b2c8b424ba9a1ed3c5a1dea54a1 : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8593fbe8e37a437abc148fbcd255727b : SUCCESS in 31m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ca99e333c453477694447a10d445c7b3 : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c653c6add904181a2a5f1db7d7e8f12 : SUCCESS in 51m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/47536fe772f843c0baf3aa85e632a0a2 : SUCCESS in 1h 20m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/01146faecdc54d699774409082ee54f8 : SUCCESS in 24m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44aa7e9ca37e4f3594aff0eee33cfbe0 : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/177abeb228644d559b6320fb1ae493d1 : SUCCESS in 1h 56m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/498eb87873b5467284866f331d66b228 : SUCCESS in 1h 23m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/78861fe46dab4236851ef6476fbe5d8a : SUCCESS in 39m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f9747f6ab44a11a2b8c28cd4ed0835 : SUCCESS in 32m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/999a3253cfdd41e09ce50ffea82e2ddd : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/434bff93760b4f54bdf366efe4d917a7 : SUCCESS in 26m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bb8c4817ef8c40229931e7227d4c0f95 : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7c95318f58d4bd1972788f8f729e4a1 : SUCCESS in 11m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a7e6089973214edbb09f666bd6b51e2d : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/636cdc22aac24f1da237556f5a254337 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":6},{"id":"551c43f0891478debc6296e5726d2745d137c919","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 23:25:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"263afda582161264963cdb30bd895ac5d74eb6c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 23:27:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364acf9a9d1a4b1fbc45cd1bdfc26659 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c1c74904a14694a233dafb068227ea : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"02e01e983bc35193a5fcfe367a492c798cff0c80","revisions":{"f46f5c0c591cb0284ca61a98a3e07688f2761031":{"kind":"REWORK","_number":1,"created":"2021-10-09 00:08:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/813266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/1"}}},"commit":{"parents":[{"commit":"580485790a5a8568f054cd47bbcb87bbbd063be8","subject":"Merge \"Import merge_configs and merge_yaml from Kolla Ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580485790a5a8568f054cd47bbcb87bbbd063be8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-09 00:07:00.000000000","tz":180},"subject":"Add support Ironic inspection through DHCP-relay","message":"Add support Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f46f5c0c591cb0284ca61a98a3e07688f2761031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f46f5c0c591cb0284ca61a98a3e07688f2761031"}]},"branch":"refs/heads/master"},"b8b88b0c17160cc63d5c5796cd1f860a9bdadda0":{"kind":"REWORK","_number":2,"created":"2021-10-09 07:59:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/813266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/2"}}},"commit":{"parents":[{"commit":"580485790a5a8568f054cd47bbcb87bbbd063be8","subject":"Merge \"Import merge_configs and merge_yaml from Kolla Ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580485790a5a8568f054cd47bbcb87bbbd063be8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-09 07:58:56.000000000","tz":180},"subject":"Add support Ironic inspection through DHCP-relay","message":"Add support Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8b88b0c17160cc63d5c5796cd1f860a9bdadda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8b88b0c17160cc63d5c5796cd1f860a9bdadda0"}]},"branch":"refs/heads/master"},"94b3cfff33bd4423483d58100dd5ebb92be5fa61":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-12 09:03:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/813266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/3"}}},"commit":{"parents":[{"commit":"580485790a5a8568f054cd47bbcb87bbbd063be8","subject":"Merge \"Import merge_configs and merge_yaml from Kolla Ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580485790a5a8568f054cd47bbcb87bbbd063be8"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-12 09:03:11.000000000","tz":180},"subject":"Add support Ironic inspection through DHCP-relay","message":"Add support Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813268\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94b3cfff33bd4423483d58100dd5ebb92be5fa61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94b3cfff33bd4423483d58100dd5ebb92be5fa61"}]},"branch":"refs/heads/master"},"b46cd2f4b9d0b901f778b56a3e7804bb257a54a2":{"kind":"REWORK","_number":4,"created":"2021-10-12 19:51:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/813266/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/4"}}},"commit":{"parents":[{"commit":"edb7101d7e68efa686158b032ead4594053e2ecc","subject":"Merge \"Update Kolla images namespace to openstack.kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edb7101d7e68efa686158b032ead4594053e2ecc"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-12 19:51:19.000000000","tz":180},"subject":"Add support for Ironic inspection through DHCP-relay","message":"Add support for Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813268\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b46cd2f4b9d0b901f778b56a3e7804bb257a54a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b46cd2f4b9d0b901f778b56a3e7804bb257a54a2"}]},"branch":"refs/heads/master"},"0fab4e0bc7782472bb7937747d660e2b63ace640":{"kind":"REWORK","_number":5,"created":"2021-10-13 08:53:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/813266/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/5"}}},"commit":{"parents":[{"commit":"edb7101d7e68efa686158b032ead4594053e2ecc","subject":"Merge \"Update Kolla images namespace to openstack.kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edb7101d7e68efa686158b032ead4594053e2ecc"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-13 08:53:05.000000000","tz":0},"subject":"Add support for Ironic inspection through DHCP-relay","message":"Add support for Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813268\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fab4e0bc7782472bb7937747d660e2b63ace640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fab4e0bc7782472bb7937747d660e2b63ace640"}]},"branch":"refs/heads/master"},"02e01e983bc35193a5fcfe367a492c798cff0c80":{"kind":"REWORK","_number":6,"created":"2021-10-15 09:26:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/66/813266/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/813266/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/813266/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/813266/6"}}},"commit":{"parents":[{"commit":"e97a8fe63d2cd57579ba4288234e6edad495e37a","subject":"Merge \"Use merge_configs and merge_yaml to generate Kolla configs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e97a8fe63d2cd57579ba4288234e6edad495e37a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-15 09:25:55.000000000","tz":180},"subject":"Add support for Ironic inspection through DHCP-relay","message":"Add support for Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813268\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02e01e983bc35193a5fcfe367a492c798cff0c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02e01e983bc35193a5fcfe367a492c798cff0c80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
