)]}'
{"id":"openstack%2Fkayobe~813786","triplet_id":"openstack%2Fkayobe~master~I4ae457a784423dcce96e51b4bb21219831acc854","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-21 14:21:30.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-21 14:21:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ae457a784423dcce96e51b4bb21219831acc854","subject":"Remove stale config dump from seed hypervisor host configure","status":"MERGED","created":"2021-10-13 09:14:21.000000000","updated":"2021-10-21 14:21:30.000000000","submitted":"2021-10-21 14:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813786","meta_rev_id":"2af03fef57ce08621c3852a1bff2af97bb5aeac8","_number":813786,"virtual_id_number":813786,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 14:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-21 09:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-20 15:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-21 09:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 10:33:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 10:37:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-10-20 15:11:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-10-21 09:33:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f6c4607bbe64d14c3e1adf099cd7a6a14684ba1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-13 09:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e46caebf1e68a98f6037a6e2d422f0f82eb9a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:33:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa29627997664d00816d3f08682bc625 : FAILURE in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6bcaae7f314616b362dc4dbdf6693f : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aab4e380d1642fb90dd7da09656f8e1 : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d20e9579ac56481cbe2ad9d047229b77 : FAILURE in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/893f40741e9a49048ac4a89104af238e : POST_FAILURE in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1b3ca252004568844ea083c851bf9a : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4611175ada6431ab9f92117255c3616 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47cba001ecbd467b9f324b8e8a4c21f9 : SUCCESS in 14m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfa3e14ffc784b55814cf609734caea4 : SUCCESS in 28m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2808580697004df6b1048c1496cfccb3 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/678ffdb55c3349d7bf2e0fef42d6497f : SUCCESS in 43m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e187400ce6a241e382bfe6caafb10542 : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/249d6527bce845858c9f3d61ba740362 : SUCCESS in 21m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31e64b1bec6e49548ebb32ed8a19f5b8 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dff7dc25d35541a69552cccb5e2479d6 : FAILURE in 54m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b57bd01e4fce4a1eaf2e6b1a35f9d2c7 : SUCCESS in 1h 17m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4b9ab76730804bc89582106ee54bc340 : SUCCESS in 54m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4407d27814814bbb8ccb1a0804cce14b : SUCCESS in 27m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d9dcd51ea2d441785c279b7f0a09bc2 : SUCCESS in 31m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc6d972823c841eda3fb1bd6a2262330 : SUCCESS in 22m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2d068eec29b745c581318a34f2e39e35 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6779eb02ddc64828affb9059e35d0905 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"42c843201004f9c100bcb4f2c1260cd41b6a55bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-19 09:56:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0404a9226489c0bb434979d63439f506bfc8c649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 12:02:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9c8b0a264554e2f97bd643ff6cf98a4 : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f4cba540ac40ee871862a09d312726 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2837445b957b446a951c5d564d4611bb : FAILURE in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b2dda8d9b0e456aa014c683f96d0a2d : FAILURE in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a337a232db4cc8888f6cb048842d02 : FAILURE in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f3e51844d134ed29e01b5c88788e0b3 : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3acc2ae5f41d4dc5a14c64037c34e4d8 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c6c1fc25acb44f7b761b15e7299c9ab : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33c1179ddb8949049ccc9c9a6d1d804a : SUCCESS in 33m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4ebe061e27b043dca3d5f794549fd4ba : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23452630b7cc457fab508de89aefc3dc : SUCCESS in 50m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d4492c0c55d24477bfa002b4d98acc53 : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/16a5adc6a595483ca2644091350c1514 : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e50087f6bf4e4d3cbc1bb63c3b29efdf : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1f422982df294a309aa1d9a66b5ab4b9 : SUCCESS in 1h 59m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87c7642926204ea58060442bbbaeca78 : SUCCESS in 1h 21m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35b657e884684601aed590a5486fbc8f : SUCCESS in 37m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50dd67f504fc4a94a1007cae2b6a73f5 : SUCCESS in 30m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d977a7521134310a177691cba366c32 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36fdb31d6404d67ba9e4646a995be2d : SUCCESS in 24m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e09b607ef94447d29db99f95e1306229 : SUCCESS in 16m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a8f2e1934d14e6d8d924329a559e57a : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e7fa9ed67c04dea9c29f3fde36244fdbadd0fc02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 06:31:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c81fc1cac34fb98552b61dd711de0b1b8d992648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:45:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ceced0d868234aa8b8043730bd134726 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aefa53ed341649148b43086afc491c3a : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28954405982c4f34bd0b7701ab8606e5 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70ed2be8679143fc89dd0e1820a83d20 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b08bbd153814a9eaa7adeb0b03f054d : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a7f531a5034d9784355f5ed2e5e4ec : SUCCESS in 11m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f81f5a5fbf34e6985c54bbb74eae362 : FAILURE in 9m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2db0e894013942e38c80bb1b177d964c : FAILURE in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37eb8615bea84d87b8fbd531474ae7ea : FAILURE in 8m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b71738b4e5b74d269f7f69e3ac5b89e8 : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe2d4851501a44fd9d14c45902464cab : SUCCESS in 45m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b279e8e5994444d7ae5e4d5cf718193f : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/37c115fd5f5f4f2b93481a21eefa4af5 : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d667e8ed77ec4dbba7c44664b82df954 : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1abad16079ac472d9c78205873acaaba : FAILURE in 1h 05m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13f6be93dcd4e1cb53c0a310254e517 : FAILURE in 9m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9573510dabb34c11b34debe05efff641 : SUCCESS in 35m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/256ed14eae7847c9bab436fb340d8dcd : FAILURE in 10m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7b3dbc36a5594ac49d775b210fe79a42 : FAILURE in 9m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135e1b6e7b74427f87e4c381dca18dca : FAILURE in 9m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1fe904cb068744e588ae9fe28973e4bc : SUCCESS in 18m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/586f1e18ba1f4dc3914428274141a16d : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f0480852ca190ffb1acdd32d0ea60e0685f94773","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-20 09:07:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e6b553cf52786314850156f6999022bfe7668c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:24:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/456b5b630da94e2c8965ee6fe332ca74 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a2b2197a5447bc99aad1f313c02cde : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37895d3c3b324a1fbb6abb30fe93dc66 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41c36778c448471e9ad1366db7493463 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29810d7d32fc4c61b858ac3024318689 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb93db0aab04e759273f03d42fc8829 : SUCCESS in 10m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/741a9dbf49424f7aa1dd0782e73c2b46 : FAILURE in 8m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/972ebe92bb934052820ab2664bf95cf3 : FAILURE in 7m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8d91bfc63e04ca6810c999694140ea9 : FAILURE in 9m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b8b31de4d0154e9b8193b6feca271c73 : FAILURE in 9m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2694646a47684e4f93f9c77b2b083d18 : FAILURE in 8m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ac132994f65243a3b3be8f7602bbde66 : FAILURE in 9m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5464a71e289e45f9b7514ba5a27af640 : FAILURE in 10m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fbfaaf5692644e1aaec80bea08b7fad : FAILURE in 8m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb85ba39dde24ed2b7eccf47985b4ef2 : FAILURE in 10m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e448bb69c4642f5809183b17392bf66 : FAILURE in 8m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/20ff2857b1884a2fa9a51ee2053d55ad : FAILURE in 9m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ecdda96fb0240b18f4a3099ccb3bfdc : FAILURE in 8m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fe463f92d76423c8dccfa9cbed90be8 : FAILURE in 9m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4e02dcea1064951bbb87fed0b39d36d : FAILURE in 7m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e9bd8a5eb0742928c10efbc0f46d927 : FAILURE in 9m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b2f39bd0d841e4b1745586cd4928b6 : FAILURE in 8m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d9d7f5bc0381063e96c161d657f93164abfd4dbd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 10:37:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a75831748036b05efc08da6ba5c9a8c686e922c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 12:42:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdd5e358e6e6491c9d67df40adee806a : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a47d6c1eeb9466e8c25a0dc321d500e : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c272801cd5f4319bfdadd8c547121cb : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58ab1b3b4e3a40ca997b9c67cb874b3e : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/415432ea783f4c74915d2df88417db97 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f908287370ef46cabc0e7057831e8eab : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ebdd06eaa91d4945938d4a3269bc4ce6 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c4a594547a64615b501539387888ee6 : SUCCESS in 14m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/007ecfcc45984050a292ef646caa5998 : SUCCESS in 32m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6442a00326bc4ca59fba50369b5f4799 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a729c3e90a4681ba78ded81254712d : SUCCESS in 51m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8b8a271c603b43a39f34f58c13994f01 : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/194862321cc34c10895dbd1ab07cf08c : SUCCESS in 22m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb0b64434f1948a6bb122cca7057e917 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc25f7e1ec8f45e091a0f63e7f2e6523 : SUCCESS in 1h 59m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a33d3fc977440559553e069c17a9ebb : SUCCESS in 1h 21m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f7c9ce731b404aeca8d726486d55f1b3 : SUCCESS in 35m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27c77bc824c342b682ac925fd05e42d0 : SUCCESS in 33m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/067830a0558c49bdbfdd777c2d0f94c1 : SUCCESS in 32m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/215201056aee4479aca2c84c5576b19e : SUCCESS in 25m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c1b500d38194cfd86dcba490f8a3c8e : SUCCESS in 15m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13f80f9dac04ec39ce702dc81b844f3 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":2},{"id":"086c232deb07151233dd51c8f090377a1956f2a7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 15:11:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ae9c1f8daebc22ad311429eeca41e41293f5e64b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 09:33:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94b487de8dd4aefd3c328b1b47d4f6bc01c8b574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:34:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41066ac847edcbce46fb1025efd370a40e381fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f72a14d2fd4c78966091c96535a2ea : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd863b0e62194f408d99af26f9a7ff69 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cee63fdd0532475091891ce35c704b84 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da13fc6d3704779947ca7e37e73fcd7 : SUCCESS in 11m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9e24fa922db47f0866a567675cb7575 : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7695df6f1e9746e889725c41a88be863 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af0251ebf2c443858f52ce5ecc2f5acc : SUCCESS in 32m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/21c1ee1a39c3489c9bd760b19f674792 : SUCCESS in 58m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ff594d976974a91b7bf1f500b16fd64 : SUCCESS in 52m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/03cc8b0cf1be4bf5a5eb7a92b084a067 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5625d8fae6fc45589044c25da97e4738 : SUCCESS in 25m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b600e695ce61404d8a405d0621bc72a9 : SUCCESS in 17m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1882345c746466992094743d84e8260 : SUCCESS in 2h 01m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/728487e40d084998b5a8a36a8538d5f1 : SUCCESS in 31m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4f4e29758c2446e84fc2de58b9b450f : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f7c7732cd2834feaad166753a5d50628 : SUCCESS in 27m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0cd0c8bedfa044e7abb73bf65a020283 : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b435dd2e6bd249bcb4f2b3b9178b13b9 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2af03fef57ce08621c3852a1bff2af97bb5aeac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49218b34db27827a0a364b739a157a94df4b1134","revisions":{"7d312a764b1326412331c8869c6ff41c31c0b4f1":{"kind":"REWORK","_number":1,"created":"2021-10-13 09:14:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/813786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/813786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/813786/1"}}},"commit":{"parents":[{"commit":"edb7101d7e68efa686158b032ead4594053e2ecc","subject":"Merge \"Update Kolla images namespace to openstack.kolla\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edb7101d7e68efa686158b032ead4594053e2ecc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-13 09:12:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-13 09:14:03.000000000","tz":60},"subject":"Remove stale config dump from seed hypervisor host configure","message":"Remove stale config dump from seed hypervisor host configure\n\nThe config dump was removed from host configure commands in\nI85670be7242bc436f73c689f027670b0938ba031, but somehow the seed\nhypervisor was missed.\n\nChange-Id: I4ae457a784423dcce96e51b4bb21219831acc854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d312a764b1326412331c8869c6ff41c31c0b4f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d312a764b1326412331c8869c6ff41c31c0b4f1"}]},"branch":"refs/heads/master"},"49218b34db27827a0a364b739a157a94df4b1134":{"kind":"REWORK","_number":2,"created":"2021-10-20 06:31:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/813786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/813786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/813786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/813786/2"}}},"commit":{"parents":[{"commit":"0b1a052b0eb534446891bde2862dab80f921ad14","subject":"Merge \"Prevent Bifrost from using firewalld\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b1a052b0eb534446891bde2862dab80f921ad14"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-13 09:12:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-20 06:31:35.000000000","tz":120},"subject":"Remove stale config dump from seed hypervisor host configure","message":"Remove stale config dump from seed hypervisor host configure\n\nThe config dump was removed from host configure commands in\nI85670be7242bc436f73c689f027670b0938ba031, but somehow the seed\nhypervisor was missed.\n\nChange-Id: I4ae457a784423dcce96e51b4bb21219831acc854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49218b34db27827a0a364b739a157a94df4b1134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49218b34db27827a0a364b739a157a94df4b1134"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
