)]}'
{"id":"openstack%2Fkayobe~814704","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I7049eae6518f818f9e180dfdb6f515d527644808","project":"openstack/kayobe","branch":"stable/wallaby","topic":"bifrost-firewalld-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-21 14:21:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7049eae6518f818f9e180dfdb6f515d527644808","subject":"Prevent Bifrost from using firewalld","status":"MERGED","created":"2021-10-20 06:16:06.000000000","updated":"2021-10-21 14:25:45.000000000","submitted":"2021-10-21 14:21:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814704-bifrost-firewalld-stable/wallaby","meta_rev_id":"41ed1b7b80d813648d0684672c182e8f5a38c3c3","_number":814704,"virtual_id_number":814704,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 14:21:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-21 09:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-20 08:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-21 09:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 08:04:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 08:42:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-21 09:27:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"32e4d62eed62dd1c0467fbed0063882167c9e7ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 06:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70fdfe348eef205728e4faa3aa286162a6dede51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:04:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b20cda1e3a240d1b8616aa8766ca408 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63b14dcb38f48baa3d6031452b5d41c : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26204fe4c64f4dfbb2807aff28a154d3 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ff8beefb4e41b6b986acfcd6e4ca32 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a28ae81030b44fda10f4d2ba1f76224 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0523915fb04979ad81c9879153f33d : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9e497bba68d4bd3872bac3494901ddd : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdf77f2a2bc54aba801d4c7b820a7355 : SUCCESS in 8m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5dc4e85510414c87898723b01ba3508c : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3354c03010e34482867bf732d67bfa74 : SUCCESS in 21m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9d67f507b98f4bd98dcf743acca3a5fc : SUCCESS in 1h 15m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a70e50338b604f48bd0748eef5f8ea5d : SUCCESS in 50m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/329c2b82ca1b40c9b3bc56c73a8605f9 : SUCCESS in 1h 17m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/14336feb5e764e5e9d8f4d47efbb7580 : SUCCESS in 23m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c9265f4a1a454ab95665a3fda09585 : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c234769481474af198294d02ef5daa69 : FAILURE in 1h 05m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/032db1c0499a48309136a4109d5a00d3 : SUCCESS in 40m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2398ab616534056bc53dde163adff75 : SUCCESS in 32m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d5e6d83e5bae4898b20f830c087745e7 : SUCCESS in 29m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3a4d9e778b1049f980cfc1a6e696dda8 : SUCCESS in 18m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69b9ffddff7c48ef9f45dc60a55e5658 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e7420247072f96e5e942d9120ed195a0e10c8f93","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 08:26:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"561b5c4b354517429cd82eed25360e34d11b92d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:42:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a0d32e7a766431483566f25b906a2fd : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be910ca10d94bf6ab07c41a120097df : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/502efd2ef85e4d4fae9934927a310406 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e683c1e7094173888bed8cd43a0977 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/559d0395572c4ccaa21ba47e2a3065a3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9a2d2e0f9d4a7b99c6dc80f4a3ca54 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d51eb196f9842c5a9adae649c80686d : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60203f339a394deeb134a280823afd57 : FAILURE in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8edc3398e8c94e0a8e710df8ae550464 : FAILURE in 7m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3589ef9f67e14003a0bbc997e85001ca : FAILURE in 8m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d19e09e44f5d4ad9958c67a90c6ffc05 : FAILURE in 8m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/870dc72361eb47f29a8c5de76653c0dc : FAILURE in 7m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2488ead1cd334e7491ca1de3a56e64da : FAILURE in 9m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2098efd757e24fea929a23a1c58c9476 : FAILURE in 9m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f356880b0ec541d18827371ce7b65831 : FAILURE in 8m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e3aed6f9acf4db3b2b9fb010ecaecca : FAILURE in 7m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3db67934f49403a938bab231adbfc34 : FAILURE in 9m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23faa6b7b9b14890a18e58a740ff6889 : FAILURE in 8m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/711ee28dccab4c79b597fd168665d0a1 : FAILURE in 8m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1bf1413c333f444493841cd793a8fcaf : FAILURE in 9m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0e90e5ae1d5477690b3d970fd025342 : FAILURE in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1b6f963b36c6deec18533bf4f56b1676a8eaba25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-20 08:42:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3fdb4b48f02a34692efe3e2ac6843ef2943bb88b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 10:37:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c175b4fa7125a371123ca507bf03a480add0ea98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 12:27:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9329e012ce754a4b94f0a89a54abfafb : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0c77e9a756477295b0d9a6b12625d1 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce699546fbd44f9b48acdc6e23742f1 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c28c1ac1a6054b5ba962addf4f86063d : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcb97cd069424391a15fafb17e91ab23 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f8397542c44e1180dfbff46db37264 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259a7769d5ba4146a0a939d907a8f6ed : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b388f4ee1db74da29e39d00526ace850 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b353a5d1bf247e8a93534233ab9a393 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/764455069472486293a67e233c6ac750 : SUCCESS in 17m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5a6dc0e367cd49109280ff8d0eeb6cd4 : SUCCESS in 52m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0ed2457f7364bd7a2954aa00133b600 : SUCCESS in 48m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3f927abd7bf441c1b44dbe5b0b4d1b59 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d074259bc47a45859bae753d782da7d7 : SUCCESS in 20m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8db38e7329e47cda977d38d1427b557 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f01fe7ef0e74437cbaa5f3424ab07997 : SUCCESS in 1h 49m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/202e171452c943588e96347c18b94654 : SUCCESS in 38m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75945afc32814be6ab4541efcf6281a8 : SUCCESS in 34m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0a28bb1ed8b4ccdbe8c642adb5b70f9 : SUCCESS in 30m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c9ae67e662c444ba8cb47e44dd696aad : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca3c6dd630b74ec3899b45989099ff84 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"af9daa55c3d0ebb90a12503c97749f86ae081895","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 09:27:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b8f7b55788bee284cb7cd80b05189f9f6078c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f140c0bdfbed44d91f706de45277e65164ffd16f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9235d4933940ef8995612611f21f12 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5815e9f3376f4efba319d4e47a3ed561 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3aa5c8a5e4145a7b690fb4377a96744 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74027ef7dde8416191e43816920f58e1 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d18893fd23042a29d2485fe541488d8 : SUCCESS in 5m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b6cffca6b7a4453a1e261d474e92e21 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d287dafc2d14c1b918964d52eac47b3 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72dd47302197436d8141f26d9edc489d : SUCCESS in 19m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/43c40bb24a3947a390c047186aa20c1e : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84cee38f5de5435c8a6d40eb74d233c5 : SUCCESS in 44m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9ff999514464181ae0a2457bf40b607 : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c6de8f87aa844c1ab88bc69431bd4a27 : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e3210d9c5af4f5ba6b8486d7ee73831 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33e1fd87edaa456d81f50f2cad426120 : SUCCESS in 2h 10m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/54734c8a27794ed484fc29d6d3f3af84 : SUCCESS in 37m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c646b55697224ed2b7ae21ab2c0064ca : SUCCESS in 37m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6497dba26a86401b8609114d83a4c62a : SUCCESS in 32m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9b5342e260124e5686df5fc163289927 : SUCCESS in 21m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61e4a418db2948218533adb26244ec06 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f304a81ff1fff3bc8bacf59889ac0fab3a8ebec5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"41ed1b7b80d813648d0684672c182e8f5a38c3c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:25:45.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551758082dfa4bb0a5208416771071e0 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e837075bd057469cafc7f289255e3a65 : FAILURE in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adb52541ef69bac127f48bee3e9eb38eea07aeee","revisions":{"adb52541ef69bac127f48bee3e9eb38eea07aeee":{"kind":"REWORK","_number":1,"created":"2021-10-20 06:16:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/814704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/814704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/814704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/814704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/814704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/814704/1"}}},"commit":{"parents":[{"commit":"5d69db6ebabc3dd6e4d3e68a8c21c942eed63593","subject":"Merge \"Add support for apt proxy setting\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d69db6ebabc3dd6e4d3e68a8c21c942eed63593"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-20 06:15:51.000000000","tz":120},"subject":"Prevent Bifrost from using firewalld","message":"Prevent Bifrost from using firewalld\n\nThis is to avoid conflicting with iptables rules configured on the seed\nhost by Kayobe.\n\nA new variable kolla_bifrost_use_firewalld is introduced to configure\nwhether Bifrost uses firewalld.\n\nChange-Id: I7049eae6518f818f9e180dfdb6f515d527644808\nStory: 2009252\nTask: 43442\n(cherry picked from commit 9f6c912b3418b21c06f53860e786b5b56daddff6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adb52541ef69bac127f48bee3e9eb38eea07aeee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adb52541ef69bac127f48bee3e9eb38eea07aeee"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
