)]}'
{"id":"openstack%2Fkayobe~814705","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I7049eae6518f818f9e180dfdb6f515d527644808","project":"openstack/kayobe","branch":"stable/victoria","topic":"bifrost-firewalld-stable/victoria","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-21 14:21:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7049eae6518f818f9e180dfdb6f515d527644808","subject":"Prevent Bifrost from using firewalld","status":"MERGED","created":"2021-10-20 06:16:36.000000000","updated":"2021-10-21 14:25:23.000000000","submitted":"2021-10-21 14:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814705-bifrost-firewalld-stable/victoria","meta_rev_id":"7b25e6761e19468449972d9ab1fb8fe41e7c216f","_number":814705,"virtual_id_number":814705,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 14:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-21 09:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-20 08:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-21 09:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 08:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 08:43:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-21 09:26:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9e30aeeff4c65112358e6729121cab4052932df8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 06:16:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77b035b39173629b59675811e0de0f6ddea80a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:17:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a00a5ee59f414ce6a01b165e4b6ff7d8 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b2b41e186242e4925e6e41941a9c06 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99969bc4116b4e19ba5de98e54c432b3 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ad3156a2a864c78b26ff3677a849d1d : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2117e03ef4bb4a98a47be0337e706bec : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259f0b4f484d4d7d998bf9e962f03a84 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c628ac8d324463b8a765f6d5860b502 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed93db68909f4c4aac837fb9aa7ead57 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ad3dc667d6246169b23ae777fdb84aa : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dabc74c7ff344cb4869d1639fe72d5da : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f01857b82bbb4a49ab70a9bec78438c9 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3bb5823fcf60434684a2899bd916c961 : SUCCESS in 54m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/75f38b593abd490f97cdbb29a7f3b2d0 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9378d252c1d94bc3b6358964e8028cde : SUCCESS in 20m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6969c2796eb3452b9604f1237a18163c : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7aee3feff8674c429e0526ef4751c4c3 : FAILURE in 1h 17m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/13a09eb5d92f48ed81c41a093ce496b4 : FAILURE in 1h 17m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bd3454f392df4ccf96b9d79a9b809a3c : SUCCESS in 31m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b45e957d07984b79a64d3b3e23713ab1 : SUCCESS in 36m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a4242ba92de7422f9e046f6e23640c46 : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc82766378a1427da5af7e5962a4f7ae : SUCCESS in 29m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5e26d8e12875423596a95a4108cd64af : SUCCESS in 12m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/55cbbdcb63e14aee97e63324e9aa1c78 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"af82cf3e74561aaeb8803f8041404de683f1b188","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 08:26:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"740ade86b674a49c67f3786cda0c6620f785d15b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-20 08:43:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f9de5f8d99f47b35880d6779f66f1f89c60d13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:00:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d60ed5aa4c7848148e8afe158fba3c30 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63cf9e338bc24bc2bce3dcf3c45f5cd6 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50492afa2c804ec19f0c4ae373edc57d : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/511aaa84a42a456285f5b9185e6303d8 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd4d3d0b83a497abbd8ccb6939f5f46 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48b258f64de74f4d9cc8f12944b7633f : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a259153f87af49cfb7bccdca89f91142 : FAILURE in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71b18b7a6d0c4d349ee69a6b30920ef2 : FAILURE in 6m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5466a710ed54332a4e0fdf9fe1ce14e : FAILURE in 8m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cd2c0681829349bfaaab85adbcb95b97 : FAILURE in 8m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2b2830c123e540a38ba1990596f3d2a3 : FAILURE in 9m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/85e9f0ff2d174712876979459dc85bcd : FAILURE in 7m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/db8f2693a7db462bbb41895552b51302 : FAILURE in 8m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/06f7ef8662184d768257d86b5595611d : FAILURE in 9m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4bdacc92edba475098becabae870440e : FAILURE in 7m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dbe8fed409d24481b2da1f9da3876777 : FAILURE in 7m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/442d69e4de534bff8ad7be37cca75e02 : FAILURE in 8m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f761d8d4d3b24601a6363a429b9036a4 : FAILURE in 6m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bab0663162b04c5588562c1a8e36c85f : SUCCESS in 28m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b087d36b70bc4cf18d7e382a206d6292 : FAILURE in 6m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a7ad403dc8814b10837eea4149208180 : FAILURE in 8m 41s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d8708d86697242c18c547e822ce74bc6 : FAILURE in 6m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ebe84c0d5a054df58b562eec18bd5fc1 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f0804275fcf0114f533cad351dbcd9d925377b23","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 10:37:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18bf91354e6bb0f22a57d0a3283922df8616e823","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 10:37:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"897677ff175b58eac2384c77b99c505025a8b2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 12:57:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/399c490841ed4ae3947d7a8692763194 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e65a575c9745faad4a1f8f22ca5863 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b98bc321e76f4f75bec1b998416edd84 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd207fc91f1c4270ba5b7194afb325b5 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395fdba59f954adbaf6ab362685a74d9 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04cbe856bdc6488293e80a54d23ed18f : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb3b37e9ff6e4ef5b25db1f79a206e30 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bc3da2f418244d88b975d2731e4b122 : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17aac19d33134d50b2486e5d0e2db185 : SUCCESS in 21m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3ee1fa87bcdf4bb689a4efe15476a34f : SUCCESS in 59m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/73d95a2e837749cdb6f4b06d8b0afedf : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/94a230c39e354856a8f96bbf00c04670 : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1fe56cdbb4be41c192194865edc91cb9 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f2fd438bbe904aac9d4412baa562f7b0 : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/20fb3e9e6a224feb8b20c12a0f5ac22e : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0bedacfaad154edd8ff639d32b912bd6 : SUCCESS in 2h 00m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c7d8918f65214ef887b02c6141f517d5 : SUCCESS in 2h 14m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c5eab301c5714f09941a0b9a2e6125e4 : SUCCESS in 30m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/55f3b3ad1d21411dae1093698e227dfb : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c959fdd0f243432ea95a2b9d9b3a7811 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c61bd3ac5e6649d790627dce56915835 : SUCCESS in 30m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c82bb515f21b4cd7aebbcb9609fd5497 : SUCCESS in 13m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9684efa70a3d4cf2a8751bf8b3718100 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"758ca1a999a6a864256eb4adc3a90cc8a10ec53e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-21 09:26:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0863efae78284052f680b656012e9e1d1e75af9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:27:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69c331542cae605952b25b898a8f47c885e912f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ead9d975fe04a2eb0f71c6f4392a849 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc11d0f324604646aabf38a575d2b70e : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2543ef4ed9f4684953038505c946d20 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66bdb149bd9c4457a789c54d13521354 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f7613dec5c54ef48556bee014879554 : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65d030973d354fb1a863378a5c2a8c2e : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3b50e6a5cf94bc6981b667648a321c2 : SUCCESS in 8m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f120c0c1d1b4069b04e2c4a28e5976d : SUCCESS in 19m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9ac5e9356e47401bb09a327e9e5fd3f5 : SUCCESS in 58m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/202c9ccb9baf48579a40a70c30192df2 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2fb6aa88b67741e9bb4ef75e86dcac90 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec80b25597f94a22aca1c027db86fc17 : SUCCESS in 56m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b2a9002f5ea24402ac77a3b020c6b629 : SUCCESS in 14m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a54ef523911f449dacdf3a27dbae0266 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d49ecdae5c914d9c80b207141415a405 : SUCCESS in 2h 16m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c4d752b8e814385aad8f6ceaf868322 : SUCCESS in 2h 05m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8d73e77fcbae400a8f1598f8cc3c0a12 : SUCCESS in 24m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c2cc92631ad14c70870d477bbf3b5ffe : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2d9a496f4aa4029bf27a5dffcc61db6 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e60b86c9a6024a72abc13f2dba7f70c9 : SUCCESS in 26m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/923d5f0c0e21455eb2ae9c44fd0cf705 : SUCCESS in 12m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/34fd190eba2c4da4801371c1b08e61ce : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6d91a210393af25298df8cd39cdbac5d995fbd95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:21:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b25e6761e19468449972d9ab1fb8fe41e7c216f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:25:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7aedcf266341aeba307bee710ba31a : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a372db41fc4400b35fa795b00bb8da : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e99a3e99902862f2fdfbfcb016c4fd00ddb7769","revisions":{"3e99a3e99902862f2fdfbfcb016c4fd00ddb7769":{"kind":"REWORK","_number":1,"created":"2021-10-20 06:16:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/814705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/814705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/814705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/814705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/814705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/814705/1"}}},"commit":{"parents":[{"commit":"459b6880500694666b3b9047330023ab1a5f16ec","subject":"Remove CentOS 7 image resolv.conf workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/459b6880500694666b3b9047330023ab1a5f16ec"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-27 09:40:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-20 06:16:26.000000000","tz":120},"subject":"Prevent Bifrost from using firewalld","message":"Prevent Bifrost from using firewalld\n\nThis is to avoid conflicting with iptables rules configured on the seed\nhost by Kayobe.\n\nA new variable kolla_bifrost_use_firewalld is introduced to configure\nwhether Bifrost uses firewalld.\n\nChange-Id: I7049eae6518f818f9e180dfdb6f515d527644808\nStory: 2009252\nTask: 43442\n(cherry picked from commit 9f6c912b3418b21c06f53860e786b5b56daddff6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e99a3e99902862f2fdfbfcb016c4fd00ddb7769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e99a3e99902862f2fdfbfcb016c4fd00ddb7769"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
