)]}'
{"id":"openstack%2Fkayobe~814752","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I478d6304c3f8d207bc7062a0909147a169003027","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-21 08:11:05.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-28 08:07:49.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I478d6304c3f8d207bc7062a0909147a169003027","subject":"CI: enable DNF tests on CentOS Stream 8","status":"MERGED","created":"2021-10-20 15:14:08.000000000","updated":"2021-10-28 11:51:15.000000000","submitted":"2021-10-28 11:48:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814752","cherry_pick_of_change":812301,"cherry_pick_of_patch_set":1,"meta_rev_id":"48cc4b8e9a2d61e6b1b0c6204c4f39b52a2535f4","_number":814752,"virtual_id_number":814752,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-28 11:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-28 11:51:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-28 11:51:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-28 08:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 15:14:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-20 17:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17a3d451addfc05ccdb18722d9d20fee6fc44a7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-20 15:14:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6ca47e78d746d472a8e33451dee7a5a7ab2a941e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 17:08:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6109fff8bc1e4c1fa2dac8b43fec3a46 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90cc7f5510484b6383a0e5be1bb44543 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5776f03c83347c7a881ba173cf4ebe1 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d99a897f950e4a098e38d1897fc4a607 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0342a822ca214eeb88199bf165ad30f3 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a805d710bcb4477a7a8437f1b205448 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fc3847124a74c99ac14d62821ebea2a : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5a7932992884f0cb59e2f43c12385c9 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91b55d1e348f4ad0b47eea3f1ecb8ae0 : SUCCESS in 21m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c29fa3f6d2934bb6a036bb8b268a7691 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfe58693c04f403ca3ff569a2a9196ef : SUCCESS in 49m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/98a5155d58a746adb498498bef25b94b : SUCCESS in 1h 14m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/78dc306b3afb44ffae9e3aa85226da68 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ac7a679d6a74e549acb097fdc3372fa : SUCCESS in 16m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a1bc34a43d5c4ec987b399019d774576 : SUCCESS in 1h 51m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1db4e12528384ca1bae3f2f01712b735 : SUCCESS in 37m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475a16badfbe49db98d5926bf2eaaef4 : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25d024108efe4c55a260d4b551c4f0d3 : SUCCESS in 27m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5377f14d3c1d453a9a116dc5e551522e : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e34f44bd817f4d1587aaa560d421cead : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bfd81383f11bcab180f0bfbe3441d67d2fff1198","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 07:32:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cac44beb9a7a5a8cf187b16b150a6d0e8315a973","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-21 08:11:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3dde476234323b50ec4fc535fb1c2d27c63a6e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 08:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14f43fd5693b3c475460a43c01d2536ada0ff69a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 10:26:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf77e8d7957143e4a3c19e14986a5296 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e14fd4247a648f49c6531dfb39d3c55 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8081b4c20a9a42f2a53ee157c936d3fe : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de4f76a3f534e5b91222a4d202bd761 : SUCCESS in 8m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26a7b35f40c64953bc0f0499da412578 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5a824b2b50946acb8303d443a4757f2 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0c5fe26d73a44f1896aab235e3ca9d0 : SUCCESS in 17m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40bad02aa7f44d32b47550e778010f1f : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/463a1d0b27b64e418889a751870def3e : SUCCESS in 42m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c712577026194a2eb5da7fa1a39807fd : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ec17868726284ec980e698048a60f3f8 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05474317de78456a891d0ee75da809b6 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2696f52138ab4223bb2ab3afc93bdbd3 : FAILURE in 1h 16m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5bf3b4cc736a410dbcf60bd03f003621 : SUCCESS in 38m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/346774869c794b2e8eb07fef0438b6ce : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b62465cbcfba4a4e893e1f99820408fe : SUCCESS in 25m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7c7c82b89eab4572ba04e23a3ebf7f6c : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b21db721a2649d69f3f6c8128dce829 : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0b5a233a92da1fc6244175641b6511f8ef998303","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 10:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78c483a986d28a2237c483a952b47efebd12857f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 12:10:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/964d2d77fbe24c3aa4ccfdbca65cfea3 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253ce1020e664ef2a5c8b9d3fdb9e378 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d624b45ee67942e3afcedb389f23e0b8 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db7400fba3d4e5c817fdf9b408ca772 : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc359b4a32b440b3a71dc9f3bd108d44 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c064ba37f89a43c490d9da3681017165 : SUCCESS in 11m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0009299112ff413d87b321ce4302066d : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/603eddf4022b4e10bdb006526ef643ae : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f310800379b4c3da67cefce00161d20 : SUCCESS in 18m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/84ab2fcd30d0458ab0736ce6a6cc4a7d : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6a95502eeed423a95a2a273a4b7932e : SUCCESS in 53m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e8a31251109842d7b05860e118d714a9 : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bdcddf142f234bf19e8c1d3395e07204 : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87b1be77561c40ac8eaac1a199c0664f : SUCCESS in 18m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/83cd717fd7ca450a829e3941f4854187 : FAILURE in 1h 21m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/317954e1baac49bcbe3aed79a055b469 : SUCCESS in 37m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9154fbe689aa4cda9480050c02afcdef : SUCCESS in 31m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/299aac7d4a564f60a73ff29265283ae3 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd51b79b2cc94033b2edd7b6c5054088 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dfce8248e3f463a817bc6591d4fb5a2 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0f460cfc52245875ee4ba409233fd74889e8a685","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 12:59:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f074f9f28d3c96da8ff0022f0f66864cb07c9a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 14:20:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecf84a6cd9c24f2cbc79a93d10eeb43c : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ed41e5875f4e609b16cc9b71f95457 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b4852cc63894d959bfa561768d05a18 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4d53cc0694744b4a7fb396251ecbe1b : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8a72471e9b417cb2e435a7726f2e4d : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2733bc55b7914e7e9b6f6211a9cea8fe : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9da06191bd204299ba5f296438037166 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74b4909139ff4319a19528193f28e090 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47bb1aa6be594ebd81026ee636cb2a4e : SUCCESS in 17m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dded467074794849810b99ec812b6121 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66058243f1b04704aad04923097fa211 : SUCCESS in 43m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8ad90646d23847daa33da9a0535d55ac : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/68778fb070084953bc43d947388df7ac : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b0538d4c54345ce80c61d4d3128c7ba : SUCCESS in 14m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/958d8841468d443abaf821af5b817c64 : FAILURE in 1h 15m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ee99297d49d4630991d04b68027f1ad : SUCCESS in 33m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6053ba3dc95410b980396b16b8a269c : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/42af6af20ed7427eaf500759e07b5f49 : SUCCESS in 27m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ac043174c21c47209865099fe4132fae : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff1c3b9dff0463697bc8c369b8fe44b : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"12172619453b8540e3e467091ced51f6336928ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 20:53:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"42dfd46948f561b0a94d815d8bdfd8230af085e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:35:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44179ca952d64b07b913b8bbb5be96c5 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf9c559644c34436884a229e975950fa : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81a0cca4b9c041a6b808bef39fe2196b : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d53162362b4fb1a04f627b787dc9d1 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d514490f5c94ec8885792f489196920 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f6e89e2b5c241f5bbd2424a695a83c3 : SUCCESS in 10m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f680edf48a54d4db982ed96777a2c48 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad1dab066b1440bcb6ca13aa4ada68fa : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/625345941a7046d9a79bd08b2376832b : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0328509d9f4b4a8487ec3a809536f99a : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a9695ea50a045cbb28514d4724e56b4 : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b1e831a286a449c0a0f3e6d61496f07e : SUCCESS in 1h 11m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/13192cb1444f4effa1a49cbb8ea55b70 : SUCCESS in 21m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f2fd620221447b4b19abd6b54ac0515 : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/de405bf6e57f4991ad67364dc8382d88 : SUCCESS in 1h 40m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e67586000d914e25a7dac64a7d3ab266 : SUCCESS in 28m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48b641f38e9a4473ae2dfdc691c73389 : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f9a4aaab19c4655859b713db0a09646 : SUCCESS in 28m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e460548d8fc44a479e74f050aa93f4ba : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bd6721c3fcc46689eec4c0bca48c774 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"600d67c1affe6bed556b71c1711f18124d93a25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-22 07:56:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b6742c216c1d1f72d4542149374c3f4fbef8ff21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:25:21.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5d2d804f4bbb25be0a09a9d1ced8394c1745d5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-27 09:45:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bd30d758094536b708488d8e52f3be7a1ae9503b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:43:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff3ffd83431e4e82874a7fdad61eace5 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f6be7f1f94406ab336c71791ba86c4 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcde0b23aba94cc39f594c798ea72882 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b47d56867b744cdbc669ca612cfb5a0 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36fa47eb005246cc920eee99b1f374ad : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903d3cb345a74050bf2ccfa7034a13aa : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f96ecf0ec8d4b64bba9ddadf85eb98e : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23592ca9c1274a90833d244f84028b4c : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f73db570b27e41cdad9a72e8042e33ca : SUCCESS in 20m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8a006ffd35784454a0452e992a089f3c : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fdbbe880c674627a15dd5caafced5d0 : SUCCESS in 42m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7bcd3fc7cb4e4299bed97ebcbb2dba30 : SUCCESS in 1h 11m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b38110006c9d4d8984a2dabef885071d : SUCCESS in 20m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc84074564d541b4b6a73ec1c785989b : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b51487a9537d4441819b84503b0ce7a4 : SUCCESS in 1h 55m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e3a30fb67d04c0a9aa556de9facec5d : SUCCESS in 38m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4450cf59c18c46b5a4169328782a0f46 : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/01e986f5d22046febff5146e249bd5bc : SUCCESS in 33m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/85c808b9da4a48929f68f9e21b07ea7e : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abe473bde4ae40529c36855b58370c76 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":4},{"id":"e0eedaab0b1620c9e657f8ea183b4bd7b834fce7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-27 11:49:04.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f58627dcb000bee839a09ce40e5e6545de990287","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 11:49:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7c445a55a256aeb7742f2f2a9729c88bc3188f54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 13:25:13.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161a9e4ac7dc4915bf0ba55c7775739c : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd571c0a03b94682921e0da43246671c : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4189ee69668b4162803e513276a32616 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3227968d0c93426383f27e2fecf6d83d : SUCCESS in 8m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b8f7ac4eb6e4a0d9cc1c26d35bc3640 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe9c9b784fdb4e02b423b12705602ed0 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c430f21aa0e842849d826861647dc715 : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4fdb7b123cd947a18a3cd1df82920aa6 : SUCCESS in 1h 08m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aeb67ac2b0e4a44b2182e9127588ef3 : SUCCESS in 43m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3f0609d237ef400eaea7a5f3123f99f7 : SUCCESS in 1h 15m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/33963b9df33d446bb0289db012e004dc : SUCCESS in 19m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10c37db095544f45af416b35de82ddea : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1049cb4ece6e437aa2998a6969a69e58 : FAILURE in 1h 30m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fbfbf858d53b4acc9c6c3aa6cd2ae590 : SUCCESS in 36m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f0da3d37c64c2aab23196a335eb025 : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e57bae5388424467b1d86e47e2663ff9 : SUCCESS in 28m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/448a735454cf4c7389429c02496e6e3a : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/487dfc32ddb4479496dcce983394301d : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":4},{"id":"841c00880078a2f0f1555bcd3b69a0334ed72e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 08:07:35.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"7f5207a55815260e600121408cd52150ede15fc3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 08:07:49.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8759ac9cd96f1b738fa252371f812d75dbec17d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:58:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ff29b191f14afdbcfdb117a2ea1f8e : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e3e266a94f46af86b6572715499711 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc068a119cd240719b02ef940afbeb9e : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22f5f093b90f4f85820f73d7871e5a77 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e20809217324b44a9e7dbc32ab6916a : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e50a93491b4ab7bb2242a83f99c577 : SUCCESS in 10m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa4b792ba55843abab13db60c053eca4 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6ea1cfad75d4e09b2f81a9618802918 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/339804ccd69b49bdae78000b44d49c7c : SUCCESS in 24m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7030da9f6cf4425fb0a7f33fc5a97aba : SUCCESS in 1h 47m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/596e40cf8093433cafaad88800574845 : SUCCESS in 56m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d52cc977e7ac4f30aca1f0112d94356b : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4bdeea50ca6a4611a9657a2d17436ac4 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85f11a7c20e448caa001c9ceb02a56c4 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2916475b4224944a3b977a777cf1fcb : SUCCESS in 1h 44m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/38bdc297c0ad415d918591263d56c5fb : SUCCESS in 35m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0ee82cccc6f4c06bcea1d720b008796 : SUCCESS in 32m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/745fb46710fd4487b8cf47fbc6a8f632 : SUCCESS in 29m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/510172310e7b4af2adf22d0f9ad23645 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8d73ecea09445dcadeb6793e1b82a17 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":5},{"id":"1c4eeea56c766ab544dbed27050dcf4ae47c2e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:59:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"17aa5a51e0f24bc99b5e03da28c10a57edd0ec1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 11:48:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a480838bb194b55939645598d9c9599 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78cc4fe00a2540fc96673c4dc144f68b : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d13350023de450fb7df0b15a42ba9b2 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ff90f147e44ecbb40d1f051be4421d : SUCCESS in 9m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de42e077d2404af6846c65feb9f914c6 : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5caafe1cdd7a4664a15874a290586591 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7de128309bcd4ae7902e5ce6e5977fd6 : SUCCESS in 20m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3f8fea33f8474e24a7f035e9ae5e1c0c : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e451f656edf46bba7391789165646cf : SUCCESS in 46m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/56f2abffdd8e4ff2bfad40d857caabf1 : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/937c8abe790c4e63b08797b55137a180 : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21e2f4c13fd04224b930cb5e52a7eec5 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7bedf45e44f4bb1bff9412d526349bc : SUCCESS in 1h 44m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/936f2cef35b646698a99c82a08d4ac9b : SUCCESS in 34m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1fd1afd32924c62b9f001a8fd27ee69 : SUCCESS in 36m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e0075c67f544971862971de8a048384 : SUCCESS in 26m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0fe6b6a3c2d44f4da13386c4d2aa34d5 : SUCCESS in 15m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98999e4f8f704572b3d2bf7c7a4859bb : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":5},{"id":"f28b1eb732133ed3922bb391920c7870e08db5e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 11:48:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b59260685276266a3c1379fa8d85baead7e42c5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 11:51:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19d160593644409badf0c89182a2884 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fb86cd16dba16bb09f335c0025c19e24572570f9","revisions":{"4fbbc75b79a9eea9db85d867be883524cf884948":{"kind":"REWORK","_number":1,"created":"2021-10-20 15:14:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/814752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/814752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/814752/1"}}},"commit":{"parents":[{"commit":"5d69db6ebabc3dd6e4d3e68a8c21c942eed63593","subject":"Merge \"Add support for apt proxy setting\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d69db6ebabc3dd6e4d3e68a8c21c942eed63593"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-20 15:14:08.000000000","tz":0},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n(cherry picked from commit 6a69a4f4db9f66978d9b593b0f510b1046fa969b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fbbc75b79a9eea9db85d867be883524cf884948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fbbc75b79a9eea9db85d867be883524cf884948"}]},"branch":"refs/heads/stable/wallaby"},"b505aab3234190376a8737dd842b5ce06bd322ec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-21 20:53:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/814752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/814752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/814752/2"}}},"commit":{"parents":[{"commit":"a452bb79384e08ca37c36460856c5ad2a951c3d0","subject":"Disable heat in upgrade jobs to save disk space","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a452bb79384e08ca37c36460856c5ad2a951c3d0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 20:53:05.000000000","tz":0},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n(cherry picked from commit 6a69a4f4db9f66978d9b593b0f510b1046fa969b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b505aab3234190376a8737dd842b5ce06bd322ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b505aab3234190376a8737dd842b5ce06bd322ec"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"5d64e0c1d7b12c05de28c6e292c2c4e76861cc72":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-22 07:56:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/814752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/814752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/814752/3"}}},"commit":{"parents":[{"commit":"277121552300907ac6eb1308d1b4a64b48b4a2d5","subject":"CI: Disable heat in upgrade jobs to save disk space","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/277121552300907ac6eb1308d1b4a64b48b4a2d5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-22 07:56:14.000000000","tz":0},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n(cherry picked from commit 6a69a4f4db9f66978d9b593b0f510b1046fa969b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d64e0c1d7b12c05de28c6e292c2c4e76861cc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d64e0c1d7b12c05de28c6e292c2c4e76861cc72"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"b1174b85cbc984e45826aff5213a46a4c2adbc10":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-10-27 09:45:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/814752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/814752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/814752/4"}}},"commit":{"parents":[{"commit":"13afea324e9d21292f0d3c928ab10ac7be788189","subject":"Merge \"Remove stale config dump from seed hypervisor host configure\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13afea324e9d21292f0d3c928ab10ac7be788189"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-27 09:45:17.000000000","tz":0},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n(cherry picked from commit 6a69a4f4db9f66978d9b593b0f510b1046fa969b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1174b85cbc984e45826aff5213a46a4c2adbc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1174b85cbc984e45826aff5213a46a4c2adbc10"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"fb86cd16dba16bb09f335c0025c19e24572570f9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-10-28 08:07:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/814752/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/814752/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/814752/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/814752/5"}}},"commit":{"parents":[{"commit":"cc103b718f7c3937fd75fcb8664d9814386fd72d","subject":"Merge \"CI: Disable heat in upgrade jobs to save disk space\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc103b718f7c3937fd75fcb8664d9814386fd72d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-04 08:48:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-28 08:07:35.000000000","tz":0},"subject":"CI: enable DNF tests on CentOS Stream 8","message":"CI: enable DNF tests on CentOS Stream 8\n\nThese were inadvertently skipped following the move to CentOS Stream,\ndue to matching the distro name on CentOS Linux.\n\nBackport to Victoria.\n\nChange-Id: I478d6304c3f8d207bc7062a0909147a169003027\n(cherry picked from commit 6a69a4f4db9f66978d9b593b0f510b1046fa969b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb86cd16dba16bb09f335c0025c19e24572570f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb86cd16dba16bb09f335c0025c19e24572570f9"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
