)]}'
{"id":"openstack%2Fkayobe~814920","triplet_id":"openstack%2Fkayobe~master~I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9","project":"openstack/kayobe","branch":"master","topic":"ci-disable-heat","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-27 15:18:24.000000000","reason":"Removed by Pierre Riteau using the hovercard menu"}},"hashtags":[],"change_id":"I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9","subject":"CI: Disable heat in upgrade jobs to save disk space","status":"MERGED","created":"2021-10-21 13:51:23.000000000","updated":"2021-10-27 16:04:59.000000000","submitted":"2021-10-27 16:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"814920-ci-disable-heat","meta_rev_id":"f29c306f7fad0160ca7e01d9225b2f8ea1472675","_number":814920,"virtual_id_number":814920,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-27 16:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-27 13:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-25 08:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-27 13:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 16:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 08:16:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-27 13:58:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"641e5267feeede6f1eb9a3faecab46885c99da57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 13:51:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22789577d52960ae96b9419c52c57296af41e27e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 14:09:44.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/kayobe/+/814932","accounts_in_message":[],"_revision_number":1},{"id":"af272f490f86b50b773b5da2dcdeb85b56a02927","tag":"autogenerated:gerrit:restore","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 14:09:53.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e6f1f6d6cdeaca9c46f34e65bf1d992280cb435a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 14:10:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d8cfee6cbe9a06d92fad907abbb02dd22fd222b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:16:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a78c666484884934a9dcc1b06f8be6b1 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210ca36484b04106b9374978f2d79814 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/368bb9b24e2349a18f26da3c0832427e : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce53f3ff8a664b749481eed1b80041bd : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d75af2e24b5455f89616cb222c3deda : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1009f15e03d4268ac3981b1740d3356 : SUCCESS in 9m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1628cd5a73d74cd8a2c40ffccd2bad02 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9c67b2cea9a4b798b00b611b96a2700 : SUCCESS in 14m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65bfd6b434e949c5a8fc7890e5b6e8f6 : SUCCESS in 35m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8fd91c27df354ffea9d5cbc27781d80f : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba3e9928a499406fb599139841a5273c : SUCCESS in 46m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c09ceafd8d254e7494966783fdcd21f3 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b0602d4d33f842f79070a2a01bbd5250 : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3881020d780b40579b98b3ab517fefbb : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7ec76457b1b466abdb13b4e7c01c558 : SUCCESS in 2h 03m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce017814bcd74306b5d5fd0167abeb76 : SUCCESS in 1h 16m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3886041c86a4066a8f9d7ecfe20167d : SUCCESS in 36m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35ab5ad4b42b404e99bb5834c4371e06 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bff8aad0d564f0f898ed915d7fbaba7 : SUCCESS in 32m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bc414b0b17b4b1b9c25c2070ecc83ae : SUCCESS in 25m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1b217860c55c432c88414b0b483eb3ba : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49a97fa145604dfcb71dd704de101b40 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b5902d157f08b14ecbe0ad621d6bd13dc2318620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-22 07:55:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"24c432bd3e7f9bb4d0e9d623096bf16c8a888027","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-22 14:39:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4276c279965dd6dc25945ca4af6582939e5f0798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 16:13:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f233f1b729462aae7a779f27615fda : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad4a63a9c3c49d49fc3865938a59acc : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb82f03b62294734b01db5440d31bbbe : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc6cc0e648f7458db4d15941c4369be2 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d3f388e0c144e3943bbf454a8d042f : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548cab71e0e14dd4a03c0bb2af8b4d7e : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92e42ecbd855412c98c1ff0aa38f54c3 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5cc4f3ed05c94f9fa43e84de016e9f98 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d64ad936b9cd46de9440b774fcd520be : SUCCESS in 34m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f1c3317cc42d4213843031be1d1a9229 : SUCCESS in 56m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf90691f9c61451ba874c3f608228f27 : SUCCESS in 36m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2e0d4743f2d463a8230a5b5e851d9af : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/31fd6c1eeb604f5fb93322d06854c874 : SUCCESS in 29m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01fcbbf5a6054385bcdf3022f3493985 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9051c059c3bb40a19ee049cf818dcd23 : SUCCESS in 1h 32m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b55bdce3b054f0e82c1b99a0078be95 : SUCCESS in 1h 19m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/85e4dcb69a674806888703efe765639b : SUCCESS in 43m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4627f8af7f494febbaf1ec2bda6529c0 : SUCCESS in 32m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e298e81c31a4e80a5333fb65ba73b0c : SUCCESS in 31m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ea0ace1f32d441fa3aeac925848964b : SUCCESS in 26m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a9303b05bb214ea4bc74970e93c486bf : SUCCESS in 18m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/108d6ff9ed3b4e69937020f4aa5cdc90 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"474a88d262dbce8d8658af0cf19d937c7dc5e09b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:16:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a424827bec878f595b9476c97549852c1ec9b9e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-27 13:58:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3449a653bfc29b254898be849b2be03bf0a56ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 14:01:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"686019b5851d2540e7ecb44a1ceeda47c5dd9fbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 16:02:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf905ffd6aa43c8882a50515729679a : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7252f86c6f6b40ef8484709d8aba4a47 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc2a78940f564af2b1ad2099d3adcf2d : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d9b5499d38427c962921f2187bb4cd : SUCCESS in 10m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b254bf316e94d30895281c7ff7076bc : SUCCESS in 7m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83a1ed1bce7c4d1fb70704b0adc42a8b : SUCCESS in 14m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5bfed75853a644c6b05660b5ff228cd5 : SUCCESS in 29m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0b275bfc79d841719b286b5ac1b64163 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af84b91e5779489985545aac3b00485e : SUCCESS in 45m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4c2e7e70597244c0a42b19e219f37e50 : SUCCESS in 1h 11m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/891fa5f9357c4b50870931e3403cc228 : SUCCESS in 27m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d83fdd686d4627935a2135a957191f : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7b97e0ed15fb47e9ac42831457b42f03 : SUCCESS in 1h 58m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5f2eda544e424cb4b7e567bd49bfd8f1 : SUCCESS in 45m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/374600576aa7490682cea172105aacad : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e071b46e58e843ac966210991a12b6fe : SUCCESS in 28m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8cc9f94436ce4f8db80cada5073dc93d : SUCCESS in 16m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b318d6774d14958a515c2b62af38a8f : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1f7434dec9828f422e4d5611c1e0b98aedb6028d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 16:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f29c306f7fad0160ca7e01d9225b2f8ea1472675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 16:04:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69d85bfa6b947fdbd1b6d4da2d7c865 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06a51ceaec37652df8573e48886a9203ec6e9c31","revisions":{"e1664fd18d0596344dbf7c8095c74133a91d6df1":{"kind":"REWORK","_number":1,"created":"2021-10-21 13:51:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/814920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/814920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/814920/1"}}},"commit":{"parents":[{"commit":"0ff94df3533e0adb64be3713eca0a15036cc424d","subject":"Merge \"CI: enable DNF tests on all CentOS versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff94df3533e0adb64be3713eca0a15036cc424d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 13:49:44.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 13:49:44.000000000","tz":120},"subject":"Disable heat in upgrade jobs to save disk space","message":"Disable heat in upgrade jobs to save disk space\n\nWe see frequent failures of upgrade jobs on stable branches due to lack\nof disk space. Disable heat in an attempt to free up extra space, since\nwe do not need it for our testing.\n\nChange-Id: I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1664fd18d0596344dbf7c8095c74133a91d6df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1664fd18d0596344dbf7c8095c74133a91d6df1"}]},"branch":"refs/heads/master"},"06a51ceaec37652df8573e48886a9203ec6e9c31":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-22 07:55:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/814920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/814920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/814920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/814920/2"}}},"commit":{"parents":[{"commit":"0ff94df3533e0adb64be3713eca0a15036cc424d","subject":"Merge \"CI: enable DNF tests on all CentOS versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff94df3533e0adb64be3713eca0a15036cc424d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-22 07:54:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-22 07:54:00.000000000","tz":120},"subject":"CI: Disable heat in upgrade jobs to save disk space","message":"CI: Disable heat in upgrade jobs to save disk space\n\nWe see frequent failures of upgrade jobs on stable branches due to lack\nof disk space. Disable heat in an attempt to free up extra space, since\nwe do not need it for our testing.\n\nChange-Id: I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06a51ceaec37652df8573e48886a9203ec6e9c31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06a51ceaec37652df8573e48886a9203ec6e9c31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
