)]}'
{"id":"openstack%2Fkayobe~814932","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9","project":"openstack/kayobe","branch":"stable/wallaby","topic":"ci-disable-heat-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-28 07:58:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9","subject":"CI: Disable heat in upgrade jobs to save disk space","status":"MERGED","created":"2021-10-21 14:08:12.000000000","updated":"2021-10-28 08:00:49.000000000","submitted":"2021-10-28 07:58:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814932-ci-disable-heat-stable/wallaby","cherry_pick_of_change":814920,"cherry_pick_of_patch_set":1,"meta_rev_id":"455d1e0c11c9f4de71319f28da19ff4f761e8893","_number":814932,"virtual_id_number":814932,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-28 07:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-10-28 06:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-10-25 08:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-10-28 06:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 15:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 08:16:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-28 06:10:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6e7866f0e4d2c2e88faccc8d5ecfc487d69a014e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 14:08:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df4a1b1177a6713fcfa982720af42cbf3938cfa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 14:09:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1eccb7682f91424a52b4218d4db25afcf54023d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:29:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e5b8c06f0c64fd4a856cccd8deff0db : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464ceaacf8044a2f91b0efd7f6558cef : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feaec889d8074259ac4c2332c30c087d : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b943596b94e644ec9c91760365dfa5c0 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecdd7e992d554396b3c6a27d5d107c25 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b27b309c9743e29299c2c8b59fa8ff : SUCCESS in 9m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a0c1ac2843246aebd7545b5d9adfe54 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/366372b76ca04e8094e48ec45c0a2023 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa3345b031d84c7bac76db83db749e84 : SUCCESS in 21m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9120549415b54c42838ae12aea8ab2bd : SUCCESS in 56m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48771db35f8644a3a37528f489ef52e2 : SUCCESS in 51m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6a867b62125a473c8837c5f2f8e80842 : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9851ebed187b41f7b9335116427af5f5 : SUCCESS in 20m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fdddbbe6b3a4924a64578636c6cb5c6 : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/90f74420434248c592ccd447c5a0511d : FAILURE in 1h 11m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/855a1e1614124c98bba7c39626054baf : SUCCESS in 47m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7d3ebdbb58048e0928f851f21b14997 : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9b2ad3d0bd74226bc6d59f4793909d6 : SUCCESS in 31m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e1671fe1dbfb4dcd9c4e0065a3eead3e : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7430579c348240ad938e48aa0e15b071 : SUCCESS in 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f727fe45197caf93f2abe765bcee830acfd6e7f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-21 20:49:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f067d4dcbfb63ad26b3b4f748005e2af6c49f7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:35:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45cce9c9d6ab4e629cda5b5fc07a6a26 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9552287cd5641f5af670eb0c13bbd09 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1418079854de408f98bccf6d8a42080a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb7a0762c3d2406ab8758404b6560a8a : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/692118dac1a148dda3cd1132494ea94d : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3054ba18e248a789e0c99666aed6ce : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0aba90c17b5a4be099b83bb5b1d2da9e : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3347352b62644a4f8782533e5b5bb6aa : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b5188dcfdc3440e8de92e6e0604730a : SUCCESS in 20m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7d2218265f85477092034a58f6fd08c1 : SUCCESS in 57m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35f7804a384c467698f2479e24031e72 : SUCCESS in 47m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3993b69a5a3d4ec48f0bd469c383c069 : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/66ff97c442234b6294d1953a738e7fdb : SUCCESS in 20m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4de9f0417154c65b0f37c51a731b457 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/89b5713ba61c4172a0f0a036cc924d1d : SUCCESS in 1h 44m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e30bb9ab8ef64b72b65a108326bec96b : SUCCESS in 37m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129f86731b0b440fb9f321340f40cf9f : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3eef2e602a748b5a15651b7bfbe1748 : SUCCESS in 27m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9fb649ea29b64c5aa83be0dd2860df6e : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16d35509579141d8a825f83202c25227 : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e8412b5c3413bddff00a4d776503942cd0b47150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-22 07:55:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19e1cfc2c78b7e3cba9a9a73cb93d29e0395b6c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-22 14:39:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e107cb2aafa1d7a818e1e3b4ebde801e67c48788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 16:29:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b652861a694e51a5ded96781ead5ea : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860fdc7ac01e4f3f8fc0541e34e16d9c : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de200a4fbed476bac86689a638aec20 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b9a772b8d0441499334f36234bd043 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceeaed81def849a8b3e8a1d52694ae83 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc057f34afe481bbea3f2737fb3bda4 : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75cc2f043abe4f51a7f4aacb243f0524 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32df8675f99c45699090bdcfc416b0c6 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca30d914e9d04eac9d6f638b06f44f51 : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e9d2d34586ba486cbda1ebb59bad212c : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3c504d1342f4986bf574c81b8928666 : SUCCESS in 49m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b5f543bcd0344f92ac5b6ca616bce0f3 : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/12c361cdbee24af0aba559bbd4918db6 : SUCCESS in 23m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fe534792fca4f80a87b82a122e5aaae : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c80dc4e947c14a66b34a557118811792 : SUCCESS in 1h 45m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e0f41fe691724e1687dcc2eef4105aaf : SUCCESS in 43m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dacded5dfa2f4fe79428665a9f4cd1c7 : SUCCESS in 29m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6fd46d0b8e374922aaaebe30b42fd526 : SUCCESS in 24m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2c07f65cf77c4078bd9afe1688540da8 : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27bf495277cf465bafa485eebbc03a25 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":3},{"id":"909252957af74b32a8c92a5791d9c65b4c234035","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-25 08:16:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"80d95bb825a19e30101ec0926c996c5767699ba1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-28 06:10:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f4bbe66f2e1819cfd4ad6a5e0909a1268eb72ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 06:10:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e2e39f2729301d7e6b86f804a56d40a57aab009b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 07:58:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8cc98d787a4d6ab7b17a0dae8e532d : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a335df135cb846b5b75a9ed7bf856d2c : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e77371b6ed4eeca631d355f3d6c3de : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdad0b9ecfe4a9eb1c5179bcebd097f : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b219a62d3294450b33ac5861c158f2a : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f4b9db864ef43ec9d7080a91537c9f2 : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8cf507be8d024ba9b0920c029af7ad44 : SUCCESS in 20m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0e1f3340004544f580e07e19ab6eab7a : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5115a376e9824e8da9d3da28a80c043b : SUCCESS in 45m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8390703a73334e07b1d3d3dc4bd6f100 : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/42934c7b411343a3a5f12282e32b06de : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f0766ce1614eec944112ec2c2beda9 : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dbc7a35f57ed4a20b1b2946cd4a7464d : SUCCESS in 1h 46m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/971b03f619c948a8b7139175e95f83e1 : SUCCESS in 37m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/316132445c844fdaa71607520c9559ac : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ef11479f9fc849c587c0ed1984e2f724 : SUCCESS in 27m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4333e90f6ee4e298ee92617135932b2 : SUCCESS in 20m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56b9810d290043948531f7c95ab7ec1f : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":3},{"id":"fdc36b9d99d1e79f15018489884cb2c459e9fd7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 07:58:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"455d1e0c11c9f4de71319f28da19ff4f761e8893","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:00:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848eacd385d34d82b7b957fd9f1ee991 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"277121552300907ac6eb1308d1b4a64b48b4a2d5","revisions":{"41c399a94340e9f74d5a9bc65ef57eee8c504367":{"kind":"REWORK","_number":1,"created":"2021-10-21 14:08:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/814932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/814932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/814932/1"}}},"commit":{"parents":[{"commit":"4ee1f33b2af230ffba9322486bae8ab90624ece9","subject":"Merge \"Make setup module arguments configurable\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ee1f33b2af230ffba9322486bae8ab90624ece9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 13:49:44.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 14:08:12.000000000","tz":0},"subject":"Disable heat in upgrade jobs to save disk space","message":"Disable heat in upgrade jobs to save disk space\n\nWe see frequent failures of upgrade jobs on stable branches due to lack\nof disk space. Disable heat in an attempt to free up extra space, since\nwe do not need it for our testing.\n\nChange-Id: I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c399a94340e9f74d5a9bc65ef57eee8c504367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c399a94340e9f74d5a9bc65ef57eee8c504367"}]},"branch":"refs/heads/stable/wallaby"},"a452bb79384e08ca37c36460856c5ad2a951c3d0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-21 14:09:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/814932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/814932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/814932/2"}}},"commit":{"parents":[{"commit":"4ee1f33b2af230ffba9322486bae8ab90624ece9","subject":"Merge \"Make setup module arguments configurable\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ee1f33b2af230ffba9322486bae8ab90624ece9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 13:49:44.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-21 14:09:12.000000000","tz":120},"subject":"Disable heat in upgrade jobs to save disk space","message":"Disable heat in upgrade jobs to save disk space\n\nWe see frequent failures of upgrade jobs on stable branches due to lack\nof disk space. Disable heat in an attempt to free up extra space, since\nwe do not need it for our testing.\n\nChange-Id: I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9\n(cherry picked from commit e1664fd18d0596344dbf7c8095c74133a91d6df1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a452bb79384e08ca37c36460856c5ad2a951c3d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a452bb79384e08ca37c36460856c5ad2a951c3d0"}]},"branch":"refs/heads/stable/wallaby"},"277121552300907ac6eb1308d1b4a64b48b4a2d5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-10-22 07:55:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/814932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/814932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/814932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/814932/3"}}},"commit":{"parents":[{"commit":"22d03c37c46f5ec539ad9b17eb18746a386da6b3","subject":"Merge \"Prevent Bifrost from using firewalld\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22d03c37c46f5ec539ad9b17eb18746a386da6b3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-22 07:54:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-22 07:55:24.000000000","tz":120},"subject":"CI: Disable heat in upgrade jobs to save disk space","message":"CI: Disable heat in upgrade jobs to save disk space\n\nWe see frequent failures of upgrade jobs on stable branches due to lack\nof disk space. Disable heat in an attempt to free up extra space, since\nwe do not need it for our testing.\n\nChange-Id: I20c99ca9dd6e78b041e4662c9b1b54eac904e3c9\n(cherry picked from commit 06a51ceaec37652df8573e48886a9203ec6e9c31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/277121552300907ac6eb1308d1b4a64b48b4a2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/277121552300907ac6eb1308d1b4a64b48b4a2d5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
