)]}'
{"id":"openstack%2Fkayobe~815840","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id912ccd7c4bf2a4ccff8d16f2a9aa6ed2127a335","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-28 13:30:53.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-28 17:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id912ccd7c4bf2a4ccff8d16f2a9aa6ed2127a335","subject":"Fix --check argument for overcloud host configure","status":"MERGED","created":"2021-10-28 13:20:07.000000000","updated":"2021-10-28 17:40:41.000000000","submitted":"2021-10-28 17:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815840","meta_rev_id":"56f3a98fb58f7aebbe4a35daad8a3f606ae57fff","_number":815840,"virtual_id_number":815840,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-28 17:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-28 13:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-28 13:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-10-28 13:30:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 13:20:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-28 15:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98ddd004de251e5cb0e56d0b0276b5523043821e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 13:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3ce17bd93ef51229a33c6330192f0928c02c126","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 13:22:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5a4cf5aaf6ee3c4f069a224557a0c9a77b73f34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-28 13:30:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40f6619cd6642500730c7fd31ff159ebe9ed487c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 15:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/667073226cda4715a8b0fbe4ca44d0d8 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb3bfa53c6e456f81e5cf6a90b984db : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46356aea6414cb484f04ce9ab9a8024 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b87b7ea710a34b3aaf481e1a71f81965 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d73163a1454344a184df92c576dee3d9 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed0cd778eb04d099f30113c8b427fa3 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3349cdc122ee44779b3b85ef315bfe0f : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8689fff6ed434e159a48e0f09c92317b : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1264e1b4e9f94306b0225a716fef1a09 : SUCCESS in 13m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/846e766e09664c1cb8d344ba358b1c88 : SUCCESS in 36m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e1f5929477640a1bb948b34f603cc07 : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f875d2fce4a04668a7c85861a74e6da1 : SUCCESS in 44m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e6b4b0a6c4b2434ab862ad04414fb84c : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/23cee73a9a1749f28d474fcc7557a464 : SUCCESS in 24m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813bfde9e4ad49c8aa0c32a7a6e52f55 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5b70704be2bd4e5bacf94bedf7b3ebb6 : SUCCESS in 1h 47m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbf8c7c5b0aa43b1acc7d6ae268cc7cf : SUCCESS in 1h 18m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8c6225982b340e39d7e78fceab3ed00 : SUCCESS in 38m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a22f996be3ed4a33a946be1cc286039c : SUCCESS in 33m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ddff54449a9410e8287a4475bf661e0 : SUCCESS in 32m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e09e07411af740b79c3bb3dd2073ec88 : SUCCESS in 29m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/83e1f2afed62498295bbe09e35a1399e : SUCCESS in 20m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c9b530af47c4b1b9f0d7bf0128549ae : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5baf8568b8558b8cb21aaf95aaed4cf1397af101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 15:21:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0da29a4a0a5896df38ef70269cf0d3ec5ae9b966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:37:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16f8ef231734a3ea92ac9bd89ef41ef : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/082177c6d0c54a8e9d99e68091d9db2f : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25b47965f8314601ab6742cc212abfa4 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f19bf71afb4215a36d49214fcab0ca : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43ec43a8f9f4446b802690730b2eacea : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e900c46f063143ed856feb3671e2f025 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8b5258d81374e5f879c47cfaac7f428 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8799144d2ce14c4e8914a8668d7fed5c : SUCCESS in 30m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ca2a6f8e91b5487d8171964c769b89a0 : SUCCESS in 1h 13m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7801481514004e9a866a7d32e2f2cc99 : SUCCESS in 54m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9b086e7dea394d38863581db84c15159 : SUCCESS in 1h 14m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b962b677c784f80a426a5b815c9033f : SUCCESS in 25m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9b42a8dea524359846637371c2aea2c : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9bb2fe96d5e0418ba4e2c21575358b1c : SUCCESS in 2h 15m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/376a296cb8984ccb8a71a1b87aca5b8b : SUCCESS in 36m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c94fefcd531400db8c6994e2509f754 : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b9705ba64ac49ddb1cb429109efcf8d : SUCCESS in 32m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bb2bb4e7edb44c0ca1fa4f63c2b43ac9 : SUCCESS in 16m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac91b358557e4cb0a3f4e084bc4bfec2 : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"942ea515c05de65e5a19548a0fb0fd4da823341f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56f3a98fb58f7aebbe4a35daad8a3f606ae57fff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4e0b7dac4b4945b4d28e79ce1aa856 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d784bb130d4598b928008efeeff464 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b207c1a3421e6e449250d59094485d21adc08488","revisions":{"b207c1a3421e6e449250d59094485d21adc08488":{"kind":"REWORK","_number":1,"created":"2021-10-28 13:20:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/815840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/815840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/815840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/815840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/815840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/815840/1"}}},"commit":{"parents":[{"commit":"3a7b54fd8fdcb3e352e4decfc1e75539c9032e1b","subject":"Merge \"Fix link syntax in release note\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7b54fd8fdcb3e352e4decfc1e75539c9032e1b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-08 10:08:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-28 13:19:55.000000000","tz":120},"subject":"Fix --check argument for overcloud host configure","message":"Fix --check argument for overcloud host configure\n\nWhen running \u0027kayobe overcloud host configure\u0027, there are some tasks\nthat we need to run even in check mode, to ensure that subsequent tasks\ncan execute. These tasks should not change the target systems.\n\nChange-Id: Id912ccd7c4bf2a4ccff8d16f2a9aa6ed2127a335\nStory: 2004798\nTask: 28951\n(cherry picked from commit 533e9ca912e882578b4a413ab96399f80b4da0fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b207c1a3421e6e449250d59094485d21adc08488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b207c1a3421e6e449250d59094485d21adc08488"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
