)]}'
{"id":"openstack%2Fkayobe~815841","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I758aff1d3ef714f1c8ef82d29dd2217734a9aae6","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-10-28 13:30:27.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-10-28 17:37:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I758aff1d3ef714f1c8ef82d29dd2217734a9aae6","subject":"infra VMs: use wait_for rather than wait_for_connection","status":"MERGED","created":"2021-10-28 13:20:38.000000000","updated":"2021-10-28 17:42:01.000000000","submitted":"2021-10-28 17:37:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815841","meta_rev_id":"4a09780f749a0e68cc3ad56462944d4debba3860","_number":815841,"virtual_id_number":815841,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-28 17:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-10-28 13:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-28 13:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-10-28 13:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 13:20:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-10-28 15:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b481fba491fb8ae118da6854564b2219ad563eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 13:20:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da1766897a4bb2a7f15bee69f598fe3440d46590","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-10-28 13:22:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9703be0dca6dfce3425cad279526e38aa67dd2d3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-28 13:30:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07fdb7ca685a23da7d6d8dc70077fcecce7f06b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 15:33:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc46b8ffe9a0489cae0bba01d7bb6df0 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68dd42a197bb4874b109fa86cecdb944 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b3c49c855c348e6a2c24feeb5910ee4 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c5bab191a642149f28eae412a80b37 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9dda6744187462d9a1de14f88f8d01d : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ee61a2500346a5bb6db95dc3a5a495 : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f3316d064e342ae91f3d67e48994205 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7d7cf7620cf45f3a3d2a8d4f92528a7 : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d084869f2114e999c9edee416d3d602 : SUCCESS in 35m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f3f8797f5c4d4635822accc8ada3d9fb : SUCCESS in 1h 19m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d59c80e9ac74484f8238ea772329803c : SUCCESS in 53m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/52686ca9ac2f4ab18722382df65684ce : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5fc25e6ac46e4ea5a66fe3e9a5feb848 : SUCCESS in 25m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c36a7b57493946e09851dc388797d62d : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/72a0817ab7f84133bee52ea6ae8f4e83 : SUCCESS in 1h 56m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00fee276b49a406b88b372391b44088c : SUCCESS in 1h 18m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5e32c7b20dbb497d82fabf0daeec89b1 : SUCCESS in 39m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2a7c644b88e41579cffa4212e0f99ef : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fda510a951e14321bf1349755fc12ebd : SUCCESS in 33m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dfbb5b932f244e8b9d5319954a93b86 : SUCCESS in 24m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/800c7190efd04127bb8062db9e38f8ef : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c1c22948adc4085b47dff9c271958a2 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"780d611def9acbb3d42edda440a729f78f9c7400","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 15:34:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1340a49e6df55f39fa53c380940baff13af0e283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:37:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa7bc7f84c647e48509cd40d37c029e : SUCCESS in 11m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f4a3aca3bdd4b179c6b86983e53de0a : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4e05b4523fd4a6ea52e4228bf7d5075 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246cb94673014c1898fe29e6af9c0914 : SUCCESS in 11m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/315bfc4195554efcb77920443d72802d : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/933e2eb5e4e64bd38a77f599c5c5bb94 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2645a6f4b30340bca1b49a79f3754fab : SUCCESS in 26m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d3b48422cdce431e8906598c807e10e3 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/117c5986eb4f4642bb2aee707b2f0aad : SUCCESS in 51m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/047aaaf97ae547c49fe7f8a2c4cfccd5 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f9d185f503b0440f84af19d43c5c6c3a : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d350dcbb67543e5a03081ae90d58906 : SUCCESS in 17m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/43cc14d14e67443da33ccf38ebbc87cd : SUCCESS in 1h 54m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0bd197da55ab4c80a09b27bf023a9ff9 : SUCCESS in 35m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eae5cb9dbf94a8fbcfc67557de55810 : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33cb9d07a00f422a8da6560421d2a3f0 : SUCCESS in 35m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47226c435ea54b95a2c42db86c7a9369 : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e2dcd78d9f4db1b1e70e8871869bba : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b3b741e4072a00458a7e892a3023d0dcb386e656","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:37:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a09780f749a0e68cc3ad56462944d4debba3860","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 17:42:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783f45bca7c3436784b899c062d6a8f1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a071757b2c2f3587903ebddfe5a52a8e8faf7a6","revisions":{"3a071757b2c2f3587903ebddfe5a52a8e8faf7a6":{"kind":"REWORK","_number":1,"created":"2021-10-28 13:20:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/815841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/815841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/815841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/815841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/815841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/815841/1"}}},"commit":{"parents":[{"commit":"3a7b54fd8fdcb3e352e4decfc1e75539c9032e1b","subject":"Merge \"Fix link syntax in release note\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7b54fd8fdcb3e352e4decfc1e75539c9032e1b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-08 15:52:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-10-28 13:20:32.000000000","tz":120},"subject":"infra VMs: use wait_for rather than wait_for_connection","message":"infra VMs: use wait_for rather than wait_for_connection\n\nwait_for_connection requires a python interpreter to be available, which\nmay not be the case in some images (CentOS Stream cloud images, cirros\netc.). Instead, use wait_for to wait for the SSH port to open, then rely\non the bootstrapping process to install an interpreter. This is the same\nmethod used for the seed VM.\n\nThis change also syncs the SSH wait timeout for infra VMs which the one\nused for seed VM provisioning (360 seconds).\n\nChange-Id: I758aff1d3ef714f1c8ef82d29dd2217734a9aae6\n(cherry picked from commit b87d89aae39cceedd4f70b447cddf5ada615bd36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a071757b2c2f3587903ebddfe5a52a8e8faf7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a071757b2c2f3587903ebddfe5a52a8e8faf7a6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
