)]}'
{"id":"openstack%2Fkayobe~815895","triplet_id":"openstack%2Fkayobe~master~Ief9adfcff28d269fc80a073cdde536e8e803ca37","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-01-06 17:02:12.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-06 13:41:25.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"Ief9adfcff28d269fc80a073cdde536e8e803ca37","subject":"Generate kolla config when running database commands","status":"MERGED","created":"2021-10-28 18:04:43.000000000","updated":"2022-01-06 17:05:43.000000000","submitted":"2022-01-06 17:02:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"815895","meta_rev_id":"5b95b69815e3b5c1e93d8e639b88840afef3c9ec","_number":815895,"virtual_id_number":815895,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 17:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-01-06 13:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-06 13:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-06 13:45:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 19:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-29 08:33:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-06 13:45:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"8657c0b30966461114a90330ef133b4ad714b4ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-28 18:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d7c34e143a0a8780834d51b0a4f566630d004ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 19:55:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97115077d04a420e880e1bf203edce21 : FAILURE in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a0f554221a489198e97ceff9a773ee : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03ad803f62b34114b1b352f44be32fa1 : FAILURE in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffa17f2809143fe93ad1f2378a21b24 : FAILURE in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddd478517b1c45a0b9b37f8a05bde779 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14ec4f9da5740e99e450377de2999cf : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de62646b20e0409aa52c4c5746ef94ee : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/614a8cab0bee45cf8fa20c0f5cadf516 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/986e2a6c799c40da869a47522827b023 : SUCCESS in 13m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b988b4a731bf41e0b2d65e4ab728da5e : SUCCESS in 28m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d2dc0ae313c747fb945f29769ccf3539 : SUCCESS in 1h 13m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56abbddcdfa04abfb55fe4bb1a6a781a : SUCCESS in 48m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/54756accf4a143b193972fd11c8365d6 : SUCCESS in 1h 17m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/54b3d274baa2410dbd6425f5d0adcb32 : SUCCESS in 22m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37c7370fa45e49898447bc0f52448615 : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a5c3e5522ff14fec873d4ca8373d83a5 : SUCCESS in 1h 45m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da49c31f1c1545b29b40840cadd9e588 : SUCCESS in 1h 21m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8e611503a61b46aea2c030339da085bf : SUCCESS in 34m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b60a06345344074b125f3dff99a77d7 : SUCCESS in 33m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a460d2756c56416a8ecbdfdc3d1efc44 : SUCCESS in 29m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/174c60464c65445889771bd2ddeaa76b : SUCCESS in 29m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81721253c9184fdea92a97d9ac05b766 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8be342d4d9348aa8e5846e48b4d4ffb : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8a9b3ed408572987828c4acb31aa7631846d0296","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-29 08:33:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71c0db2c9332947847b783ca24abe8809d6f033c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-29 08:44:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93ae5f3f63d1942996e24367005476e3bed7c3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-29 08:47:28.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"66441b2856be9edfeb8ed6769a7d7fe81dab5601","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-29 08:50:52.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"310a8dd5e95772d75d504bbfdb8073bb05c5accb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-29 09:49:20.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"497669c7af339664167cd21c7c387456abcd8995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-10-29 09:50:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"abc6478255ada111ec6861a45526dfa83e1deca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 11:40:31.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b377e29b024a467a8e223df2526bb586 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d02b5e1c7948f18cda0a9ba990c92c : FAILURE in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2150c1a9efc4299a80e7542e727beab : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2959df03df4bada5292aff71ce6aa3 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4012bd8161b14daabeb6bce68fc3f84a : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c265bfba454791af65c6d89e4603aa : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/208ec6cc9be446fc9dddb83d018badd6 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68f7eac7b27245d4a207c9c9dd59c612 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a455d67cb29f4d62ba3bd36afbfaf274 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4e64611474f34019931a3a1229421537 : SUCCESS in 26m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cc8dd1d3bd5a40d191f610e576a27085 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a31a725232624620bc0cd185860868e2 : SUCCESS in 49m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b5cfd313d6ee4c0ebb8481827e72917e : SUCCESS in 1h 14m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b9c9fe53d7f4afb8157d0b558ae1988 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62431de9aebb4b6f9d7b8b978faa38c4 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc01bd7293494aa3a558f82f0f1b90f4 : SUCCESS in 1h 45m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a5e0e40f37c4c8bbda54db20d13fadd : SUCCESS in 1h 16m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/befc7c4dbd194f8bbdb2e9f6c686b279 : SUCCESS in 37m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0371412b82f4984ab1a400975b90452 : SUCCESS in 30m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc017ea348c24e69a75137d68afbfe0a : SUCCESS in 32m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51c34d3e3f94eb6ba82db6fd05d2f0d : SUCCESS in 25m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4bd23562008e43aeb46be3bc68fdb3f5 : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2433a57669646be8989fb4202052099 : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":5},{"id":"f0edbd42691997eb9263105e779349128d5d7517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-01-04 15:56:55.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"1c6182cc07f86da89f51089cfbc2557727cf834a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-01-04 15:58:14.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fcedeb793bee59ee0341290764ef26539c8e4196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 17:56:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7132650f2f6f4b7ea4851c2fd121469a : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f6fcff76704ba38089b3b5106ab891 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16ce72c0334944f4acd853d0c3106608 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9439e46afaf740448423d425ed5632ac : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7169cab0fb3426eba5a23e53573663b : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01479f4597d34dcba26f9706eaf4ae66 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/657afe280c13449a9133466498a92abb : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/470799580b7749809f67e9daeff833b6 : SUCCESS in 14m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94984a4acde54d2fb1cce1df6721e6a7 : SUCCESS in 32m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/37de9c7c825b44bdab4d41ef044d84b6 : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d852c17ef8741ff97c89d7f3f2af4d6 : SUCCESS in 48m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0b64c8272eaa40dd905b0538fdb9f9b5 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e4c5e171e894470e85b5e12d3fcd484d : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa86cbfb6a514ce4b1cc660926082e7d : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c55e6e1cf7754f5fbb157aa54937c95c : SUCCESS in 1h 53m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca83c3a8d75e4af29e07aa78cac67ab4 : SUCCESS in 1h 13m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e115ae97a5dc42daa8afb878fd07124a : SUCCESS in 54m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e1d16a0221b44bd9da25879d12d42ff : SUCCESS in 46m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/43b5a63935fc4ba28335e1b126bbddad : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa8ed594635445d9ab3700f0e7deb17 : SUCCESS in 25m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d4bc397af9248ff84a0b20e0c856b25 : SUCCESS in 23m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d8536eefa8f498885c90ccf9da230f1 : SUCCESS in 13m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c33a3dcdcd0c4770aa0efdece5738a25 : SUCCESS in 18m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82f049e5ca5341ffbab83c8e4c1b7074 : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":7},{"id":"f3fa16309d8acd05350d1439df0a8246d6e0fec0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-06 13:41:25.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"5739e3b9f77a8d56c0f3bc88ad38e3bd97a139aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 13:45:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3629a66ff0c0837ac142ced19a9a6fac9bbf14a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:46:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c7ebed356572fe160118f7d64764e04fcf758993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:02:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be0e958a74f4e3d85622c3661cc98b2 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12ea498afeec4a53a763bc7813c0e371 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49ef162c5eea48aa901f73163e7d59e1 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009b328b2ea94081aa144912a74b19ae : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378d587c09ef46f4b0222e751c8d6652 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2661791a15e84a8c89ea02a568eb73ee : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57f23f357c8b4c2e85e5422c2da78428 : SUCCESS in 13m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8008d34f38c64039ac93987ac65e39d0 : SUCCESS in 32m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d18527940bf74e60843f8dc55e86d238 : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a626ac488084730bb1f2bd32008cc78 : SUCCESS in 50m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f973834985f44311bf94296726ffc9ed : SUCCESS in 1h 09m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4874a55d2de041e7ba891f0d27a439bf : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5405c322b4de4d6fbaa1ae833c7ebcaf : SUCCESS in 18m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/760eee2966714035a1544b496d7a1e69 : SUCCESS in 1h 58m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c31de2b0e740b98d91ec36f99e9830 : SUCCESS in 1h 13m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4d7c4cf2b1cd4fc4a5ba8ba578b51a05 : SUCCESS in 50m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9810feaaaee4d069352987f94ed7b1e : SUCCESS in 40m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8acc83c3fdd3488d892f9d57b6fee743 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b05eeb2bacc549f597ac20c47007f3a3 : SUCCESS in 22m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7343594e3cc64e9a8216212aea3af3d9 : SUCCESS in 19m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10427d75f694cd4960d4c29db48e2cf : SUCCESS in 12m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c0fd7c2f5b7410994a88fc1baca2a5f : SUCCESS in 19m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf7bc17c2914453581898afb484190e2 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":7},{"id":"3fe387bcb2891659b4f903ebfeac169c7beb831b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:02:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"5b95b69815e3b5c1e93d8e639b88840afef3c9ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:05:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c5bb0cf0dd476aab2e2e0eedec793e : SUCCESS in 2m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffbb732f0dfd4079a17bd336186215bb : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4321471a13eb949c2f5d57784557ddebb9b682f0","revisions":{"c83e6a8b78f32c3d5c9eb104e4c731d2fa1619e3":{"kind":"REWORK","_number":1,"created":"2021-10-28 18:04:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/1"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-10-28 18:04:38.000000000","tz":60},"subject":"Fix overcloud database backup","message":"Fix overcloud database backup\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c83e6a8b78f32c3d5c9eb104e4c731d2fa1619e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c83e6a8b78f32c3d5c9eb104e4c731d2fa1619e3"}]},"branch":"refs/heads/master"},"f575bbbe5a823ecabd2317f9b72f858e71203de3":{"kind":"REWORK","_number":2,"created":"2021-10-29 08:44:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/2"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-10-29 08:43:59.000000000","tz":60},"subject":"Fix overcloud database related commands when no kolla config","message":"Fix overcloud database related commands when no kolla config\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f575bbbe5a823ecabd2317f9b72f858e71203de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f575bbbe5a823ecabd2317f9b72f858e71203de3"}]},"branch":"refs/heads/master"},"db4a1fc87d18ed23e473002e3aed34a55a46a1e0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-29 08:47:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/3"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-10-29 08:45:11.000000000","tz":60},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db4a1fc87d18ed23e473002e3aed34a55a46a1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db4a1fc87d18ed23e473002e3aed34a55a46a1e0"}]},"branch":"refs/heads/master"},"75f314078d7a78d62eeb827b4aa44462237dd49e":{"kind":"REWORK","_number":4,"created":"2021-10-29 09:49:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/4"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-10-29 09:49:20.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f314078d7a78d62eeb827b4aa44462237dd49e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75f314078d7a78d62eeb827b4aa44462237dd49e"}]},"branch":"refs/heads/master"},"94695c0667505a29ac183203ecfc29934512ecb3":{"kind":"REWORK","_number":5,"created":"2021-10-29 09:50:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/5"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-10-29 09:50:31.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94695c0667505a29ac183203ecfc29934512ecb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94695c0667505a29ac183203ecfc29934512ecb3"}]},"branch":"refs/heads/master"},"ba4bc48f6820e09cc35144a7fc958dab70b965d5":{"kind":"REWORK","_number":6,"created":"2022-01-04 15:56:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/6"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-04 15:56:27.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba4bc48f6820e09cc35144a7fc958dab70b965d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba4bc48f6820e09cc35144a7fc958dab70b965d5"}]},"branch":"refs/heads/master"},"4321471a13eb949c2f5d57784557ddebb9b682f0":{"kind":"REWORK","_number":7,"created":"2022-01-04 15:58:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/95/815895/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/815895/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/815895/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/815895/7"}}},"commit":{"parents":[{"commit":"97a7dc6b229bbaad3e693cf4c2ffcc311058e031","subject":"Merge \"Fix --check argument for overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97a7dc6b229bbaad3e693cf4c2ffcc311058e031"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-04 15:58:14.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4321471a13eb949c2f5d57784557ddebb9b682f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4321471a13eb949c2f5d57784557ddebb9b682f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
