)]}'
{"id":"openstack%2Fkayobe~817282","triplet_id":"openstack%2Fkayobe~master~I84e336bec37eac12c659c7a9d9fc4ddd191a379d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-25 19:22:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84e336bec37eac12c659c7a9d9fc4ddd191a379d","subject":"[release] Set previous_release to xena","status":"MERGED","created":"2021-11-09 17:32:00.000000000","updated":"2021-11-25 19:31:40.000000000","submitted":"2021-11-25 19:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817282","meta_rev_id":"11087d2c548ab9b054677a2f4777e09687a161a6","_number":817282,"virtual_id_number":817282,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 19:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-25 17:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-09 20:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-11-10 14:10:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-11-25 17:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-09 19:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 20:22:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-11-10 14:10:38.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-11-25 17:11:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a208451f81ca2fe56a5847725386aa6f32782cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-09 17:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c47cec763b046eb12d6c71fdb42f8a66b806a1cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cb09db1698146cea62ac0cbe9a81b2a : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb92278262b4ab5a951215f90bd9356 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/603fd559a0214844857457511c11c5a9 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0388f3d4e59427f91319f4b3b97c453 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33eb14de1b2e480c84d3850ad08efaa4 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb59752607140bf9d7e75f15d71f89f : SUCCESS in 10m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef6571f1759a40d3af99d04eb14feef0 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d572dcc0bffc48d3b80093793dd3d28a : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ceb7d7474c140b0b1b989554e812441 : SUCCESS in 33m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3910d4289ce04035b5b1fc6a03209622 : SUCCESS in 1h 06m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b70c736b69544d238b10d7da6d34eb9e : SUCCESS in 41m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f10f8d67b6904a8fb06bcc5a9252baa1 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/06472313874045b2bb65a9b3f0ede3e9 : SUCCESS in 23m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4447986994374c2ca8053225887ab56a : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44deb20186e4455bb02c7c654de8c4e3 : SUCCESS in 1h 43m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/363b4e04616e42c1a2ab0aa9c7f74412 : SUCCESS in 1h 15m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d157483c95964c1bbdc76afb77cf67b0 : SUCCESS in 40m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521f27d5038a4c7297461199d874883b : SUCCESS in 34m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/828cf380c92341ad973c76a388d9ce7a : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f77fd18f09904fba8753c2840bcea16e : SUCCESS in 24m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0aa35e5fff7a4fc39ece3b97820c7ba9 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ece2870d36844cd4abafeb146a7f7ceb : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"65f8c3a51b43b4e2e21cf95816608a9d9aa7790e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-09 20:22:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64ff9a5a9d95b0881291e42974469cca1ea3df55","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-11-10 14:10:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b07105ad17fcb0c9f9a803e951b72ea6531cf185","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-11 08:54:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d4d8fa599e82941612a4db013bbbce5b4a4309d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-11 08:55:12.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"af6f15270a87ff0e7e7c39a4e39fd78108677a20","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:11:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e957a5d4d995d7deda6ebe218fd97b452eb1b6a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 17:16:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f947f6b3927ad270a58d3f9118e8171863a513bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:22:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1f7c52be6c4449b0cef17a181ec9a7 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96ec3f2bd710498dbfcc2205e1ab166c : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/397b377e8c24426982e398515eecee0a : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da7bc336770d4b0aaa67399ded880b59 : SUCCESS in 8m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b456e6a6141e4ca68e09a897f0c40aa0 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97579dc827474a5db2b2b3fa6f82ad0f : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c39e07dbda804f2c8a5bcb135be69333 : SUCCESS in 33m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a78551f9fe39496caf818c33bdb2226b : SUCCESS in 54m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95795416578b46bf8a35241d7836ea2f : SUCCESS in 47m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/abe0f055c0cf48068baa6a31d65fd892 : SUCCESS in 55m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ba8b6dd2d65f4fce92461b07fdc25f34 : SUCCESS in 23m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dab17b8f191643f0bf87801ce95fc895 : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a4adc6b6e1be4a629e973abe3d4596e6 : SUCCESS in 2h 02m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0163a4b4969427eaf9603c96d703cc2 : SUCCESS in 1h 17m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3a2568c0fe5044fcb5fe8509ca936aa6 : SUCCESS in 36m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f1a3c5b01374dbfbeaaab4b3179c7f0 : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7026c5747c224c07adaa16eff89504e1 : SUCCESS in 39m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e23a07d7225649999a410c7684bca41e : SUCCESS in 27m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8769635c98b94f139d1e244101aaf3c7 : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba1066870bf24ec28d506617b67b0f43 : SUCCESS in 13m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c81dcc54f5f549cfa88b7c5a5b98b4dd : SUCCESS in 18m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894be2d4b27549329f702100e6be8d7c : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c77801657991fe4bd87e481cc8fc01ef4d2ac606","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:22:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11087d2c548ab9b054677a2f4777e09687a161a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfb561bfff44e2199c4d7494f985d43 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf7cdf1e35a0ac963e35ba28a4b1d28127d3c21","revisions":{"4bf7cdf1e35a0ac963e35ba28a4b1d28127d3c21":{"kind":"REWORK","_number":1,"created":"2021-11-09 17:32:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/817282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/817282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/817282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/817282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/817282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/817282/1"}}},"commit":{"parents":[{"commit":"7f72fdbe5510ed087e91b91651d2d70ba6ecb8a9","subject":"Revert \"Use OpenStack Xena release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f72fdbe5510ed087e91b91651d2d70ba6ecb8a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-09 17:31:06.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-09 17:31:53.000000000","tz":0},"subject":"[release] Set previous_release to xena","message":"[release] Set previous_release to xena\n\nAlso remove some Wallaby upgrade compatibility code in CI playbooks\n\nChange-Id: I84e336bec37eac12c659c7a9d9fc4ddd191a379d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf7cdf1e35a0ac963e35ba28a4b1d28127d3c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf7cdf1e35a0ac963e35ba28a4b1d28127d3c21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
