)]}'
{"id":"openstack%2Fkayobe~817673","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I7dbd643fec72223bfda20cbbc37809ae78073804","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-17 17:36:02.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-11-17 17:36:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7dbd643fec72223bfda20cbbc37809ae78073804","subject":"CI: add Infra VM jobs","status":"MERGED","created":"2021-11-12 08:37:15.000000000","updated":"2021-11-17 17:46:02.000000000","submitted":"2021-11-17 17:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817673","cherry_pick_of_change":813048,"cherry_pick_of_patch_set":5,"meta_rev_id":"82240b098d12187e771f51ce6bd246498b0b9faf","_number":817673,"virtual_id_number":817673,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-17 17:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-11-17 09:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-11-12 14:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-11-17 09:45:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 08:37:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-12 08:58:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-17 09:45:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d8669757f46d3d05b863b986e408a23152cf10e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-12 08:37:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"81ea4bf9e77eb162b254c47e6fa5db7974e22bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 08:58:07.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"cf854b4df8236819b1e553862083a461efda59e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-12 08:58:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c80ff74625a72365c1037aee3bcf2d898093fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 11:06:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c36ee32f644544ab9cf3179615af9d46 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efda2976579249baa04d48b1b8f09924 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d060d4cffe34cccb34c56969eba1855 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc78fce2d55d44d4ab613ae89ff97815 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f069d24fb9d84c74bb5e51cfefb2a9b8 : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5592d0be184b4d1c9c994aede8c4b145 : SUCCESS in 11m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce5aafb38bb94a7e8a65b719acd02e4e : SUCCESS in 10m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c30cc93c29f484a989e25d4cdab380e : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6901b7956d6647e29291fa5b41ef733a : SUCCESS in 31m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2675bbee54ef4a809dd57d07c61abe45 : SUCCESS in 1h 15m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064dbf0f54f74e59af2952328449c483 : SUCCESS in 45m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/906f62a98ac649c6bc4d3be9071f0fac : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2a2f3e771a254cfa9bc9a672f4946715 : SUCCESS in 23m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59a1d0a383eb45f1846f4fbe950f4c38 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d2d0edcb7b474687a33c7a91c54fb167 : SUCCESS in 2h 01m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b7ccf15df9b4d4db941d65cb6002a99 : SUCCESS in 1h 31m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d25fe2f44e154cf3ab74d0c4e11a567e : SUCCESS in 42m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b43daa968bb148b38ec1536f8ec15f04 : SUCCESS in 38m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6aecf760b6f54df6aa35b0b7f8be8250 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23dd7e8344c441c1b40360f595162231 : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f8e53ab3835244fc88f1b8c9a07a9355 : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d3a1b3d230d4c30b4167bb86e50e4f3 : SUCCESS in 13m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/898b846e36fa4164add63f278e0fc262 : SUCCESS in 18m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a39d0c32745c4904bb30d08c84327c74 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ef91ded993e76afbda226fdd0feff51efe6340cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-12 14:01:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a9b69badd3fe28ba00b9a82b9f59f0ee1823a6ee","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-11-17 09:45:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e81832754fc20e9e2683adff6a4f7b82dec4a21b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 15:22:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ffdb253ad6eb4cf84efb76f629b0a644debb0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:36:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad4f544438148459af3440bcd0b4f5e : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad3892806f3240eeafdea3b8136b40e2 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f2d9bdd8f084d3abbaac99bf1d886a8 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49d922cf76749a196875fdde27a0839 : SUCCESS in 9m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/630da0f4913e42ebbb3cc706a5225b03 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/771f7a5ced974c79a0928ac089a65075 : SUCCESS in 15m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/699dc2b711804632864ecc38b9d431ee : SUCCESS in 41m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d309a843f0854474a424073fdb56a3fc : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded89f1c939d4a468be30dc10fd55c0c : SUCCESS in 52m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/507d70d1a2ba4efb85ff16e3817b5831 : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f0b3741ee5424bdcaf7f56d4fccdee0c : SUCCESS in 26m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/008e0a5fdfa04f0c8a005fb29995f2e0 : SUCCESS in 16m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c777bf9778234ccfa759b75508e10e02 : SUCCESS in 1h 43m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/612a1be375c945c2b60d4c1666564b8e : SUCCESS in 1h 22m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5c4258b8f10b4c4a848409a5a417c04d : SUCCESS in 1h 14m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a46438f789744cc08a5eb4cd1460c8f2 : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16515c9de0d14620b3363f04e3f8f164 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a87cad9ad8e4583bd580c160a7c4f90 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/14ec7483d18d4f86b97ffa0c6fecdb99 : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc82d766f72148a88fdc174386525308 : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e7ab5e79d9ae4ba7917ee631786a4649 : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e40b260093da42f9a664c5003e1e472d : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0a34fabe69ff4dbf2cb65ae63c8146ae8ddbee68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:36:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"82240b098d12187e771f51ce6bd246498b0b9faf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:46:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8213ffc81c6045e58836c491832e4645 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dbf0ba524274f1714faf97536d89346934cc72c","revisions":{"62780fb415b7f89766be29ea6d46e3dc2c5428f4":{"kind":"REWORK","_number":1,"created":"2021-11-12 08:37:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/817673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/817673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/817673/1"}}},"commit":{"parents":[{"commit":"98e7817ea7815efea2a828c581020147c59184e9","subject":"Ubuntu: add upgrade jobs to gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98e7817ea7815efea2a828c581020147c59184e9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-07 15:51:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-12 08:37:15.000000000","tz":0},"subject":"CI: add Infra VM jobs","message":"CI: add Infra VM jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/812670\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/817672\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/813212\n\nChange-Id: I7dbd643fec72223bfda20cbbc37809ae78073804\n(cherry picked from commit 104e383fc7b2bdd3f0b26036d169afd63313db24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62780fb415b7f89766be29ea6d46e3dc2c5428f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62780fb415b7f89766be29ea6d46e3dc2c5428f4"}]},"branch":"refs/heads/stable/xena"},"1dbf0ba524274f1714faf97536d89346934cc72c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-12 08:58:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/817673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/817673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/817673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/817673/2"}}},"commit":{"parents":[{"commit":"98e7817ea7815efea2a828c581020147c59184e9","subject":"Ubuntu: add upgrade jobs to gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98e7817ea7815efea2a828c581020147c59184e9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-07 15:51:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-12 08:58:35.000000000","tz":0},"subject":"CI: add Infra VM jobs","message":"CI: add Infra VM jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/817672\n\nChange-Id: I7dbd643fec72223bfda20cbbc37809ae78073804\n(cherry picked from commit 104e383fc7b2bdd3f0b26036d169afd63313db24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dbf0ba524274f1714faf97536d89346934cc72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dbf0ba524274f1714faf97536d89346934cc72c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
