)]}'
{"id":"openstack%2Fkayobe~817964","triplet_id":"openstack%2Fkayobe~master~Ifea5c7e73f6f410f96a7398bfd349d1f631d9fc0","project":"openstack/kayobe","branch":"master","topic":"ansible-path-envs","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-09-08 14:48:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifea5c7e73f6f410f96a7398bfd349d1f631d9fc0","subject":"Move group_vars to an inventory directory","status":"MERGED","created":"2021-11-15 15:38:08.000000000","updated":"2022-09-08 14:49:31.000000000","submitted":"2022-09-08 14:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817964-ansible-path-envs","meta_rev_id":"4c4ab95cb04ca7c59ecc352db478fed3af7a659b","_number":817964,"virtual_id_number":817964,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 14:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-09-08 12:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-09-08 09:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2022-09-08 12:43:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-15 17:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-08 09:16:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-08 09:24:07.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-09-08 12:43:10.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"42361311d4a3620d090805eef06096dab289b2e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-15 15:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb5ab179c546e729d8583cd9a6fc921135293bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 17:52:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fad368f538f4c97b1562fb277ffafd4 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994ec35585c34ed38c190becdb28ae63 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45d3e63b52394f0b8e401a6bd3022704 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6db737b0601407aaa4047a192dd81ca : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fa10c09dacb41e39e779cabb0606b63 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52a1be10a7b424bb06187f97e340ea5 : FAILURE in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e525abd3ecc340b09ebeecb7a98631bb : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60361f97ee8d4175a7fe9a4db6c5ddbd : FAILURE in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98c06974910b416abda5e380989b4ad3 : SUCCESS in 31m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6ef1d48074a04be081a61ac3f3404349 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c4c5d0345f5497fb23ea7102829b523 : SUCCESS in 48m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c18db43e40bb4d9c956814a9dad9ea6e : SUCCESS in 1h 14m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f12272d25bee41d6b8b0f0d77953fdb2 : SUCCESS in 26m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7084a83fc8e495181f3443a59a946da : SUCCESS in 18m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9cde858652a94197825ed0d7df386238 : SUCCESS in 2h 00m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3793df0da27e4775882efcf5aad52d67 : SUCCESS in 1h 23m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/818be76449774674bf106a2518d5a7d5 : SUCCESS in 38m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb78d47e93c649f3bab756ab193618dc : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65dbf16e775143f8b101474f12ec14ed : SUCCESS in 33m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f11d3771169465494e366b9cfd3921f : SUCCESS in 21m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/27d64998b7404672a1800beefef56f90 : SUCCESS in 16m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df1558b57c5b47f29ebebd1c5bef0e35 : SUCCESS in 14m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e5e98deb0f742ccab06876370524389 : SUCCESS in 21m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03b7d89c67194bbbb345499491511720 : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9735acf958b1302c7d963aa3308a1d2a153b9e69","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-08 08:25:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8794c13f558b0d3e1321f735ec98e99f32991343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-08 09:12:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f93eeab1246f85b7b6fe3f48b78a7fc3659ae52c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-08 09:12:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"2d50ba5f13a34ac3480b268b443c9038aea92389","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-08 09:16:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c65450c668c48ad5a29ebc5827190d9bc47a6e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:18:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c14e6c8fb254b1f820605d64610998a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a8b52d2224d4d378b4947c01e28a3a4 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3629b817f2744131bea98cb0bae7ab73 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af59163dc0774fe5ae80e2352b923de0 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd35015f16114ebe9ca35e2a3a034505 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3264936567db468c867377940f2e08ce : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07890c3149d4461996a304e5dfc83dce : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc8a257e1044cdfac44b699f01c132a : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14acf31936614087b44eb7badc342d57 : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fab4f28cfe84a6d99373b9f9e7d57b3 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e164174aec240729b81ed7d77edeeae : SUCCESS in 17m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2f3540a557024d27821fde57a992621c : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/32f0b531d6834ba6aca195f2d2b9c0a2 : SUCCESS in 1h 15m 12s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/614df2931f234498b8df1e14ef85700d : SUCCESS in 55m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3d3e9467027849bc9e3a116792a16e7a : SUCCESS in 1h 07m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bfe49b38957544eaaaf04aeec79279d1 : SUCCESS in 25m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/38fa0bf5f678475584ed13a98b640015 : SUCCESS in 23m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b54463bfbe154f35937d0f8783718fae : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bae97e3098ef445a84da1a44be35b502 : SUCCESS in 2h 01m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf1269c29a3a44d8ba3c9cd3062c1f32 : SUCCESS in 1h 11m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c8ee7947bc16418b94a4d787fa4a40ff : SUCCESS in 22m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/fef91120a20f41c5961f71faf0869a9e : SUCCESS in 22m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f58a0c7b0554f9a9d8597408cae426c : SUCCESS in 16m 15s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/10e23ee98bfa4e4a88b699e1feff4c9a : FAILURE in 31m 55s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ae683ea4c0c47f1a0b3fb819e7cdd8d : SUCCESS in 43m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/600f9b10be5445a3a647d06e23994a4f : SUCCESS in 27m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/58882d2cc9e0404fb7c30128ef9adaf7 : SUCCESS in 18m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/43a03b8f7f4b49e7b0d0a2790d370f38 : SUCCESS in 18m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c3af98a42042f795ba3710ce9bb8dc : SUCCESS in 13m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cacb15a522634f5a9d93d7c807bd3493 : SUCCESS in 21m 28s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8628b230e911410caad568857f8a617c : SUCCESS in 23m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57553786bc584171a2870ac35b76c59f : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":3},{"id":"cdbcd6aa4a6bacea90a7e594a2502af4a4f1c6fb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-08 12:43:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ca4244b28046fdc055d4fdd72c35258870e86cfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 12:43:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dacf9137d7f6600f64880b7530b2ec7d56ba61ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 14:48:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f010c8e0d47b4e0c9fa77cad5977acda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23b2b068707d44269b8ba1dcc52a0c24 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86093fc1c7a44cb6acf2e85b1b0d804c : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be784326dc3b4b4f8934677c5b2f0505 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc22f747916459b9706e3cdd368c5d1 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95fca8a473f5433d9e5ed569a77284aa : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5357bf029f5a4fbf9575760508d6284e : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bbec001e33543d8a21d61d94fdc5225 : SUCCESS in 16m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90e7baf7ba0b46d8a5638f3ef5431a63 : SUCCESS in 19m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79caaf0e93424c1d94314e780a22a1f7 : SUCCESS in 1h 17m 57s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/969db29ca5a9432383f1e1a55e2486db : SUCCESS in 1h 12m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5236198ee5c4898b517e4a923fc5e6b : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b368d05d8da94cfe9434b23811284022 : SUCCESS in 1h 12m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b362024efee248f1a7c90f0133481c4c : SUCCESS in 23m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/becb7ba6937a4cbf9ad10771c5d575d4 : SUCCESS in 24m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea8f3c3a27da425f8b9f7c553be96ddd : SUCCESS in 20m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ed82751117a48c1a33fa46ccd0ae60a : SUCCESS in 1h 59m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ddc4be05b54e5792a89563f59b37e5 : SUCCESS in 1h 02m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/67200a5de6c54f74acea359dc63c776c : SUCCESS in 20m 25s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e21d54381d914c9f9048abb3bf3d73c1 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3aa5e982aa94503b55ae882322d4617 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8fef55f5f6a74a2287e7570a2f012616 : SUCCESS in 36m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfc8f79a83ff4b88b8b9f8b1ef795187 : SUCCESS in 26m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5eaf7867fb2049f782c98d2c564708bf : SUCCESS in 16m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/35f9b09fa9764afca296283275a9fb62 : SUCCESS in 17m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8952d00155864d91b31229aec6e956b4 : SUCCESS in 11m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/396d71fe5b054d6e98ccea522221e48a : SUCCESS in 18m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2b3d0f63a0064abca6217cb2aa1366e6 : SUCCESS in 20m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddc712b18124471e85349e0da9097e26 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a04ff16c7a1cba011629bc3379beb23943503f9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 14:48:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4c4ab95cb04ca7c59ecc352db478fed3af7a659b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 14:49:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7666fc77ba4407b2bc4bcd1fe81ce8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87b50af598e42e7a29492c6810b35c0 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ad8feb4a0a746598cca6937465cd39b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"af3b719b73cb1805c10cf6d6343e7eded63dc186","revisions":{"05941058ee261261d48ff73ff8af8a2c63bd097f":{"kind":"REWORK","_number":1,"created":"2021-11-15 15:38:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/817964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/817964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/817964/1"}}},"commit":{"parents":[{"commit":"8df7b75c06b32a21ea313aa42a63e7def6305ba8","subject":"Ubuntu: add upgrade jobs to gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8df7b75c06b32a21ea313aa42a63e7def6305ba8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-12 20:46:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-12 20:54:45.000000000","tz":0},"subject":"Move group_vars to an inventory directory","message":"Move group_vars to an inventory directory\n\nChange-Id: Ifea5c7e73f6f410f96a7398bfd349d1f631d9fc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05941058ee261261d48ff73ff8af8a2c63bd097f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05941058ee261261d48ff73ff8af8a2c63bd097f"}]},"branch":"refs/heads/master"},"b4b22903578f3d7bc1fd7dde1969382c4f0c52f5":{"kind":"REWORK","_number":2,"created":"2022-09-08 08:25:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/817964/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/817964/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/817964/2"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-12 20:46:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-09-08 08:13:18.000000000","tz":60},"subject":"Move group_vars to an inventory directory","message":"Move group_vars to an inventory directory\n\nThe group variables originally in ansible/group_vars/ were playbook\ngroup variables, due to being adjacent to the playbooks. Typically they\nprovided default values for global variables in the all group, as well\nas some more specific groups. This has worked fairly well, but results\nin (at least) a couple of problems.\n\n1. The default variable precedence rules mean that these playbook group\n   variables have a higher precedence than inventory group variables\n   (for a given group). This can make it challenging to override\n   playbook group variables in the inventory in Kayobe configuration.\n\n2. Any playbook run by Kayobe must be in the same directory as the\n   playbook group variables in order to use them. Given that they\n   include variables required for connectivity such as ansible_host and\n   ansible_user, this is quite critical. For Kayobe custom playbooks, we\n   work around this by symlinking to the group_vars directory from the\n   directory containing the custom playbook. This is not an elegant\n   workaround, and has assumptions about the relative paths of the\n   Kayobe configuration and virtual environment in which Kayobe is\n   installed.\n\nChange-Id: Ifea5c7e73f6f410f96a7398bfd349d1f631d9fc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b22903578f3d7bc1fd7dde1969382c4f0c52f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4b22903578f3d7bc1fd7dde1969382c4f0c52f5"}]},"branch":"refs/heads/master"},"af3b719b73cb1805c10cf6d6343e7eded63dc186":{"kind":"REWORK","_number":3,"created":"2022-09-08 09:12:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/817964/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/817964/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/817964/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/817964/3"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-12 20:46:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-09-08 09:11:23.000000000","tz":60},"subject":"Move group_vars to an inventory directory","message":"Move group_vars to an inventory directory\n\nThe group variables originally in ansible/group_vars/ were playbook\ngroup variables, due to being adjacent to the playbooks. Typically they\nprovided default values for global variables in the all group, as well\nas some more specific groups. This has worked fairly well, but results\nin (at least) a couple of problems.\n\n1. The default variable precedence rules mean that these playbook group\n   variables have a higher precedence than inventory group variables\n   (for a given group). This can make it challenging to override\n   playbook group variables in the inventory in Kayobe configuration.\n\n2. Any playbook run by Kayobe must be in the same directory as the\n   playbook group variables in order to use them. Given that they\n   include variables required for connectivity such as ansible_host and\n   ansible_user, this is quite critical. For Kayobe custom playbooks, we\n   work around this by symlinking to the group_vars directory from the\n   directory containing the custom playbook. This is not an elegant\n   workaround, and has assumptions about the relative paths of the\n   Kayobe configuration and virtual environment in which Kayobe is\n   installed.\n\nStory: 2010280\nTask: 46233\n\nChange-Id: Ifea5c7e73f6f410f96a7398bfd349d1f631d9fc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af3b719b73cb1805c10cf6d6343e7eded63dc186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af3b719b73cb1805c10cf6d6343e7eded63dc186"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
