)]}'
{"id":"openstack%2Fkayobe~818075","triplet_id":"openstack%2Fkayobe~master~I879df0ea3f357c2bb45a5e7331a3f2569eac63e9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-22 09:11:38.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I879df0ea3f357c2bb45a5e7331a3f2569eac63e9","subject":"Ubuntu: support host package update","status":"MERGED","created":"2021-11-16 12:17:27.000000000","updated":"2022-03-22 13:32:51.000000000","submitted":"2022-03-22 13:31:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818075","meta_rev_id":"86c37d9c15ee4c79d6b7da26af95dd3f1416f623","_number":818075,"virtual_id_number":818075,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 13:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-22 13:32:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-22 13:32:51.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-22 13:32:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-03-22 09:11:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 14:16:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 17:10:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-27 18:45:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-01-06 13:42:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"435cf75b835b0688212a5d39503bc62012da03d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-16 12:17:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7466c623c7a890f790108f5a28f45dba48b4545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 14:16:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ac4c1768cd46809cbaf0580d19ebef : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42552c75eb1145d1a0d57d54ee26edd8 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b09a4c7deb5c4360975c589b6f0bee86 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adff2463251849de9b8c3fcac307c2bd : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a46adc87580454796c9f8954177568a : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5c001bf69b4fb2af7407d58976e787 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/529ab46b1d35452196dfbe2b370b6fb7 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60c0b381ac9d4c6e9d109f69d4f6580d : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f755c43af6884b37ad3e93a7b2b7046e : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39231eb2ac6a4eada14a20d48b9ac87d : SUCCESS in 31m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9e889a4a47514fc58903e2472fdf132a : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a3cef816fc543c0a6b8acf56e860145 : SUCCESS in 51m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f87b38af20324021b2677d1a13c219ad : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0c9a8b196bd24658af68b5504dbcb10a : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c241606aa7cf44eabf6cf973321ae856 : FAILURE in 15m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5aff3f6678d94d408d70168c06883fc1 : SUCCESS in 1h 55m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e131ece2da704465b2432a3975ba4308 : SUCCESS in 1h 11m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7b4b2bf2880341c4a128baaf5a52b4fa : SUCCESS in 41m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a32dd737fc84ed9a3bf7ebb735d96df : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c3a96fc7eb53462fb9afb8ae637718e3 : SUCCESS in 30m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ffc7366785d4e3fb09597eb7b0e3a5e : SUCCESS in 30m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc0c62423028478d9c4b741e25a9d3d0 : SUCCESS in 16m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8acda450acba480f81307e078aaa5585 : SUCCESS in 13m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/122710c040bf4469970e489dd9811ab1 : SUCCESS in 19m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07ef64cc77d44794b2883d6858ad3d81 : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ebbef34b86f1dbc6d0425d3b84e432b60400ddba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-16 14:47:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e0454e5bfdbd7148233a2d28fdc00b530909f0ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 16:55:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7cd3eca3c304ff0b415af5f1dbc9fab : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f071f7bf12d54d4c823a563b0145889b : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e36a8ff0c3d143969d5a222f4e6b188c : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24402ae88bb40bfa9ac1842c800e913 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baed09bbcbaf4542914dc9194b577e6e : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76f84ab692145b9aadb5b4c9b6ed5f2 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484c6382efe142e2b85956ada19829b0 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb136e7679f04b5cbbc7f22a7a774bd7 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee900056a7f54096a2a71ad3f450b249 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e9fa18aa1f549e88d3ea0234e8d6815 : SUCCESS in 28m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/223bfb64e1fb468f9e7849ea5fc6e025 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9188046af1844b9b94e17b0225f652de : SUCCESS in 50m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fbf5e64cac65436e980adaacd47c076b : SUCCESS in 1h 15m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ca73bacfe29b4a3db39a0ef4c15ab117 : SUCCESS in 26m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/248491b4ab2b4022a8098709bb4d726f : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1747f389057446a5ac541b13265626b2 : SUCCESS in 2h 01m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddc2d90eb03842ccb5754537503c2721 : SUCCESS in 1h 18m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b539ed2ed43b498bab44022f5b2ea6d1 : SUCCESS in 39m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05d344df49454fe88f6a9ee11edab54b : SUCCESS in 43m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f61570e9ce2547f18bae571cd253d03b : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3b2f21d3e2f4501a2f89efca2b8ac29 : SUCCESS in 23m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fed4668d232945288710cddb07ad6996 : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc0df56267e14cccba32de99f963484f : SUCCESS in 12m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a8802cc734a4f3287f0e037ccaa1b20 : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae2615454329461facc5b807c4fc124c : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a06aa3297aa2e9cf84f61e0ce3ba87656849c9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-23 10:55:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f3359f480f2bc2a747d7619529a420500ea17dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:52:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/599ca6093bdd4ff59e04b6e65f7b260e : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937340ec3fc04e378f29849ec336d744 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c0fce0abd52424b931b41aa685efe00 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba031ed64c94b7cbcf24ab749103d3c : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/664bdc92dc9948a2bf476953460146bb : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b12cb4484ae42e89ac4e287111db90d : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/284b5fef2e4942a8972834a9b6794a7f : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d20ed370aa6c47dab64f094aceb708df : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b6973e182a99421792697de5d8818713 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c89370495f3f4a81a3fe4dfa2c80c2f5 : SUCCESS in 33m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8e32d8e2a523413c8b997aa10f959582 : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01bf0b5ed0654ce589d4daddac744455 : SUCCESS in 47m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f60aed4db2c467b8d01142b05e0dc87 : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9f181531c8384412aec64863e571ad1c : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a90ad62aa9924ef8b9ad289304eaa69c : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa52878b475d4192a78595fb62793e03 : SUCCESS in 1h 54m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37dee81929384bc99e250666be7ab394 : SUCCESS in 1h 20m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7543acf6f2db469298dce12c6bc480cb : SUCCESS in 53m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac94986f5d04d3babf8574e95628ba4 : SUCCESS in 35m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98449318516f4a65ab393c27cabd7efd : SUCCESS in 27m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fbfc72fb0b24bbe9bf334ece5bd7cb6 : SUCCESS in 25m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b9d5eb304d2e4242a4204e1919aff2c5 : SUCCESS in 28m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f97cc6768a4fa2b875a6d163283174 : SUCCESS in 12m 44s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/21cdca9112f64431ae6dbb6d6f7b09ba : SUCCESS in 26m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abec7b17b5854c00b4c243c5f89caea1 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4305c0df881d52d814b802979de9cf61452a27d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:10:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3ef5c87f4ee7f51bb6b4c0b797b2474ed56c82bd","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:10:27.000000000","message":"Removed Workflow+1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"16a9b7a067a0ba76b12a27b5fa2109c68836d489","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-27 18:45:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f3c8ac4d02bfaef38a963d582e6e15fcc48bde54","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 13:42:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"699dcb624d6affd5f7fde047d18870a234e6c950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-11 12:38:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3df619c20a9b50fd8903da6b942800ae73e197cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:31:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79bd08e84da4402dbc460803eda42c69 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07357bc7195a4390a28ed3d48463e484 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca377872cbf41b9b597ce928f8a8050 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48d9b2e63f15474ba662fdf32d98f0f5 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614ea4c370e747da9fb05b855758d612 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3250447112214c3d989262f3c2daec7a : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28c00934f9594271a5abf494fafeadc4 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a962784bc80b4695bf3dd014c9cc61e2 : SUCCESS in 14m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2a739463ff64643b0790ce77a72df23 : SUCCESS in 31m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1a469387e84a4c528b1b7fc2c066a9a8 : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/041f760d4d75496d962e3e0d952b01f2 : SUCCESS in 54m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/08f2c002be374cfbbda51b46a0a060d0 : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7910b18591d4f609647d113b082a779 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/423725d54b794fdda2b19f0c59dfe8a0 : SUCCESS in 1h 47m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dc130c7ceb043719a88299f3ba33650 : SUCCESS in 1h 14m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/752e0a9106f441c29020e2f36e837b15 : SUCCESS in 47m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b29f145551634b1091a3cb972c5d599e : SUCCESS in 38m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1535aa6d94934618bbb8e1637d1d0149 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/905457c77f8341179df2919fdc7d369e : SUCCESS in 28m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f099c3cc4eec4735b274b2cca267e3fb : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f16c3a8f8e74783a346186a6c7f4d1c : SUCCESS in 12m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/42787aa48daa40209ddf31a1db5a27fb : SUCCESS in 18m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7bea8a9197f4c36a49c5b0897c591f6 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":4},{"id":"b3e39c2b931c81b7dd7353609da379c0fdf8ffc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 20:47:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"cd84ed9d2bdbf8826db1ce2471f47607a5ba888f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 22:44:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819693025da4497b85e1e18f2238cf63 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8feb3315f4a541efa67899ba63851ac3 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2cf338e4e04bc6a994352c7aa7e9f5 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db5ed4af81bf45cf9c3050a485a205a8 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae72648804d04bd087fd757181604787 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4df2024666f14cb2b1ac4bac37cb3adb : SUCCESS in 5m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4899648e2fcf4bfab623501511060b5b : SUCCESS in 8m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8831f2726c314a808b1810ce44816844 : SUCCESS in 13m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06dc6ce56bb14ff79e9badab694545d9 : SUCCESS in 34m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b431700b8f6343d4adeba98dac964ed6 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/df021bee055a47c2a405964a9cf15f57 : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d313f3fd68a645358e21825ad38502ff : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4679607d12444411a7d63f71ed25183d : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/68290a965ef644d3b6c0c69d04bba9ef : SUCCESS in 25m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2461a51acff14d90a0f2158889ef6402 : SUCCESS in 31m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e37ca2471404305b073099a5032db3a : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/73004eebeab6479381cdf8c77146a561 : SUCCESS in 1h 55m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70d537e8a78a4fbb931e530859ca8cef : SUCCESS in 1h 19m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb0b135d28a84597b4291ba4b20da2fd : SUCCESS in 47m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/78a6e48521c74f869f404099a9b735d5 : SUCCESS in 49m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95a7989bd82d462e97bc1d69c8c300fa : SUCCESS in 34m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e2d762e3ce7406cb9fa9749f5e60f96 : SUCCESS in 44m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df30a33dffb40ae93acf4373c226963 : SUCCESS in 35m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5fc177e0e761408dbe8744eb701dfa3d : SUCCESS in 16m 42s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3e9ad6a9b5974a5c885b4f819f057ecf : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff0ac67289c44a8aa189f17ecdd0d7bc : SUCCESS in 13m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9dc3aea1b92b40ccb5fbdddae1fb696c : SUCCESS in 22m 39s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ee043bf4d0ab4d0d87ec2404922f4435 : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a624a3ce1f4240d3b6c63061476b6f83 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":5},{"id":"79a6ac9bc29b26b647ce70bfacadc55ceb89e492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-22 09:11:28.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8f620948275769ec8fafb3423cabab44e5ed2444","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-22 09:11:38.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6997929e74fa67b16891c7b6411ea2cda41ae7a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:02:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf38aad248e498ebd275af451995758 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fcb088c9bc44b63a1f0944b2cb2532c : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b0953297f2b42a696b5e3ce9ca8241d : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c66ff9d759944df6888d3fc1c950ff81 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1edd05b3c6f4800b2cf5ea9ccf52099 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d873be64a2d24ac4a80bea9ea0ce7b29 : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20d30a8cdfed4b5bafcc0fef3c50cd7e : SUCCESS in 8m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9d580b621904d6697d0982852aa4bc7 : SUCCESS in 15m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/780af325a6ac4f88bb9ba9628272537f : SUCCESS in 35m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e676861cc47d4e4b9563b0247c0f1658 : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/beee4a3dd2ef4380a2fb77717b92678b : SUCCESS in 1h 12m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c78f175abf54631a077cbb997b9800b : SUCCESS in 58m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0479ba3af674f1a950844d15de63577 : SUCCESS in 1h 11m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/36de08a35a4e45d2b16d81c66115949e : SUCCESS in 24m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f832e728469e439d9aaa4df2ede32a5f : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d30cb8df96594aa38c1f860b678aa51f : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed888c300843408a8467179402767736 : SUCCESS in 1h 49m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32834dd83d054a228f35d3f1ff8ac0c7 : SUCCESS in 1h 21m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b796fe400abd497db803900e8d0985c2 : SUCCESS in 46m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/309125afa66541eaa50ccd725549b284 : SUCCESS in 42m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b33b3bbf89f74e0db3ac880dfdf4bd98 : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b65b9535a5534a3d9de7e8901af17fae : SUCCESS in 37m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6ec8c06623744cf8af49d666920b1c8 : SUCCESS in 28m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47b4544e99c84213b756224ad8bd610d : SUCCESS in 17m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a222f55026254ad280d39ce756595d8e : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67671028165d475884e366718f2154ec : SUCCESS in 13m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8eab1826f7f34db0a701a9272c3f87eb : SUCCESS in 18m 00s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f4bb1424748145e4a60fe9ea0edbf8ce : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26ff44b0463641c18d682ded6159e5c6 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":6},{"id":"f47d194d10e050c1e3eafc63caa56075978cb110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:03:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"547a93501eadd3ffe71d27f3627fe218071442be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 13:31:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadf32e04c4444ca89411c1946e9b8f4 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c60d60dbe8cf4eb9bb888f9930391438 : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/743f3bec3dee4d75a5dd2855d0589554 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174789ec40ef4499b7c07fb76ecde496 : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d9cf7b06334a4ab7ef6395928733d1 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/728e1effdf054580b25e399cf5a3970d : SUCCESS in 9m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc12c8daee13472eb02d2b8b678f54fe : SUCCESS in 14m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da6b97d111894dc0b1ace4b9e62e8fb6 : SUCCESS in 33m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/239c5d00946b4a1695c1d4f988b1feee : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2c4059932306419a93ff03b3817d0cc9 : SUCCESS in 1h 06m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8306a38ed4a473fa20e8d098d92972f : SUCCESS in 51m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4a660aeb501c438bb8696deccf389da7 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/38d7116d84de4b679392a263a48b5103 : SUCCESS in 24m 29s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3d5bc364fbf046078bf58ec80e4d93fd : SUCCESS in 20m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c571b78b0f44d58927fc73d037815d8 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/11b8c3d7d2eb43dc86fa3fcb2ec888c3 : SUCCESS in 2h 01m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70fa765001f4e69bab0e3598d5f2cbd : SUCCESS in 1h 13m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/386a65d038b34bac9e58fac247c55cd5 : SUCCESS in 48m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/233eb42beb3948fa9b6ed1e8db827048 : SUCCESS in 37m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4071de2339834ad09c9e9ea04011437c : SUCCESS in 33m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/acffb14d5df74f398b7dd10593369112 : SUCCESS in 41m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81198a4005a14249b0d43b84842d8e0d : SUCCESS in 29m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5b29ac7f927647b590e070b1b075095a : SUCCESS in 17m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/50375936e2a0487881c226ae2f4014e0 : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48aeb10d40e24970b46a02678c8efa6a : SUCCESS in 13m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d67759dde80446bbb2d46aac4984a7c8 : SUCCESS in 19m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8736a46650bf4a9ba15c408a476567ab : SUCCESS in 17m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b6e2aa8980474e99f8d5699e0422b4 : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":6},{"id":"a3f6d1ada7e2d911a554faee747cce247132a74a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 13:31:12.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"9e320f650379e3431b614f7b0c8bed9f2fe2c14e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 13:32:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d63d5d1fdff4010879b980f62fac392 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f402a754d14a4b68b4f8c6dce37b9885 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"48a6dcc82d86976a01da9880b04e0e6cb2c13353","revisions":{"97b2c75b62351453d4bf2c273eca10cb0fbc6779":{"kind":"REWORK","_number":1,"created":"2021-11-16 12:17:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/1"}}},"commit":{"parents":[{"commit":"b6fd87912d2fc2d7dbf9e4f13a4ec1784459c9e6","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6fd87912d2fc2d7dbf9e4f13a4ec1784459c9e6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:17:13.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b2c75b62351453d4bf2c273eca10cb0fbc6779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b2c75b62351453d4bf2c273eca10cb0fbc6779"}]},"branch":"refs/heads/master"},"e08514b7fc1c0017c61316c2ebefb8f68a4f5155":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-16 14:47:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/2"}}},"commit":{"parents":[{"commit":"87d7952910e460848c8fcf94ba1002051b9b6165","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d7952910e460848c8fcf94ba1002051b9b6165"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 14:47:22.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e08514b7fc1c0017c61316c2ebefb8f68a4f5155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e08514b7fc1c0017c61316c2ebefb8f68a4f5155"}]},"branch":"refs/heads/master"},"e9376a8074f706b398d54425f9b9d17a5595a9ef":{"kind":"REWORK","_number":3,"created":"2021-11-23 10:55:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/3"}}},"commit":{"parents":[{"commit":"87d7952910e460848c8fcf94ba1002051b9b6165","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d7952910e460848c8fcf94ba1002051b9b6165"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-23 10:55:40.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9376a8074f706b398d54425f9b9d17a5595a9ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9376a8074f706b398d54425f9b9d17a5595a9ef"}]},"branch":"refs/heads/master"},"b9ea42cfd0375b204534b53026ce194d96cc0ac1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-11 12:38:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/4"}}},"commit":{"parents":[{"commit":"3c1127d2ad5738c159b0e690045b48dd485e5e59","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c1127d2ad5738c159b0e690045b48dd485e5e59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-11 12:38:01.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9ea42cfd0375b204534b53026ce194d96cc0ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9ea42cfd0375b204534b53026ce194d96cc0ac1"}]},"branch":"refs/heads/master"},"08f19271e08a203ab466ef6b09e938759d24f867":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-03-21 20:47:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/5"}}},"commit":{"parents":[{"commit":"1f20a936816d2db9bccf5e2c0e1133aa1e4b500d","subject":"Ubuntu: add support for Apt configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f20a936816d2db9bccf5e2c0e1133aa1e4b500d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-21 20:46:56.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f19271e08a203ab466ef6b09e938759d24f867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f19271e08a203ab466ef6b09e938759d24f867"}]},"branch":"refs/heads/master"},"48a6dcc82d86976a01da9880b04e0e6cb2c13353":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-03-22 09:11:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/818075/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/818075/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/818075/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/818075/6"}}},"commit":{"parents":[{"commit":"4616c87010fe87caea302dc3c67d0b9b0b0f287d","subject":"Merge \"Add support for Rocky Linux 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4616c87010fe87caea302dc3c67d0b9b0b0f287d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 12:09:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-22 09:11:28.000000000","tz":0},"subject":"Ubuntu: support host package update","message":"Ubuntu: support host package update\n\nCurrently, the \u0027kayobe * host package update\u0027 commands do not work on\nUbuntu, since the main task is skipped.\n\nThis change adds support for running the commands against Ubuntu hosts,\npassing parameters to the package module based on the OS distribution.\n\nChange-Id: I879df0ea3f357c2bb45a5e7331a3f2569eac63e9\nStory: 2009685\nTask: 43988\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48a6dcc82d86976a01da9880b04e0e6cb2c13353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48a6dcc82d86976a01da9880b04e0e6cb2c13353"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
