)]}'
{"id":"openstack%2Fkayobe~818489","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ie572eda567782cc377767bd8d925145e2ad4bd61","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-22 11:39:52.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"Ie572eda567782cc377767bd8d925145e2ad4bd61","subject":"CI: always return host configure test results","status":"MERGED","created":"2021-11-19 13:25:07.000000000","updated":"2021-11-22 18:08:05.000000000","submitted":"2021-11-22 18:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818489","cherry_pick_of_change":812941,"cherry_pick_of_patch_set":1,"meta_rev_id":"b515524460b0feadc6bc68d153caec9fadcd25d7","_number":818489,"virtual_id_number":818489,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-22 18:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-11-22 11:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-19 15:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-11-22 11:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 15:09:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-11-19 15:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0326e0de8c15dbc6534b637c60cca18683fdcfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-19 13:25:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e4da58c9ed7aa43cee02f506bca95797720d5c30","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-19 15:09:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0175e5475dc180cd9f0e3ce1506062957bd2bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3dc43a8d18f469085be17f3f926ce48 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baabc156892a49409ea041297ec64ee6 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac67cb85864e4da78416006ef4f485f8 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd42ac405894565a1e4bf49d5cfc48c : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/020154a196574adfa1002b423d2f4e28 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8655cd03e0740cd9ae239866164d526 : SUCCESS in 10m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9680ec11651f426a92e70069c8a076a4 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2f3fc2110ed4b65adb0eec67c32a511 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd788f5aea574f7e94266f5586381414 : SUCCESS in 28m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef5cf5032cfc4853bca1be892e372cea : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0979acb0848d4f2db45ef24b39dc837f : SUCCESS in 52m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/33cb401cce3d488baf090da4f8c77efd : SUCCESS in 1h 16m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dca60820a69447869e7cdd2a7fef2473 : SUCCESS in 24m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae51fa6a996453a800c5a7bda118e32 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b81d31b57c54463904edc6f778ac554 : SUCCESS in 1h 51m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e1609e579b49d7b04604fd5a2317d1 : SUCCESS in 1h 16m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/586a2f188d384cc9b0950f3b1da447ee : SUCCESS in 36m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83f091e9eb6448b98d52c29f7bff3b2 : SUCCESS in 39m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/acb984e2767b486987114dd2ac55060e : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe5e221eb834871bc479b88acecb66b : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6faccc831dbb4dfdb3fbce19dd05635d : SUCCESS in 16m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f0839c95dc54aca9e89faaa1faaf85f : SUCCESS in 13m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8c9bc3b92b243169ac37aadb3de7b38 : SUCCESS in 18m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/005d4c71c3944c56b351a51e1a4d480d : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b31511d561cf17b93c0152ca8443067e207c2147","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-22 11:39:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe27a885de7d5bc0a05124dfc0144824c124ee74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 11:41:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edecfc206774df84ecb96ca74c8188c97dde8e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:06:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abaa3fe657d64a2fafda21d4a8bdca44 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faa351616cc540afb27a63348ffbb947 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7857b1b06f764a2b9d1e2785f5e3e74e : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22dbeac35264ef89bc32fa6f9517ed1 : SUCCESS in 10m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4eb323b145ce47ffb8fd90f4ca4fd2bf : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5866f5ee4bf74512b4dacab74af7c12a : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/039468004745492db479d713b9b1e7e4 : SUCCESS in 32m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/965c41fea34446d2824c998e19f57871 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd6182a18d7423bb233511da5df1742 : SUCCESS in 46m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dde500f89e784bd786d6d6db2b84c857 : SUCCESS in 1h 13m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b61cd52e988e4f6dbfec1b8458144c84 : SUCCESS in 23m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/986b40bdb07b41df97a1209c55ba8269 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40dcf276023941008abc3a2b137be883 : SUCCESS in 1h 43m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/353929dc07c447378c88279754e7977f : SUCCESS in 1h 15m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7f3c0a65be9c44beafc934360dff82da : SUCCESS in 36m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7196853fea14f9199719b2bbfe5d6c2 : SUCCESS in 31m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/726be27f751046e9b3a29cc23f18c06e : SUCCESS in 32m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86cc8ddcc5e548a0b693f8975d51f1ca : SUCCESS in 31m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/38dc6b8c82084a758c8bc507d7af4968 : SUCCESS in 18m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a146420f56884bfd803e115ea3a83b96 : SUCCESS in 16m 44s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/47e117067661447fa104adfeba4aa7e8 : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a90140b5a194808abce85ca45737fec : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0f9667c001cc71bfba29a0604e38c42b65c4b6f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:06:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b515524460b0feadc6bc68d153caec9fadcd25d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:08:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee720aef545a4643b5174cfda6a18277 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e6a66bf18a389b56d725eb2d1062e95e0ae70e1","revisions":{"8e6a66bf18a389b56d725eb2d1062e95e0ae70e1":{"kind":"REWORK","_number":1,"created":"2021-11-19 13:25:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/89/818489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/818489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/818489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/818489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/818489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/818489/1"}}},"commit":{"parents":[{"commit":"10bfbddc671162e75966b0cbe1c51293c96fb4d3","subject":"Merge \"Infra VMs: fix some playbook host patterns\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10bfbddc671162e75966b0cbe1c51293c96fb4d3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-07 09:26:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-19 13:25:07.000000000","tz":0},"subject":"CI: always return host configure test results","message":"CI: always return host configure test results\n\nCurrently, the overcloud host configure test results are only returned\non success, when they are least useful. This is because the zuul_return\ntask is not reached when the test task fails. This change reorders the\ntasks to fix the issue.\n\nChange-Id: Ie572eda567782cc377767bd8d925145e2ad4bd61\n(cherry picked from commit c2ae611ee89173bc0888b9f37d51b934dcb2ee49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6a66bf18a389b56d725eb2d1062e95e0ae70e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6a66bf18a389b56d725eb2d1062e95e0ae70e1"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
