)]}'
{"id":"openstack%2Fkayobe~818491","triplet_id":"openstack%2Fkayobe~master~Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","project":"openstack/kayobe","branch":"master","topic":"inspector-options-doc","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-11-29 12:02:58.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","subject":"Document that extra kernel parameters are important for inspection","status":"MERGED","created":"2021-11-18 22:10:29.000000000","updated":"2021-11-29 15:45:25.000000000","submitted":"2021-11-29 15:42:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818491-inspector-options-doc","meta_rev_id":"9e278fe54d28a68ccc4a232b1cce8c5195f39ae9","_number":818491,"virtual_id_number":818491,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-29 15:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-25 17:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-11-29 09:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-11-29 09:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 00:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 17:11:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-29 09:36:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2ed92646949ae127b5042cc93d7ee307e3f064a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-18 22:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac33d41f05ba7d9a255d67c8977f99037412bf3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 00:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef96d1a2fffd4f0e9262c64e63a7a88f : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d5422f3c5c40bba172f5c4d438dc13 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/717d450ec788492eb6010d6152f7acfb : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb02721a7100420cabd6c8107e1752cb : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80d0be87993a43569ab64ad5dcdd49ee : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/426aa4c36092474b9624beb7dc48cac3 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51c2f8d7bc234e5fa9b7851d72805f44 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f63a49ce5d8b49729495545ced2aa51f : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5360ec8051e44d28a68aa2132b9427a8 : SUCCESS in 31m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dfa32b461b5b4063aa2a6d7644355af4 : SUCCESS in 1h 17m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ede274a11114255a553533a4072e9f3 : SUCCESS in 48m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8aec1ff6258e424dbb7103d60dfc4fd3 : SUCCESS in 1h 18m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fa55c03696a84276b74af000205f74fc : SUCCESS in 25m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ff741b5202f4feeaeac539aafff2260 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3526df3a9070427c9008cd68fa52a4ff : SUCCESS in 1h 55m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f72717a847e422388f390c97e73a30f : SUCCESS in 1h 20m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1cc3436d6c664f339e1f6dca4001306a : SUCCESS in 37m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa6e9bfbaa4040229dd43bb8cd6570f0 : SUCCESS in 33m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b17f301a57f4f779fb71903788dcb3f : SUCCESS in 34m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a121716c1b304a18b4600db75852707f : SUCCESS in 30m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3acf081653e24b2db33745b2783ac4cc : SUCCESS in 17m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82c3a226b3a94db29bd4e0bc110be95b : SUCCESS in 13m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/96390078e7fc41f4bc78e0410aaadca2 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32e07fe8f924868aa0707aa3f6d6f3c : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fc051784368258a32561d738af1807fffd60867e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-19 08:38:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe36885bb9b52f587bc600e8913631c8948c7c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 10:33:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa12825cd6b84df7a4f7b6cee3cdcc0f : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b2bd684f5e402d9a9dd796c025be85 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/415e0d2a6ae24608ba6887fc10b02360 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ddcbba951c64f73938cf92f082ff5e0 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6f4a4a9775b4c1b80ee9b6a7e6c72f0 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6aa001d6e946278cc8ee54dbb9b014 : SUCCESS in 11m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2854933dba984ac4a663c4c7d8e932e4 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d201ba4e2baf4314b81e57ec0e5e2e7d : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfeabe58ce3644e693ae37f9e3096653 : SUCCESS in 27m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/62049121c35b40d5a570e259902d6576 : SUCCESS in 1h 41m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27b55adf25704547a1999b4be358ee5f : SUCCESS in 49m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f553b3d293bf47c4b86efb49399f7e63 : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/15e604433200418d9981ba1a0daaa195 : SUCCESS in 23m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943c309a1c9d43ffbd8c16836f48ef17 : SUCCESS in 42m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7e1354b4c0447e28751eaa1de9a972f : SUCCESS in 1h 53m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02ca4d6503d440c39fd0dc91c63867c2 : SUCCESS in 1h 22m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c7a0e19dc43946abb9dc309427952c20 : SUCCESS in 41m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b4078f40a3a417eb187a7a710c72ab2 : SUCCESS in 30m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aca0d0d3d2b045f09659dc568e9405c4 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2718d5aa26f4152aa92b9f627faab15 : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/37e545ab53954655914fd9ee058f35c2 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3ca387f9a414eb1aff99d704d10c106 : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a64a990b62af4b409a80f4d5f980291d : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ee42d8e25d44bb59c5b812b667d5910 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":2},{"id":"bbe0e6b2ede84beb80b566a69725a257fac69307","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:11:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb7f2a7924e14c47bc1a953eed83eb2fccb606f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-29 09:36:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7d2c394bd93f3b03861d5b9731a362153f9d99c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 09:37:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e40a3262f24fa677af4a8f75e4738ac30f10a8c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:28:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4031f0ec8ac14773898bfdad863ad844 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd283e47412b4e99ba6aa95bb6842b80 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a431206e5e5d450fa5887ad2cbac17f4 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93fe5d87604409eae0ecf89e3bf1250 : SUCCESS in 8m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f35c1d55c09c4d6b9492c8180d91c757 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4af00f4c55354073843b3828094952f8 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30d136f1ce12457c85b9f980f5e94150 : SUCCESS in 30m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0b32430a2f804186bce566ba71ffed5c : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2504163a3b03452f9666c28b0427fc18 : SUCCESS in 40m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/85467bae8c534c468fedd6534d7d9789 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/93e2a0153be24d63b94cfb4f5e6d0bdd : SUCCESS in 22m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e0ac49a6d7a4bf2be3834bec04e48d6 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a67eebd931b142709104b68236c1ed7b : SUCCESS in 1h 46m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714473eece08404b97c87f664f84b9f5 : FAILURE in 40m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92534ae94f644eaf955e4014a105a360 : SUCCESS in 43m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/783707b10b6e4145bc045331f786a9af : SUCCESS in 36m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bfe419e58e854cdda3a442648e8bec4d : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9526dfa5140c4a4cb3530f875bd6b418 : SUCCESS in 19m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c59cfe32f1f4295999dce4ccca6090c : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1f1053dbb5c46e59c60da31cd7a60a7 : SUCCESS in 12m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/06a9d6e5090e48759715021a37e90934 : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e3d331b738d427db948f461f39f656d : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6b8cb74aaa45615b09b60f3f72e5c2509a9551e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-29 12:02:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5f75a8cea1c57a30f25bfecbaa4aa8398314de1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-29 12:03:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"609842d026e2e3aab741d66c4e59e990a49bd862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 13:48:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/111b86a472b0484c87aa4926fe31c346 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61e6877408740f589f2932a2b094d81 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55250db319654a28b8f320361c1d8119 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cc4c310e1f94bc5a756c489fff8d206 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54360f8b72534e42bc85129cee8e33c9 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5638ff7aa5b4da08bea2118f4275af2 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53b8b23e9de6462ea531af23296f0f38 : SUCCESS in 13m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14aba1e47e154d12b35ed3c276bb2f3c : SUCCESS in 27m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1c5a896536254eebb23b776c5b6fa6dd : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d01c86fd2fe4290ac252c996c65d37c : SUCCESS in 49m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ffefc734769e4bd4a5da4901871d6bed : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/273349609ed74008882ea19b430147b8 : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de62a2a44cf04ee299b60cf6540dec4b : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/62ed9d423caa44d5ae2bf86fab7ad28d : SUCCESS in 1h 43m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45acc42eed504e2bbd65cdadc85ffee7 : SUCCESS in 1h 18m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1131b63e2b54d678490cee56c5ac2ae : SUCCESS in 47m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0649cd09bf424fa259908a81b5851c : SUCCESS in 39m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f50673b074384238b045b0ac31596afa : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f5817127f8e4f4ba91eec5d8fb90e38 : SUCCESS in 21m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8317824d6f5442679846b217f8802f2f : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47f0b0e4b47e4b22950ca66a83a8b4b1 : SUCCESS in 9m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b3811078ad2473888a46cdc258b33f9 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae7752e0273b42be9a3c4b30524e4556 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f730c4aaceb6c08cbcec6a3ec490851ebab2785a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 13:49:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f6398a05b9272091491a89cf244f476a0cf7aca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 15:42:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5379bc9f76d46e099076a1282052a53 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50a148869d654b08b8c10e6766ccd0b2 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67e174e93ae4737bd4b929d2ace9992 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708d84b7ba484344a0d4ace701ce5a0c : SUCCESS in 8m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cac9565f70247059c2788b700c8c690 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/257c7881ae99417b869e7891ce249b4c : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9ff155967ee495f9cb4c2fb79c8ce82 : SUCCESS in 29m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/13e94b88d4ad48cd9ee249e9408ca327 : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d334730d2945452c88a636e9cb21e0b8 : SUCCESS in 49m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/41a46355db2d41fda06cdb0c76e885ef : SUCCESS in 1h 12m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/de4f8ed1be8045119b34caae485650c0 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c38d0628f934dbcaa489664f0ab3627 : SUCCESS in 16m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/48deef5bba5e44479f39335e5deebde8 : SUCCESS in 1h 51m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9805705eebf4fdd9e45e9c9dfd64d9c : SUCCESS in 1h 13m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/04f17f7af409453cadd1b35f38da0088 : SUCCESS in 49m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d0a66a9d833469cb76c3e246d668578 : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2832711ef8e14e8a85221dc28aa8bd41 : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9302717353774f7d83c542e436a9531c : SUCCESS in 23m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fd17bd835e71454dba8bb60cbde8fa5b : SUCCESS in 17m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29ecb1fd2d8341e983ec7f27901888b6 : SUCCESS in 12m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e841ee20a62949d29167ed1889779132 : SUCCESS in 18m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/717bf33e8ba84a69a74ca759820c8e32 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9f94b9e313d7494b61a6ba9e8c1dd90276046f11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 15:42:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e278fe54d28a68ccc4a232b1cce8c5195f39ae9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 15:45:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d55dfeff2e1460d8bf9c8e02e10f88b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ed91dd06613b36889597df0cd4a8792b43f1b61","revisions":{"be5a28478a58a273a5f1aba534a89ae80c22554c":{"kind":"REWORK","_number":1,"created":"2021-11-18 22:10:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/818491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/818491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/818491/1"}}},"commit":{"parents":[{"commit":"5e9affb62c40a885f4d64b0dbcc4bf6576b8779c","subject":"Infra VMs: fix some playbook host patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e9affb62c40a885f4d64b0dbcc4bf6576b8779c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-18 22:09:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-18 22:09:31.000000000","tz":60},"subject":"Document that default kernel options are important for inspection","message":"Document that default kernel options are important for inspection\n\nChange-Id: Id789dede2d1886ef5ec66ebc86968d6cf62fa2de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be5a28478a58a273a5f1aba534a89ae80c22554c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be5a28478a58a273a5f1aba534a89ae80c22554c"}]},"branch":"refs/heads/master"},"7ed91dd06613b36889597df0cd4a8792b43f1b61":{"kind":"REWORK","_number":2,"created":"2021-11-19 08:38:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/818491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/818491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/818491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/818491/2"}}},"commit":{"parents":[{"commit":"5e9affb62c40a885f4d64b0dbcc4bf6576b8779c","subject":"Infra VMs: fix some playbook host patterns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e9affb62c40a885f4d64b0dbcc4bf6576b8779c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-19 08:38:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-19 08:38:27.000000000","tz":60},"subject":"Document that extra kernel parameters are important for inspection","message":"Document that extra kernel parameters are important for inspection\n\nWhen customising kolla_bifrost_inspector_extra_kernel_options, for\nexample to define which console to use, operators should note that\nKayobe defines extra kernel parameters that are important for\ninspection, such as ipa-collect-lldp\u003d1.\n\nChange-Id: Id789dede2d1886ef5ec66ebc86968d6cf62fa2de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ed91dd06613b36889597df0cd4a8792b43f1b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ed91dd06613b36889597df0cd4a8792b43f1b61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
