)]}'
{"id":"openstack%2Fkayobe~819009","triplet_id":"openstack%2Fkayobe~master~I90223a36ff3fc71585cb9302573154a5921041ff","project":"openstack/kayobe","branch":"master","topic":"docker-module-fail","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-01-07 12:49:51.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-07 12:49:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90223a36ff3fc71585cb9302573154a5921041ff","subject":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027","status":"MERGED","created":"2021-11-23 18:27:06.000000000","updated":"2022-01-07 12:51:33.000000000","submitted":"2022-01-07 12:49:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819009-docker-module-fail","meta_rev_id":"6d82b8b0845b3edc63f160595d6158163aa7f302","_number":819009,"virtual_id_number":819009,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 12:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-12-21 10:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-06 13:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"date":"2021-12-21 10:25:17.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-06 13:53:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 20:22:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 09:59:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-11-29 09:36:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-09 14:04:59.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2022-01-06 13:53:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"42d76e3a4169febb5ce373f4d10c6196895103d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-23 18:27:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa4788348283ca0f0cb1f31949b9d746e14961fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:22:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/812eedbc9226406a8a8aa40d6561af51 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94472dfcd25442eb82e0d689581b193 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92ddc1d8cad84cb0b4009bdd79ab3f4d : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d18ec693e94377a72c11e850da02c9 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c136ff7599e4047803881744e06094c : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e91ec594774698a37f1b73a73aaa69 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/119620f24c224eb5a5d1e4ec69e29125 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e979ed5b51d417d9a9441740a6c1334 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f84b9694e244d64b30709ac8bacdcd4 : SUCCESS in 26m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f4ace444780443ad9dfbbde129fb530f : SUCCESS in 58m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd1e88a7b3e4ea1abd97c35d46da349 : SUCCESS in 43m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6caa79f03fda4313b4b0f435c5e7252c : SUCCESS in 59m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a21a9a6a164b449585eb4a2fd1d4fe64 : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603300ac81914754819c3562b8707c48 : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7c0fae485e7d4d339308533830e77d12 : SUCCESS in 1h 51m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3300d137224c4890ab8bf786f3c6ac9b : SUCCESS in 1h 20m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/77a3d92b3893470ba0dc30a832876955 : SUCCESS in 35m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca2ceb0c1f114081818edb6cc910ff5f : SUCCESS in 34m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b8beb7934ffc42579fffcb6da0b6116e : SUCCESS in 53m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6f7cccb977d4df5946bced5b7d483d1 : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8bd6c6f59954b39b2dcca24689a4b4c : SUCCESS in 17m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/494d3cbfb0fa4b92adf31d3fb0cdcbd5 : SUCCESS in 10m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c0dab6964eb64649b2994bb9bedc1f64 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d531d55852d8497f8e48ee0a4ee437b5 : FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"692c75f36ea08618abc548fa4b04694081cd2d26","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-23 20:43:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1b0bdbf2b7a1b2a5051fb0346acc39ef23ec1a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b23ff29d76e44c9885b5aac2dc58a98a : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de28736c8723453a956613fa83ef6827 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9652021b35794848b23c8d47d072594a : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5748ba6d7a854b55a9168bae64e5eb21 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a82a9fbfbf3f47238499d0143f669202 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c6e8eab91543b28fcc16b9164ab221 : SUCCESS in 9m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/723ca75f03504ebea9bd62475620a676 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e6f8ca6d91e46f4a3586b621c50eb4e : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e92cff3b792c475ab6053be8b28cddd8 : SUCCESS in 36m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3589c90916fa4c929c9ad7ca5660feae : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470bbc33f5974daf9cf2230249a78dc3 : SUCCESS in 50m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f039fb655c1046ee98ae5189f2a78c79 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d76f815cf6224e33bbb7b16a44f15a80 : SUCCESS in 23m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50df0902e6774e518a315a9d3fc62804 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f948383d2b8450f9b58d97cecd06fc2 : SUCCESS in 1h 53m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c6b23f96324bbea9339fb28c15b0d8 : SUCCESS in 1h 22m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e0d02bccb4f340e1a7775097471fe99c : SUCCESS in 37m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60eba0b72511468e89950a5591ccadb7 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e64a3b61403b49e294976d64fabc2beb : SUCCESS in 58m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e33429fcc09c4f22aee88fac491ffa45 : SUCCESS in 25m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6c83d3f1ff0e4fb98a69183af711f8c3 : SUCCESS in 16m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef046fd9ccd4cb1abb259ad1154263a : SUCCESS in 12m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/76885cb054b44c2cb876f493b7200da3 : SUCCESS in 18m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b101835ce9434c0dafb66f44826e675f : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1289dec1062a8fbb1d9e3a44e424b185c3bafc1a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-24 09:59:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f97b4abe4a5e150a0d3827ea6211925adf4c3ab","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-24 17:38:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e2611b2ca1cf24558e5f4dd041436a6166a809f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-29 09:36:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d7df19959ef617489b84ffe5410dde3f6433c7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-29 16:34:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56a7530bd15a43bfae3f338594ed66844562dc5f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-29 16:35:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e1ca225231d4d4d8318ec769da616fbc145f2742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 18:25:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4116b53262044bce8aae26ad8c54d2b5 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a342a92739343cfbe1184f45a1c1bfe : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2df27f01f0524fa4ab60961ff5937ae3 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c1ab84129c4ad4a77a46077cf4181f : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f5b4aa26704ad5b3965b13082a705b : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea35e8cc6f7c46399465f7e615217037 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4364a7ef0d3249719c29f43a23a1fa2f : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c648e6ae8864e84b2a9eb907337e4e7 : SUCCESS in 32m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/18338bbb8d3e4ba0ba4b5754d6bc4612 : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2063de1ce2043aa8abf124f6a58346a : SUCCESS in 51m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/af32aa0051e04f80ab74c0f91c294877 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6c83de58a9d64d7cb2725e22fbddf990 : SUCCESS in 22m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8125cdb4cebb455a8c161900df66e031 : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/086f522ae11a46928e6c2e6eae40cfa8 : SUCCESS in 1h 48m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fcbf067d1464ff28c51c936ce94764a : SUCCESS in 1h 16m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4bae6dfd591f45beb7fe81032c50ebec : SUCCESS in 48m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/984c6006d1194cc08e4cf021d7392015 : SUCCESS in 42m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1b6536cc77d04309a742dc8dff265280 : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3897872ef8ab4faabf1ac5f34b2fa87c : SUCCESS in 23m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f863fecb2a7c4c46a5709d6555d1b359 : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6c8eaec6e34a71a7f6fd999a78021c : SUCCESS in 12m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff97d109e896436e99be5c83f0955658 : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13efe98a92544f078141f500bd02d034 : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"fb46648bd4b2282fc37118e02d9a56e69aa51c7d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-29 19:56:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f15b6cf82ff91e96e950e194812109b0c1cc0bbc","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-12-09 14:04:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b10bf60c62c3e6314a55a62058d6ada354609af3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 12:08:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8542e76aaec637fe72a891c165d946cb39eae3ec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 14:01:36.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6454ab288562c7b3219bcf6e8318b41fb26eb6ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 14:03:13.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"26baf730460e8726101b98bd5dd7d5b9e0bb470f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:01:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/357bc1a6a3c047f28d3ee89440add139 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb55ad884e564359a2d4d9fce8cb50e2 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d349359dc3493a8a1dcd343811aa23 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61f93c4a9e11448abf7bee6ff965e8dc : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419a63da37494c9bb0a35eb21c646093 : SUCCESS in 10m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b453b65bbc81418289fe6b2f7d00c606 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2bd246cccd314a1799a564fa2ca70d4c : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af2d19775e2c45f2bc68ccc41a4adfa2 : SUCCESS in 29m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eb3465ae396d41afb93557d1f1746ff9 : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbd75e93337d40488b5ea75e54e4ce26 : SUCCESS in 44m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ae85b2925c2444b1b67286d9cea408a9 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2de19216b13f4c9f97b411a574222f64 : SUCCESS in 23m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d04f19559b534295bc0a91a71d84cc5e : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/393376d1b8a24bafaab43f28b0c892fb : SUCCESS in 1h 54m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bd6bae315864bf5b030ba581052ad48 : SUCCESS in 1h 16m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8722a85c713c4babbcdc8c24db7b581e : SUCCESS in 51m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d943f40046ad4d5089af62e6e5fa9d7f : SUCCESS in 38m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a9a710f44bcb4ca99da9980c23b47bb3 : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b3e4f8efba84058a511dbbb5aa80fce : SUCCESS in 21m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/95c96e759eb64516b46bb6609a2ff020 : SUCCESS in 17m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7de76f39a7b4a1aa14b1569ef57681b : SUCCESS in 12m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0417f18ffd9b4e1bbec9338a905671ba : SUCCESS in 18m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f35ca8f8163d4e46b4fb5b6a9dedc630 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":3},{"id":"aaa81f5e233b23ddec200291c1c460f489cd6bd6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-20 19:44:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"a90e56324d578713fee92956a95466d987c18170","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-20 19:48:59.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e69827ca5834509698a48f654946eb382519b66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-21 10:01:50.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"367ead4da5ce7f70d017cc7b59b0a7d433278fad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-21 10:25:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"daf029a060ea707988063190f0e408c2a4a0fcd0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 13:53:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9ef0eb19f8c199fc9a7ce99559e0ce26ec9fe6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:54:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77f054d13083a4f989e08d1745846f8a06655b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 20:28:46.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60de4f3542549feb3e90b3061828d9e : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/108dfd957f09437296dfb1d3f67d4445 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c02815f95774ed08a07df9e9d897696 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb7a6f27e884cc89e4f6baf4e45ad96 : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aee60189b87642a488468d8821aa374a : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e39a4e8021234ea0bcdb6b35dd535f19 : SUCCESS in 15m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1df1651d2ddd4d9cb32d94fc6afb48a3 : SUCCESS in 35m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2543ff3824cd48709463f6afcf8448b3 : SUCCESS in 1h 16m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab471374d7d448695b59a37164c11f0 : SUCCESS in 45m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/488b2f4221fc45cba33e2da3948c428a : SUCCESS in 1h 16m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0308cf436ff2458790b490f5969d2a58 : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5de45d09a9d34a19899520bbe4f5881e : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7bfa2659703141168639094f62220d00 : TIMED_OUT in 2h 35m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/818f89ed05974f32adaf7193b79bbc7e : SUCCESS in 1h 16m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ae2e827633fe4642996ac29e217b420b : SUCCESS in 55m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6313977e2bb04327b1e49264b971d031 : SUCCESS in 45m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e97eebce2d64d2ba4276a93a5b1bae8 : SUCCESS in 40m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fa4798404754371b82244eb1051af60 : SUCCESS in 29m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/919b461e923b42739d36c6187abd3f62 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b035b249c814ac5b990815e6116e7ef : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c1a1629fbc254d9bbf300fdace2698fa : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ed462762a3476ab3111f72fea5591e : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":3},{"id":"200f260910c4206827b0fbedbee1ff1a49e986fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 22:12:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5fd268cbf9f5949eba9216b704dc221398b9d2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 00:16:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b1b21ac7ad345c6b5072a606d45d302 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5049c9fab1415597f840986e8af131 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/484b0ce0740743ae94fea86f2fae1a00 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b98b802b89345c0bb2933402c6968ed : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c233012d744c4e67b0878f62b4aa6500 : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97d78dbfdfa744708fdc33e40bf3aadb : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/24915b8e744547b3a0fa107a76497f6d : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50cb7693aa5248fda8757af7c74d0552 : SUCCESS in 33m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ada6540200434fb488bfe9402bb780e9 : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68b799f6bd8d480ca9fb6e6b3b004297 : SUCCESS in 43m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d1cabbae303b4ea49632875de6904d87 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d81d96e35c504b539fb82fe48d55f2f0 : SUCCESS in 23m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2ba36e4f1f54aaeb8caea8c8e315797 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d03fffde482b45b98cf0f1d5834d0bdf : SUCCESS in 1h 59m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9eb9113589147e189a488ba6051be31 : FAILURE in 34m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d9d314f6d7d342738fc100ced7067410 : SUCCESS in 1h 04m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c474f251a9f8493882832a71a88cb4fd : SUCCESS in 39m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cced3c56e8a341c98f3fc430797acf15 : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95c3d6609a4f4103b39dc21690050501 : SUCCESS in 19m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5961845d28144e549212e508ed4312f5 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7cd7f4007ce4e8ab4f921ac0715757e : SUCCESS in 19m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/406cb3f287e3445ba6344a1fb6ca4ada : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec831082a6ed42bab962674b4cc57830 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":3},{"id":"2831f425417274dbcc12648c58d68e01924586b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-07 08:53:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59a86a1c6df6362f5a94be0931cf4be03a771664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:53:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572c740bb6d54a03a19fc63cb6933c49 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106cfec78a6142e3827a3702159eacef : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3326af63916b4eb194aac6eae69aa245 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4198193e646444a945f52e3325b60c6 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e83d1f0db9c448ca49fa724fdf944fb : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33c58d707eba49f5b6890504144d16e2 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48f1c281611840809bdd440b11058583 : SUCCESS in 13m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67da1c44f46c4791aacc368fde35f542 : SUCCESS in 31m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/529fef45ee7841f39e1f2229a1d6541c : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5e91306d5f34c1f948e64bc35e5e314 : SUCCESS in 54m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/adf835a7f503417bb4f2b202e967e4c4 : SUCCESS in 1h 13m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7e3f312605d84bac92a5f5f0216966d6 : SUCCESS in 23m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b761596adfa4809864c338ea6fa1435 : SUCCESS in 18m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3cd9210dbf54a2481932194072a86b2 : SUCCESS in 1h 58m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b4050c4b89c409fa68bd1c57857cbf9 : SUCCESS in 1h 11m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4d06f756fd0e4c33abaeac4dd1e38262 : SUCCESS in 57m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b64424ae99ea4c20959228cfc40d7b20 : SUCCESS in 38m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c9d61f673f8471b876a848027f0ba37 : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa2a6920a4424cbeab904ee98d9190c7 : SUCCESS in 22m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0b009721d6de417e9d5a8a65d63a4c72 : SUCCESS in 20m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fde7ff80401a4eda98e38d0cbeab76d4 : SUCCESS in 12m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5aef99f32704172a6b3c88a52967587 : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdd3a8aa49b94f4a947f6648cc49316f : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"4f85d37d045bd3bc1ae96f5d8cdf2b7224701590","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:53:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"956bde9fea85f1725dfbdbfe939fdc302b5d74d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:49:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8fbf137af04fdb8f8623a110311fc0 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d4848c612547419cab3eb5049512b4 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b85f683f4b944e995fd1b95b35e7209 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420eeaa0f8234fefadb1691b88f86232 : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3df5a26eff9f4e3faa27219be50ad785 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83aeb61747ed4a768b22a8dd04816ea7 : SUCCESS in 15m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e5c772e808b4defbdfbb08d63c7a2a7 : SUCCESS in 35m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c23cd6355174ea6886e6c039aa2b1c4 : SUCCESS in 1h 16m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ee5ec7682fe44a28cfd940724ab5438 : SUCCESS in 55m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/081c9640e23c4a63940080e2760cc3c3 : SUCCESS in 1h 15m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e79cdf406fca4fdd953f57dde1bc73ad : SUCCESS in 24m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93c848604b454bc883347c3f7b2c913c : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb2bb04a0e924fa294c07be5bd6a31f0 : SUCCESS in 1h 54m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f82a12fd1d2e4f518376e902bdf29ac3 : SUCCESS in 1h 20m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/53f236ad514c4c2c8bc2c20df16c0349 : SUCCESS in 57m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08f92c0f194e461381906b759821c7de : SUCCESS in 40m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8d010a0443274a67a41bee45d06240c5 : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/347b74fd338d40168acb29f414119f76 : SUCCESS in 23m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/75bc97902c1c4b348762158418e8f18e : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49f66e428ad541d68d08b0420db01968 : SUCCESS in 14m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4da81303d73b4b2da5bdc45b18bbad0a : SUCCESS in 24m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26c3d61a9d744f38850b1373b275042a : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":3},{"id":"8a8350a0323fbeda894e2eb5cfc10b02ce6e0208","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:49:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6d82b8b0845b3edc63f160595d6158163aa7f302","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:51:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70850610bae4b5fba4f3d0b1d8a62bd : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bf318310e6534c6074d15229c20555b459e517dd","revisions":{"4f83f9b56848f60dc81e0ffde91055fd3271f364":{"kind":"REWORK","_number":1,"created":"2021-11-23 18:27:06.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/09/819009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/819009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/819009/1"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 18:09:49.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 18:18:57.000000000","tz":180},"subject":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027","message":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027\n\nIn some cases, for example with custom inventory, we still have an\nordering issue and shouldn\u0027t rely only on \u0027docker.service\u0027 check.\nWe should use \u0027docker rm\u0027 shell command prior \u0027Docker SDK for Python\u0027\nis installed.\n\nChange-Id: I90223a36ff3fc71585cb9302573154a5921041ff\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f83f9b56848f60dc81e0ffde91055fd3271f364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f83f9b56848f60dc81e0ffde91055fd3271f364"}]},"branch":"refs/heads/master"},"c27cc664ee4c0db97a9ba895daad6208bb33f77f":{"kind":"REWORK","_number":2,"created":"2021-11-29 16:34:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/09/819009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/819009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/819009/2"}}},"commit":{"parents":[{"commit":"9f574402f38750c042a6a81aa7813571fdae3f4d","subject":"Merge \"Document that extra kernel parameters are important for inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f574402f38750c042a6a81aa7813571fdae3f4d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 18:09:49.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-29 16:33:49.000000000","tz":180},"subject":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027","message":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027\n\nIn some cases, for example with custom inventory, we still have an\nordering issue and shouldn\u0027t rely only on \u0027docker.service\u0027 check.\nWe should use \u0027docker rm\u0027 shell command prior \u0027Docker SDK for Python\u0027\nis installed.\n\nChange-Id: I90223a36ff3fc71585cb9302573154a5921041ff\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c27cc664ee4c0db97a9ba895daad6208bb33f77f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c27cc664ee4c0db97a9ba895daad6208bb33f77f"}]},"branch":"refs/heads/master"},"bf318310e6534c6074d15229c20555b459e517dd":{"kind":"REWORK","_number":3,"created":"2021-12-20 14:03:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/819009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/819009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/819009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/819009/3"}}},"commit":{"parents":[{"commit":"9f574402f38750c042a6a81aa7813571fdae3f4d","subject":"Merge \"Document that extra kernel parameters are important for inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f574402f38750c042a6a81aa7813571fdae3f4d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 18:09:49.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-20 14:03:13.000000000","tz":0},"subject":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027","message":"Fix \u0027ModuleNotFoundError: No module named \u0027docker\u0027\n\nIn some cases, for example with custom inventory, we still have an\nordering issue and shouldn\u0027t rely only on \u0027docker.service\u0027 check.\nWe should use \u0027docker rm\u0027 shell command prior \u0027Docker SDK for Python\u0027\nis installed.\n\nChange-Id: I90223a36ff3fc71585cb9302573154a5921041ff\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf318310e6534c6074d15229c20555b459e517dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf318310e6534c6074d15229c20555b459e517dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
