)]}'
{"id":"openstack%2Fkayobe~819030","triplet_id":"openstack%2Fkayobe~master~I4884cccf8a54f6c5af37f0ee418bb3821dc87091","project":"openstack/kayobe","branch":"master","topic":"pip-uninstall-ansible-base","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-11-25 19:18:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4884cccf8a54f6c5af37f0ee418bb3821dc87091","subject":"Uninstall ansible-base package only if exists","status":"MERGED","created":"2021-11-23 22:17:21.000000000","updated":"2021-11-25 19:18:31.000000000","submitted":"2021-11-25 19:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819030-pip-uninstall-ansible-base","meta_rev_id":"00af5f48079ce7677fd4f15c3bbdff72390da223","_number":819030,"virtual_id_number":819030,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 19:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-25 17:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-11-24 09:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-11-25 17:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 00:44:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 09:56:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-11-25 17:09:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"11eb0c0f6ee534a3adea054f7edc4f7214177bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-23 22:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"955b986d88fdaa81d01f843a1f3d9c0c6377a143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:44:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c013b78712f74064ba09eb0bd7446216 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0444a02851e2406cbc048a902a693b69 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e50a5b5ecb149e58d74658f365e0cc0 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2941066b8811481eb74d90eb200d9f57 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/187bf67d43324daf9f0a57a4f45cecd9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71dc46647db8492c96ab66448dfbe254 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c7b8ee9867f457cb35bdc1e93ac9a9e : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/339e47ffeae44ae0bdd1f895185d79e9 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f1f741c6b2a4b169ef5c1965e4efba8 : SUCCESS in 30m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2451ea8d85a5404984a0543d37573b21 : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/739951a68aa44ecba0d04343ef52929b : SUCCESS in 50m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/843b74ce690c485a92255cfab5f84377 : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9cf66388e01244038626dfcb0446a3e9 : SUCCESS in 22m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ea988409e934a4795033dd4150fb363 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/405290aa76fd404ab376592cd10e808b : SUCCESS in 1h 53m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b09bcadf101c4924b61112ae17951a37 : SUCCESS in 1h 20m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2fb3eddd9e3f4a348639ffe82adc789a : SUCCESS in 28m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/144ad90e7e4a4f02a4ac0a7bdb575467 : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/48d75a8a32bd4123a1eecd439fffcb82 : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4ecfc8d42544d38e83d8feca10bfeb : SUCCESS in 25m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1d5076bb342540ea8bda00af794fe23c : SUCCESS in 15m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3868be001a6346d8a2cab0bb21f1cb3b : SUCCESS in 14m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8922597c373e480cb7db5c05ba9ff00b : SUCCESS in 17m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a309276bee054b6a84f3f33a16e7d87b : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8fe0b1ec7585aa41245b4e3a48fd139cb3dde4df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-24 09:56:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc065def7fff0f9fd1f054c927100fb6896f870c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 17:09:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64b6d6ee20a325566ff735c73ebd045a8f484609","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 17:10:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a37a08540a095d952fba2d6ff305fb75728a56c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:18:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1870084a4b94515b716f9eb3e0047cf : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ecf80600c1413cba1425258af5e462 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0a35f18967644b4856c628f97652011 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c4d2fb3aac41af97848a5870db528d : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd21b7236f884420ab08b599df533dae : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4710067a387645fe95667e0227d06319 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e9309b2a1fe46269050f76d7db5a248 : SUCCESS in 29m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ce9636a0365a4223a7a370c6ecfdc306 : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cf72e1e980b4b1a97009e46b2d1c527 : SUCCESS in 47m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eea4de8d967a441587617f9de3d2b85c : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5151fc2cc08d4c1990c9c0446107ec91 : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e1f1306bb284703af059ca2fdd7ba47 : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/23519b93457f41e58a9814a444b20c90 : SUCCESS in 1h 55m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b85f53313d48b69309d17edd944951 : SUCCESS in 1h 20m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/36f879f96d7d4f32be3cb4232e1a31a1 : SUCCESS in 44m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1468d0a2b6994525bb433779ad3dd3a8 : SUCCESS in 32m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d384298123d94498a873c55291c0c46d : SUCCESS in 31m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2175a690fbf41da8a6c06a07440fa26 : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e3352f0e6bc24613a46a5d0d0582242e : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/959527de1b904f799d2cfe68c9765e13 : SUCCESS in 12m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/99c1f528c99d47c6918864bc3705dbf0 : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f4e98589b6549f09a7e5b8d3de41c2b : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"00af5f48079ce7677fd4f15c3bbdff72390da223","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:18:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1aacc0f9fb40aa8803c6b1a27b7e2345e7a5fbca","revisions":{"1aacc0f9fb40aa8803c6b1a27b7e2345e7a5fbca":{"kind":"REWORK","_number":1,"created":"2021-11-23 22:17:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/30/819030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/819030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/819030/1"}}},"commit":{"parents":[{"commit":"52d7523c0c3e6a6f7f321459b9d7fc055282057f","subject":"Merge \"CI: always return host configure test results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52d7523c0c3e6a6f7f321459b9d7fc055282057f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 22:08:22.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 22:08:22.000000000","tz":180},"subject":"Uninstall ansible-base package only if exists","message":"Uninstall ansible-base package only if exists\n\nWe have the list of package installed, so its nice to decide when we\ncan skip the task.\n\nTrivialFix\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: I4884cccf8a54f6c5af37f0ee418bb3821dc87091\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1aacc0f9fb40aa8803c6b1a27b7e2345e7a5fbca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1aacc0f9fb40aa8803c6b1a27b7e2345e7a5fbca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
