)]}'
{"id":"openstack%2Fkayobe~819372","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I35788e69eeace41d5bcbf89922145f1c35f58ff2","project":"openstack/kayobe","branch":"stable/victoria","topic":"limit-seed-playbooks","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-01-01 13:38:28.000000000","reason":"Maksim Malchuk replied on the change"}},"hashtags":[],"change_id":"I35788e69eeace41d5bcbf89922145f1c35f58ff2","subject":"Limit ip-routing and snat to seed hosts only","status":"MERGED","created":"2021-11-25 19:29:57.000000000","updated":"2022-01-01 18:23:49.000000000","submitted":"2022-01-01 18:22:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819372-limit-seed-playbooks","meta_rev_id":"713c36feddf300022ac99adca0bec67a36084c29","_number":819372,"virtual_id_number":819372,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-01 18:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-22 12:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-09 14:05:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":2,"date":"2021-11-26 09:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-22 12:05:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-26 00:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 09:01:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-09 14:05:46.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-12-22 12:05:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d8098be90b8e3b024ae2ffd13fd5487a73b0b733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-25 19:29:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57c8ffd9a59b68c04ac175baa160149d6b082989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 00:56:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fabdbb1ee33b477385e4d7f77b59df6e : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431b3633e38946cb980f7a0bfccbd874 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faaeee699284c47bcc22aa3d736b1cc : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c139257ecf14dffa24415426e0388e2 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf9837a7d70489fbfedbf4687280b0e : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9aec52837b249d69a9568f06b9ba490 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bd9cc2eb2ab41189bf2c3e79a8e504e : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b59cf8e6b9e94c3a80c851e3073c31b8 : SUCCESS in 19m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/538a136e4f054f90ba3f0571b6e0596a : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a2755550f5c74b2ebef51d547707d1cc : SUCCESS in 57m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a8e2f68341e64662b13c1ada1820a48f : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ef1f10a9b5e4d4fbaf2c01cfc4a16eb : SUCCESS in 56m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/651c736d0ac647ba85d2d98e256a6156 : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5842a57289024a4ca319ed2675648fdb : SUCCESS in 17m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a3e173757588445cbe629d10970703bd : SUCCESS in 2h 24m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/001dfef61e85480b9af0339da9708f31 : SUCCESS in 1h 45m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/110bcfe928004d91bd80efa4361661ba : SUCCESS in 26m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fe584ff15a084845b06a1bfe6bfed6c6 : SUCCESS in 32m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b44978985ec4ab9bdb80097294ad4fe : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39dde777c10f453ab31bc184a0dee60d : SUCCESS in 24m 45s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f781f57cd706426382b4732bada7b1d4 : SUCCESS in 13m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc68e327e7394ddf97621eada1af60e7 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4f0f3c74a628420289ccbc1f9eb38148a47015dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-26 09:01:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea7eed71c9d87d3387e5a853385fbff5f68e1f2e","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-12-09 14:05:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60f9333575469c45586c822b3895bb6d52b5f416","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-22 12:05:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89f7bf240ec4e1e23b8d1adb90300815b115c552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:06:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c395f2c014ff45c508abda41d03f850ca5d55b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 13:57:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4049a0b72a9f4fe09e93c9b5dd13de22 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/289ee6840757460c9d0264818c25c62b : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1f2f0c99d984686839751e83b00d7fc : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5024c655f7c4306ad6e0648a370c2ac : SUCCESS in 7m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d37a08639b1b4099803ecaa35b9186e0 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fe2063f861e4d268c84940457f584c3 : FAILURE in 3m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7305ed27738b4a81b7f7b2f2856ae3c6 : SUCCESS in 18m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/736c90b25237483a966fb77d57136ab0 : FAILURE in 21m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/073bb11123ba42b2a488cdfed4db1ff6 : FAILURE in 24m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/91eb3c4eb67d4e868286fac82ea2b407 : FAILURE in 16m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5dea79e9b20c4031979fca29e8b815cc : FAILURE in 15m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7bf33c3ae2b54dbabc05e0b9bb744ede : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c2ecdbf79d20497c91bd2435a9c2450c : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2d966d91d1d846919341c8c189ba519e : FAILURE in 17m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b53b9618e091411388b3d86ce6cacd67 : FAILURE in 21m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/97cfce5e36e9430f8ab91c4eba4b4acc : FAILURE in 11m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/de61ff973c3742ae970858921345a8bc : FAILURE in 12m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bdaa3cc5d2a24f4b9aa26793935ea3b6 : FAILURE in 14m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0050a71c282d4bb99e05285a3eb21faf : FAILURE in 19m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f48c8b2d9b3f4b859f56c59146dba4be : SUCCESS in 11m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ee97933d27784e4ca2fb3aacee307794 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a50a33450cd081137b6e5e39f13aab437bcb2746","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 09:56:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50396fcac48148c3240e94c1321193741f839280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:15:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aa5205790374c1588a2afc219834937 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ed4fb4da8543b1a941db551e76add6 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94a47588328d4eeda079709b14548b37 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fe0f5b5e99f46c2a584434fef784aea : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0eb0ed3d3343b69b5b717c67d270f5 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1f4c47785eb45a7a5405918c81e93c6 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd1085c0aa9b43d49f387a681e7bb341 : FAILURE in 3m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ab7edb4543149dfa27f850cc29e92a0 : SUCCESS in 18m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/647dfae727c245ecae2f21a2b4ba75b1 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/75c2d07d0a9b4e7abb74206c24e92fdb : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/42cc8d947a82473f90f564077edacf31 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/053cc63de91f48c2b565fd7b1fc1118d : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/75ad25f205e449d2864c3e2d3f3b7710 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21a8fecee8ba4ab5bf5649340b4dfdd6 : SUCCESS in 16m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/07e64e8c697340dd84e7506d13dbfa02 : SUCCESS in 2h 14m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b3eb7db8db24ccaacbc358e4d724ba6 : SUCCESS in 2h 14m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b0fb9e0796a4eb79256f012368d06ca : SUCCESS in 30m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4cc9baa568c64ce298cd73aa072654c1 : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/64b2337a6030479a83996e975dd7ee4a : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1cdbcc6e3f4041e7ab84b7cf26357c3c : SUCCESS in 25m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/78f6c826e2064edcae2ac33964a0b275 : SUCCESS in 11m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7bd94fe4061434cb37ab1437b9e7ba1 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5fc553b80c27de25cbc8af6936fa259851986395","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-23 12:22:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72fe6172658f0d014aa6db4f4b88a80911974924","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 13:05:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4de65b1fc3bd564966137880336e83ab426903e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 16:53:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1db8fe780144d39bb858195c5d4651b : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa02bd8a3eb4cc88eb574c5022f3b89 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d867157ef6654ace985a7c03bfc5a78c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf75227106584fd5b960a4e8d2bb8baf : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491c54e8ba6949b395e9f252dfd1473f : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb83a299432e402e9174651942dc32f5 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1abb3bae0a024f649bb6f647266dbe12 : FAILURE in 4m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81fc4faaabb34121884c7089229a5930 : SUCCESS in 16m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/47de7dbcba7d4e39b82f2bfdb03b7b40 : SUCCESS in 58m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f4ca9ecac4ac4d01948346324374fc6d : SUCCESS in 52m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4014e43bd36c4d168449d93a51760825 : SUCCESS in 49m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/901c5dbeed2c4773807f7674bf478e86 : SUCCESS in 50m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e225d051ad1b4035a161b060acf512cc : SUCCESS in 18m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/26763dc7fb58499a8c376b4e026d8123 : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5a7f0f7c27734ec486f0079887121714 : RETRY_LIMIT in 1h 47m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33b6fe10d4f6422cb1657d1d6e736879 : RETRY_LIMIT in 1h 15m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/87b480ec171c40c2b5deda5515e89692 : SUCCESS in 28m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dc164c7ab2d345f19841ea0496e9445b : SUCCESS in 32m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/88cf2afa906546609ea63b48c5e436ad : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/84067a39f00f48b9a9d0842f21eb6352 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c64a6a08d68d45fcb238f5aad5c12fcc : SUCCESS in 9m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/26a3b58d825c4ed9b33dac032f2111d6 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"33fce680f425c63a20e333580a9fdaf244871a0a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-26 21:48:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebf0d40b9104feabe11b77c1dcf27b960f3a7a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 00:18:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a0583bf1dcf4d28b101743e178fc5a4 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0770a338ce14120b7e0efb7faf96a75 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1cb607185d7405dba7cb0f61a22de09 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91364db073644689bf44a7669b2ea0f4 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cb95b2bd7046e3931b55e760ca5cab : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6eaa5d1a07b64ac1b018ae2de5a2b12c : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de4d569efc4d4feea56900afb086af5e : FAILURE in 3m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71d5d94023d44283902ffc70142f7730 : SUCCESS in 13m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/39853eb3557c40e48cdd592e3f0ace4c : SUCCESS in 59m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c0afbf9196f742c78b08da9bbe3e3e2f : SUCCESS in 59m 33s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7f03415777194583a840a33b29303e8a : SUCCESS in 53m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/47fd52a221cb44158cce7508ddf611d8 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9f3b8279940e45ea8a4221dbf75050fa : SUCCESS in 14m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7f9c2e7160e24a7bbe79029e269c28d5 : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e46fca76d4e1452eaf8ede4d395e5bc6 : SUCCESS in 2h 28m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/807f4b4bc6364e62bd7aa5feb37989e3 : SUCCESS in 2h 25m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4976e44cf0644b2a819ed7566ce8eea6 : SUCCESS in 28m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6fe685b041af4e94ac321026c5675162 : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/37a4eadd298b44a693095f50f10b0e24 : SUCCESS in 25m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55bd7b2f3e384daeb4a17b252662f0a7 : SUCCESS in 25m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/41b7e5086c8f421fb208d45c05fd75b7 : SUCCESS in 11m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6604307ae1bc402e9d9c34c8944e01cc : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f84947c67289fad2c904711eec308608767f034e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-31 18:37:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4aa7e39e1d44a6b3ef37d2715611580a093dc723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 21:36:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5cfc0385244486e9a1e2aeffb491ef1 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d1eb3d106c4e24b727d339ae38461e : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b1fc3f015b491d9a3535e646751dfb : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4203904488a40e8b02c21dc4ffef774 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeb972fa94c94d74904d8f66035cd8bd : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/314d333fcdc6447fafcb6f118b742b09 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6071b08ef604b039b39b83c8b2a459f : SUCCESS in 6m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6d7c5cafc324d8599f5fc9f142d0ca1 : SUCCESS in 16m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ffc64bc4757047adb9d5ad5ca04da0e4 : SUCCESS in 54m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fd88d0b20b674061ab1bd21f33ab79ed : SUCCESS in 56m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3abddbaec5474a779e876766c192389f : SUCCESS in 56m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02bbb8fde3644a4a896f9b6790846050 : SUCCESS in 58m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/05e041e2868e49e5b9f19b0d5412342e : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cac8aa7ad01d4d668e78625a4d2ae881 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/150f4acee9384e4a83d41a6b5fc8b6f2 : SUCCESS in 1h 42m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c10831f80794fe29a9a3fb4abedfaec : SUCCESS in 1h 42m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab22d2c2cb22471e976c0c400823df35 : SUCCESS in 1h 00m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c4266ef92e146d3a6b28bd057eff20d : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cbf61326c5d24a8dbd257c8dd4121b1f : SUCCESS in 26m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f318d69062245d6a8fd354b0e6a0d76 : SUCCESS in 24m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f40e0452fd294227903296b5de832b80 : SUCCESS in 8m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/66598a2d5ac0404283e0554234317b69 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4ce15f4b2529b6df08a4f0c38b318f5aca84bd54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 21:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8db3be06a9bbd86c9b05e0a86e0431a222ab6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 01:28:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f87f8763c349cba81201d79a3bc2df : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90637e5aea04d14800e59671242386d : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d97fa883da84b4cb54b714cb1506702 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd9bc385d254355b7fbc98c2a1a1279 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0a1518f4dac4451bee23713d7ecb397 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f21716ae0d54b24afa9e6d49f7875b0 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c3ffe37a98242f6a5cbe98bec99cdef : SUCCESS in 17m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fef120291097458b9139b3f7df398edb : SUCCESS in 1h 21m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c82862da9811469bbeb9ebd5cc6ebff7 : SUCCESS in 56m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/23da783de8364c6bafff0f7c74f4f076 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/25b24b22128d4d4ebcde2bf507fd93f3 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/457a78b833a74027b0666417f024f328 : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c11698c881a64f74be90e071f1d0f748 : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2afe4e27b9b9423785aaf1870506b871 : SUCCESS in 1h 40m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19fb17809e124ff1b524547d29be01d6 : POST_FAILURE in 2h 23m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1dc864a3cf2a46b3a3a6c3d170f4f5da : SUCCESS in 31m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7030aa777f7a43a596400d9e6900dffc : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/524abf78ab4c46d8965fe3c0ff660ba4 : SUCCESS in 25m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24357c6db804485783605e4c546e5028 : SUCCESS in 25m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d5c8b1d1135b447f928cb7cfad0f96d9 : SUCCESS in 10m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/76218dee432c43d9b20d3fa0b7cea1fd : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d2636baddf891f302a0c91cf77c0c75053da692f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-01 13:38:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94f52034eacbdcd34a67c4f8d1344918d2cbd439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 15:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/610f88ac41c941728f58fac6a6425968 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d18f3bfbf442eca57da457d6de1713 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79aa3744dc954f5ea0d3b9b031abbfed : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93615c8b1464f138a4dec85a564c29b : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95ece6a2d59489e96e71488158b0d07 : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8dad3f7104304187a45f1b6115911f60 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28bc32f9c1b94cc4bcabea352a870d82 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/521a54828fa640aa9d67eb75436888fb : SUCCESS in 17m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9047b270fe5c4a21b6adf4ee6a68146b : SUCCESS in 50m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ba43d2eae0954a76ac0d526664ebd336 : SUCCESS in 54m 06s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9937f34bc5d548cab016b3be7b7e26f6 : SUCCESS in 52m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/665c084de136448989d2ec5613951419 : SUCCESS in 57m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/72ecf81107ce4fcaa1a429e793c56288 : SUCCESS in 14m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9cca08a63f284d1ea64f5e9c55d3dcef : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6571712f8a244e9f936949368fd81251 : SUCCESS in 1h 49m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5fd365c7654940558abc2c0619631f98 : SUCCESS in 1h 49m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/37f6428a055c4125ab5dc79833d0e76e : SUCCESS in 20m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/76fc4de3c3cf4685ad4122c11594f553 : SUCCESS in 26m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b302f3138e7f4c7caf6aff68cac4727a : SUCCESS in 25m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/268ae441bcbf43cdb70b9ea152a5727a : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/67191ad02a5f4ebeaeecb0800a1c124a : SUCCESS in 11m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a9cc8a790a54041afbda07931b0fd1c : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"94dc7642dfd3d4a0d17232ee78950c0d966d48e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 15:31:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e748c25fd31349ac455576e6029745acb66cc30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 18:22:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773bc422711a451fa9f264b6641a5f7a : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6f3e3c6e8748688bf03edf3a4b98d9 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cad0805a0724e9f9444f0b7b49cdf43 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2011a61108f455b9cbb3ddd6ade1712 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bf6aaf3d4cc404bad6aead83994ec5b : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2f589b9dee1414f816d9ef64990030e : SUCCESS in 6m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edc176616e514fc08a643eb37e836b89 : SUCCESS in 17m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4396547f6c7c4ed181f97352a731d36b : SUCCESS in 41m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/07daa1e888244c2b90bc7cf360ba5463 : SUCCESS in 49m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/41b8a4233f2149c3bcc162c76bbb4152 : SUCCESS in 42m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ceb545b98de54d01929ac059e4ad5e50 : SUCCESS in 53m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8b42ebe5f71c45289681c1b547e68946 : SUCCESS in 13m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/23043f4b92d14aa28475831bcead8f4a : SUCCESS in 12m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/acc8bd03a638441283bc4f37745427ee : SUCCESS in 1h 28m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/354ccd145a9445d0891237bef92068f5 : SUCCESS in 2h 24m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/045b19d06b584863a45c931173e00acf : SUCCESS in 23m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5f20b8ff61fe499a8777ae3386037766 : SUCCESS in 25m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/82682204ec86491aba13aff22e43cbff : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3a33c3b4174e40c6a91ce30179e97394 : SUCCESS in 25m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f50df1e7f0754a949891508548b3273b : SUCCESS in 12m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0e5cc4cf776d458e8dd5b6837f8a66fd : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ae3b9d8ea69a70ff162ece1b7d4d26305870f520","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 18:22:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"713c36feddf300022ac99adca0bec67a36084c29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 18:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d72d93f66d649199009cb9ee3130aaf : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c048c257ef1e8c7c17eb1287c7ac3f9cddbd0c7a","revisions":{"c048c257ef1e8c7c17eb1287c7ac3f9cddbd0c7a":{"kind":"REWORK","_number":1,"created":"2021-11-25 19:29:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/819372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/819372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/819372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/819372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/819372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/819372/1"}}},"commit":{"parents":[{"commit":"7bade28f73043d5f19a87724199aaf73ad3a79c9","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bade28f73043d5f19a87724199aaf73ad3a79c9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-23 23:14:55.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-25 19:29:33.000000000","tz":180},"subject":"Limit ip-routing and snat to seed hosts only","message":"Limit ip-routing and snat to seed hosts only\n\nEven if kayobe itself limit [1] ip-routing and snat playbooks to the\nseed hosts only the playbooks should have same hosts specified.\n\nTrivialFix\n\n1. https://github.com/openstack/kayobe/blob/52d7523c0c3e6a6f7f321459b9d7fc055282057f/kayobe/cli/commands.py#L610\n\nChange-Id: I35788e69eeace41d5bcbf89922145f1c35f58ff2\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 8986173237a8491847ea4292a57384c8b4371c80)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c048c257ef1e8c7c17eb1287c7ac3f9cddbd0c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c048c257ef1e8c7c17eb1287c7ac3f9cddbd0c7a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
