)]}'
{"id":"openstack%2Fkayobe~819384","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I9488a72db588e31289907668f1997596a8ccdec6","project":"openstack/kayobe","branch":"stable/wallaby","topic":"inspector-dhcp-range-netmask","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2021-12-31 16:41:02.000000000","reason":"Maksim Malchuk replied on the change"}},"hashtags":[],"change_id":"I9488a72db588e31289907668f1997596a8ccdec6","subject":"Add support for Ironic inspection through DHCP-relay","status":"MERGED","created":"2021-11-25 23:52:02.000000000","updated":"2021-12-31 19:38:48.000000000","submitted":"2021-12-31 19:37:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819384-inspector-dhcp-range-netmask","meta_rev_id":"20090432532029b9de8a7686452b29798f491f7d","_number":819384,"virtual_id_number":819384,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-31 19:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-22 12:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-09 14:07:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":2,"date":"2021-11-26 08:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-13 15:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-22 12:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-26 02:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-26 08:59:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-09 14:07:01.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-12-13 15:44:24.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-12-22 12:04:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"fc63f4f1868e904850b480b0cbbc46b92130e7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-11-25 23:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc9fcec11d7b2952f13f5b3191ec839086f3cb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 02:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97447d2736914efb80b567d10081c1ed : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05ebc82307b421f912ec97f82ddd702 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ff1d47caec431ab8a07edbc053809b : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed48c7c523e43f08c6a2f37ac9bee00 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e072eff8673488d911e03790724390d : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffcd53c9ceaf4f889ae85be3950d37bf : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7be7a8be5f684573a1722ac70538dea0 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f429d37477ba4de097d38d302b304aad : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/083594fb6c524d56bf830888ac9b3eba : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/214f735713404f5ca0fc658a1ccf6a25 : SUCCESS in 14m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/80471e92bf744e03a80adc266ea1833b : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa28930373ef41e99a2b2274ecd1eea7 : SUCCESS in 37m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/440b78b337da478aa050f13cbbb8c00d : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9d315438420b452dab34c4d5209b3da7 : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/461824c03481430a9d1f64f25e79a513 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98bc76a7d1304168b74935ba901d8666 : SUCCESS in 2h 15m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/516f27bddb0d4b1989ea3f2039dc2490 : SUCCESS in 29m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5ed39524ac0486f93ad90919b76dc58 : SUCCESS in 33m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a97aee8dd714952b67034faaefbc1c8 : SUCCESS in 27m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/32329d2327d5454ea71687951470ef0b : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f3da1162743410299d493024820b846 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2aa75fd1725d629891598b04a2b4a6761b759cf4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-26 08:59:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c46056bc570c8a522f615ecf0d497b4b89400c94","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-12-09 14:07:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c8c15f512eaa5159b72889a4b684f9942682466","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-12-13 15:44:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d8882e7d21f955a835785b2d86ac7674220faee8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-22 12:04:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7762f21ffb2020e9f50cea45b81fed025a85c4c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:05:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"314cf94f524c084e59d303595ea43962fac3912d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 13:06:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868dc88cb45747fb90540eec03db05a7 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9487ed6d6ce4ada986da5595605ac43 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b58db1fbd2d44781b83dfd43df2f0c3b : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f90556d0524f78bb71ecdd89830ca7 : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0891421a1e254b06971d761f1bc7bbbd : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bf107f7437c4e7cbd9039e0a3f53ba2 : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e1389f9704542dc847e133b6ad4c645 : FAILURE in 3m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8dfd8151c77d4d4488a40f827880691c : FAILURE in 9m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79435f53fbe84a0f9d13280c19c7f7e9 : FAILURE in 17m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80354944141e4a64adb35efdf1117878 : FAILURE in 14m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8da5578142dc43e7a7fdbfc24a7f37d6 : FAILURE in 16m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/076467d0ffe54a959c54c18a10cf95ef : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c92eabd3dcbf40c7b683b8f6214cbc9d : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3600cc5177d142ed83616dd0a76875b7 : FAILURE in 14m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/093bb06d03bd47b888263a1e10308e58 : FAILURE in 16m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45388cfb47704cfc840aa0d7dd74afc2 : FAILURE in 11m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf7dc9294690467e85b23e84f821be29 : FAILURE in 16m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7f110da084a343b58cc78f4c9dabdb9a : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dd08915c8c64c06af479e73f2acd00f : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0b2a76adadb703acaf33ddba04b0f267aafec407","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-22 13:19:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93307451d662257b25df656649d73794b48106ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 13:42:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d772b2de6cef4018a662ff2cc3005255 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451865de8e76451994ff4b78b6f5d5a8 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b642b35da2d54d24bbf8edd24603c35d : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7af34d45015f479eb66a46077e4c9b03 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093e84b30a3341d1b67aaae97e660c68 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d56fddd054b40d4943ae4c56e27bc35 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b16132b8e046f0836e7eedd18e84d4 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bec997ce76f647a49f9d1da218725a96 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b05b749bba94d7d8700d301bff79556 : FAILURE in 4m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c31f4b8a826497da4e651411acbd28d : FAILURE in 7m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6b39233556d546668f795be0e2d7dc75 : FAILURE in 18m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71b065b17fe842c98489918c840404f0 : FAILURE in 15m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/269d2c3fb28041ab9b9a1882129c9dcc : FAILURE in 18m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/326f10c1267b415ca08f500c8c00bc23 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f11235802e4243149046f829452045dd : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/755151771f74451e838667bddf953e6b : FAILURE in 16m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/63f4dfba70424157bdd9b8b125f000fe : FAILURE in 15m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d189bcc3004a47a2bf9076bc9c8fd0 : FAILURE in 15m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c3b8b86f3e6e44799549595e5d16c21f : FAILURE in 12m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cd55c2e161614880b06c1564ab510fc8 : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47c9e6f9bebd48d59ac929a071080f4b : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"345bd1613d05c203c80f3b681d49009ef61aa603","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-23 06:23:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"029453fac15910be1f95e75d01dfaf7e99d5447d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 08:08:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6c1c43c0024778a7e2c57c1c9b3df9 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2173d336a741efa0d7387f9531d4d4 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec5af1a46c064759ac62c61e9f3b4791 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26ad2e82f05f47a89555c1bb61c653e4 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a0c4bc7b6144498adec665a16e9afc6 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beea70c458704e67a186c52f95b73cfd : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce54b56b9ee4bd7b205359f9d79ab6e : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7566b081e80b46b6a3ff14e37bfdcddd : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4860312f1c754270b2d55cf3c6b12c69 : FAILURE in 3m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ce012f9ee4248f5adb80ac1401f126e : SUCCESS in 17m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c94793d705ce43da8438618c8938a994 : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d21b1b7677524d2e9c024ed95dcd963a : SUCCESS in 45m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a5b61a0d2fd74f84a467ef5d70b7b1cd : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eaf0996b7e6b4c27864919fd22aa66a4 : SUCCESS in 22m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03de3b4c69a84dc79ccfb95966b74f44 : SUCCESS in 13m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c7d6156242044c0e85eebeb5d559d595 : SUCCESS in 1h 41m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f7590c555c8f4c5f9fa0589c65216d66 : SUCCESS in 34m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e155fe0f334a44dcb1a9e00bbefce5f7 : SUCCESS in 30m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3291c3473f724275afe03312efca7183 : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/74e4d43ef4764b788413193d3a80c0d8 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec62e728570647eba9a2b91681ba833e : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8768fd71e570fea30a446af081b02aba9f6f419e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-28 14:46:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a221151a9ef8599f935872e82cbdade6d2b1de4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 16:21:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cebe71be85074b49a01fda7b0f46f294 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d36a61deaec4cbeb2954ec2b898c334 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc047480f3e4fa281fd7268399edf57 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/820614188c814c1081989266019ec2dd : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/272ec900cded49dbabc3c56e588a4f91 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23af5d796984493918fa6364db5a907 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38a6dea095394da282d1bb96f45c1c7c : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffce39e4134446e8bb0a16deec140d28 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7d7658dc3004c1f9e38cde0e7349bf1 : FAILURE in 4m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4af806cbba714dc29c0bc046c92c730a : SUCCESS in 21m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/53ea1ac1d1d94749b4ab7db86106dc68 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1785ea97528342eeabb3b77fc9401d02 : SUCCESS in 38m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8801790fd9c8407b92f475b84c9562f4 : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/568726d61e224eb1bb777f44623d04d8 : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a92e0959dce04aeb875e54f311268f02 : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52e6f4d79fce4cc4b757f99e9f8e34ac : SUCCESS in 1h 33m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ffe8c1cbb1264c9ba5d3ce7d885399f0 : SUCCESS in 43m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fec4a863d354fff801a5634282f8a81 : SUCCESS in 42m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/01ae7cdb3f524920b26c7efb2c4ea0f4 : SUCCESS in 25m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/76eafb27b87244288ef12ffd941b3c53 : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a1bb4218933419a93c1099ba85c3cce : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b16d41eb50b98e014ae259e7604f6965ea16f005","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-31 16:41:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc012c30a0f30c32cb9d655559fa9a27dd8a71a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 18:21:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d172024d184968ae68a7ad19cc35b0 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199ec5bfe3594d91b561af0f50784581 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b47358adedb4817bcb2bace0793fb20 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffbcebc7ea53444ab66132b09252a7a8 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f35ed0997bf4999ba6374440710791b : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/595ae1693c3c403ab7c82073467309c0 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f4a979463994c0db2246f7a03cc6fc2 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9206bb37ee394c8baa0827a548f9cae6 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1faa92d52348452f8c2444937c05e7cc : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a1f913a52eb4cd3a9689ee340556ce2 : SUCCESS in 17m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/160d946f113e46e2958a8beea629c73e : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2f433d6e1e14941ab9e683a80ed5ed2 : SUCCESS in 43m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/121e79196bb244cb8ae4eeffbcd7630f : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/18a2bc1770d546a59b45f72cb98d12f9 : SUCCESS in 19m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03bbbfa1bf4e49929d9986deb2f550bd : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/87242ebeb7ae498ca1f80d0e281b0dc6 : SUCCESS in 1h 40m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ccc75bdc8a949f88d5fa8d9eafe5efe : SUCCESS in 38m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2da6f7b724b494ea85e652e14d3bf7e : SUCCESS in 21m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/daad02339842456791bd9c3792b0499b : SUCCESS in 25m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/10342eda2ed843838f926b1a722f5442 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407c6218582e4204a61cb90a51382106 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3dc51f1807628fce9545468d1380d320c80e034e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 18:22:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebc421f5f144871aba7e6a5a51efab9ecd66bb25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 19:37:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a26a1186454efb8a40b04352bfa55b : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b108882b096f4a9db7e71311dd79a4ef : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0bdd6aa328b4482802808d9b098ed11 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578a4d73011d4be4baf5f3472a767e27 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db1339638b0b4e0eb4f09519bb8d61fb : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5d9b47dbd9647709e7b9f72157ab979 : SUCCESS in 7m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c77b37d041a419da23e98972900184a : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/573d5cb1c59f4798ab4a9a873c55e7b4 : SUCCESS in 13m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ea03f28a2fa744f69dca582fa55f295f : SUCCESS in 48m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e30d63908040209f5716eb1022ea65 : SUCCESS in 35m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f3dbd6121191470cbdf7f1601e72f0f2 : SUCCESS in 48m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7654a7d8186c4728ac511725f5704206 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6405b9e84aa1407ea9312236534b8f17 : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3449e0df98514b5c968b82761f2c0f10 : SUCCESS in 1h 14m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a30458be202342ca8bf21d40a8f28922 : SUCCESS in 34m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10630009ad0c485dbbd7026428b7f820 : SUCCESS in 22m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8b771d94227645a494c3a7fcb9909acd : SUCCESS in 19m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/829c68d4760945c7b3a31409b9099446 : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b1992f042324828ae5ec63ca88ae7f8 : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3f76a39e77532e32ba8970088277a3984ade8d4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 19:37:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"20090432532029b9de8a7686452b29798f491f7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 19:38:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195f60947fac4e2bb5c48b6edd03d36f : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5a1637daa9f4e71a5cde55d18a1aa32 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"679fa2a75b623687c74c819724384a0e8724ba3b","revisions":{"679fa2a75b623687c74c819724384a0e8724ba3b":{"kind":"REWORK","_number":1,"created":"2021-11-25 23:52:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/819384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/819384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/819384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/819384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/819384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/819384/1"}}},"commit":{"parents":[{"commit":"4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 23:35:36.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-25 23:51:45.000000000","tz":180},"subject":"Add support for Ironic inspection through DHCP-relay","message":"Add support for Ironic inspection through DHCP-relay\n\nThis change fix the issue when inspection didn\u0027t work corrctly in\ncase of L3-routed Ironic networks when DHCP packets received via a\nDHCP-relay. The dnsmasq optional netmask part in the dhcp-range\nparameter should be specified [1] in this case.\n\n[1] https://thekelleys.org.uk/dnsmasq/docs/dnsmasq-man.html\n\nChange-Id: I9488a72db588e31289907668f1997596a8ccdec6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/813268\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 02e01e983bc35193a5fcfe367a492c798cff0c80)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/679fa2a75b623687c74c819724384a0e8724ba3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/679fa2a75b623687c74c819724384a0e8724ba3b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
