)]}'
{"id":"openstack%2Fkayobe~819630","triplet_id":"openstack%2Fkayobe~master~I2f6c2f92903815973865ef0f5d6b867d5b995bd5","project":"openstack/kayobe","branch":"master","topic":"story/2002098","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-16 20:31:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f6c2f92903815973865ef0f5d6b867d5b995bd5","subject":"Add overcloud_dib_upper_constraints_file variable","status":"MERGED","created":"2021-11-29 11:50:01.000000000","updated":"2021-12-16 20:33:39.000000000","submitted":"2021-12-16 20:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819630-story/2002098","meta_rev_id":"e9254cefc503f3b538466c72f7e434ab7bf4f339","_number":819630,"virtual_id_number":819630,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 20:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-12-16 18:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-12-01 15:00:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-12-16 18:43:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 14:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 15:00:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-16 18:43:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"1baec2b66899bd0004990c47359c173470bde9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-29 11:50:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"500a240a3454676ccd6aacfdbde47dd1bf1bd5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 14:13:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6c8a6ba549943fe9c6070feb0d370db : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c061e0a2781410e8ddae78b8de2f766 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be1faf3294d2492b9d541347c0a7b92e : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69adb5a3fc441a190e967cbb5cb832d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd4d381591a416a8d74b7b697f72b33 : SUCCESS in 8m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8f27a6ca41d4bc09606d7030f68c9a1 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/522cd87d14734bf3b35a383b263ef5b6 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5b51b130acf4897b9b942bab80e6ad2 : SUCCESS in 29m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/211858253c4f4814969fbda361e23c02 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de3b2053f8b8408db7292a22506831a0 : SUCCESS in 46m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8327cb281b214ac1ab60a631a3b446fb : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cd1e5b984e4e49b6a04930fa34e4197a : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/569c795caa494112b17d5c64f68da2d3 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f93cae529584e0c9efc420e8b4031db : SUCCESS in 2h 16m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8e13c67be5a4d2dbb229aa52e07cf5a : SUCCESS in 1h 14m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/44a34122b2544d2bb5786eb177f3cb33 : SUCCESS in 46m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ac0b065c8874d9dbf21d6dd0c363afe : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7dd8435057f4a87b002fa26cc951331 : SUCCESS in 27m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30367b9aea254c0080346c98486f160e : SUCCESS in 23m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6ba792d64dcd489e93ad7c3b43aceba6 : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b5d2cce1bc4496c95b8cc4101b3574a : SUCCESS in 10m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6acdf9de5c7944f9ad452f22d21ac00f : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cc920662852482db5200b17062704e0 : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f37bf968b68caefa644fbd6a91ee67d6cb695165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-01 05:28:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06082c9ea0b5b3d7e3b4a0faa564eed4b1fe71f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 07:26:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d40a77b6391a41eea7b8f0ca3bba5da4 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7155e104cfd429d95b4799322a1e104 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248e1b329e824f2081f29652f09cd2b0 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/556351a89ee54a859a4e19992061acca : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9436f14bb53943ca98504392012cb0c6 : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b1066a7ef164058bd29c1c7665ff086 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/872308b679724aa2b739d2dbbba8c392 : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f5f044bcd24441aaddd1a3651bd6ab0 : SUCCESS in 29m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9d32d227409d47288046cb7dc1817336 : SUCCESS in 1h 14m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff8d68d52e1a4097b2c5345a9aade0aa : SUCCESS in 37m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f4c8b6efd6c24fbdadf8c8710f1805e3 : SUCCESS in 1h 24m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/59391dc5b6264a64aa7da7f197571b15 : SUCCESS in 22m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a1a3657be2a45a6a5b72dd5fe6d456c : SUCCESS in 12m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9eb2dcb58304f39a0b5f16884e06d0e : SUCCESS in 1h 52m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cafd9a8a43c8467b8dce742c0f779e21 : SUCCESS in 1h 07m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e7383ee7d9847b98a91b7051cfa3fc9 : SUCCESS in 48m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc7884371a94344b3d9a32d5e9d6fa8 : SUCCESS in 42m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4eb191c5f37f41b9b4045d267b9017de : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32999f2246874002a95c6000cf72cfaa : SUCCESS in 32m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a5046af00010437390472e6b4157fab8 : SUCCESS in 16m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f082db1c2511493fa47b2739cdf17faa : SUCCESS in 12m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/34d6e2b20fac4a06b0a32acb423e7069 : SUCCESS in 19m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb418619b2404feaba55b882483c0e41 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"97ca38050f8ecc66e7220bc63b65b51049da2228","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-01 15:00:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"47aaf06cadb57f349dadbafb584254490bc747db","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-12-16 18:43:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"382d3db096787bf885deda816d1aee2901af05f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:43:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b0cc4cc27a1bff6cefe1cf64d4b30f4da51b1d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:31:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d547ac77342f4c12bf929e882fce95c2 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f84510d4ec447aa042989bd02a6f40 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1fc5da91bd247529224d9de762d29ec : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3e7a130106405a8e5a6bd4e15e2801 : SUCCESS in 8m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/936b4be35095495ba0fe91d409abe3aa : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0af4192b992342d3b75f566a292184a1 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcc2cd814e0b47d5b747f70d459435bf : SUCCESS in 29m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4f5643b9eda84c2292f5de9e1246edd6 : SUCCESS in 1h 05m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7faa8a353b114057b2dd3c9667986816 : SUCCESS in 44m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c3da751e82949ac95a84070840efa04 : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bfde4c9b22684927bf8fda31bed8c367 : SUCCESS in 23m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/110d922cee4d4f09b04c4d9245927a31 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1294904c03a2490e82d11bf53c8c6a2a : SUCCESS in 1h 44m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4d1716009d74c1985c7a816af743763 : SUCCESS in 1h 09m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/661fc9dd008441958930c92305b5d344 : SUCCESS in 48m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3198310b8a84dbf8d77cbedcc88d2a6 : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99c49968383c48588ffa4454e1d4cf79 : SUCCESS in 26m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3675e196bb60420bb7a66da8df62b019 : SUCCESS in 21m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/554ff1594aee44c69a2de308ec5e1bf1 : SUCCESS in 16m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ef6b85513d14618965bebf432ac7296 : SUCCESS in 10m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5f8f131cd0cc4fc38a054712adb9749f : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413a22c238b44230b0eeaa0bb8983fa2 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1d0c696d1cdb202d26e0deff035444be3bb6a2bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:31:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9254cefc503f3b538466c72f7e434ab7bf4f339","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:33:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754155fc05c441b89d00d014033ce0c9 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad3f6f10a2dd7078c69efc1418a2c70b05dc2a54","revisions":{"e226aff4b44de8ba0221ebad8e1ffd2be3640bfb":{"kind":"REWORK","_number":1,"created":"2021-11-29 11:50:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/819630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/819630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/819630/1"}}},"commit":{"parents":[{"commit":"1fccaec4b08f37e79d307f9954ceb94307a614c2","subject":"Merge \"Document variables to configure connectivity check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fccaec4b08f37e79d307f9954ceb94307a614c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-29 11:48:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-29 11:48:39.000000000","tz":60},"subject":"Add overcloud_dib_upper_constraints_file variable","message":"Add overcloud_dib_upper_constraints_file variable\n\nThis variable allows to customise the upper constraints file used to\ninstall packages inside the overcloud-host-image-dib virtual\nenvironment. This can be used when we need a newer version of\ndiskimage-builder than the one available in upper constraints for the\ncurrent release.\n\nChange-Id: I2f6c2f92903815973865ef0f5d6b867d5b995bd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e226aff4b44de8ba0221ebad8e1ffd2be3640bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e226aff4b44de8ba0221ebad8e1ffd2be3640bfb"}]},"branch":"refs/heads/master"},"ad3f6f10a2dd7078c69efc1418a2c70b05dc2a54":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2021-12-01 05:28:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/819630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/819630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/819630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/819630/2"}}},"commit":{"parents":[{"commit":"9f574402f38750c042a6a81aa7813571fdae3f4d","subject":"Merge \"Document that extra kernel parameters are important for inspection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f574402f38750c042a6a81aa7813571fdae3f4d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-01 05:27:55.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-01 05:27:55.000000000","tz":60},"subject":"Add overcloud_dib_upper_constraints_file variable","message":"Add overcloud_dib_upper_constraints_file variable\n\nThis variable allows to customise the upper constraints file used to\ninstall packages inside the overcloud-host-image-dib virtual\nenvironment. This can be used when we need a newer version of\ndiskimage-builder than the one available in upper constraints for the\ncurrent release.\n\nChange-Id: I2f6c2f92903815973865ef0f5d6b867d5b995bd5\nStory: 2002098\nTask: 44101\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad3f6f10a2dd7078c69efc1418a2c70b05dc2a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad3f6f10a2dd7078c69efc1418a2c70b05dc2a54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
