)]}'
{"id":"openstack%2Fkayobe~819681","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","project":"openstack/kayobe","branch":"stable/xena","topic":"inspector-options-doc-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-07 12:27:12.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","subject":"Document that extra kernel parameters are important for inspection","status":"MERGED","created":"2021-11-30 08:44:43.000000000","updated":"2022-01-07 16:30:30.000000000","submitted":"2022-01-07 16:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819681-inspector-options-doc-stable/xena","cherry_pick_of_change":818491,"cherry_pick_of_patch_set":2,"meta_rev_id":"965496600cf9a0c79eac4c27cbf6885465217089","_number":819681,"virtual_id_number":819681,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 16:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-22 12:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-20 11:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-22 12:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 10:39:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 11:09:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-22 12:04:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"18383eecd158f849d5060663eceea30f618c6727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-30 08:44:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d139d2088cb509179a8bdf91d11c2b5b938c65d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 10:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d21ed3844a4cf09423ff745a64b602 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0bf14e0f5d47cf85a4f9cf0c9fc53c : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7a23b55c8a43888c265b649f8da6c7 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9486ea02275c4bb8912a268b38392a04 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e010dbdd41f84cb281169b63080e9d1b : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57efbd16637749479edfbf86c7886e5d : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f6f245711ae4e73bcac554355c757a1 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa8f4be8cbc24c8f9bbe028b5dd2d126 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e36b5e51ca14e39bd34bb5877c6c735 : SUCCESS in 27m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a385ad9058d0458e944fc9592a2212fc : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d5ab80bb97e43f98a12e2376f1bbd05 : SUCCESS in 49m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/19971871fe1942f196b2eff75999bd99 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/630a79b76db248e48810d1a6fbad3e16 : SUCCESS in 21m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dfb1b5f1d56404fb063b6e8ee6e68aa : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc9c7759b7f74dd1871c1160a1d5b6ba : SUCCESS in 1h 52m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e769f273aeb46beb3dba1344ed3bea5 : SUCCESS in 1h 18m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c6d2e17856ee40938225ad354a2abf4d : SUCCESS in 35m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea1fea2aa9446148de9b1698fef9222 : SUCCESS in 31m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6465d2de4006432f9f1f219efb99f68c : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f81140dcc86477ebd59bdb415cdfef6 : SUCCESS in 25m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c10596fa56c048a68ef6e475fdc8b97b : SUCCESS in 16m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e77974783e464b0c9d1bcc642a9c414c : SUCCESS in 14m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b18ee8effd904083bbb733089b3f907b : SUCCESS in 20m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21df6c105626422c959036042047d5cf : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4ba9d5abbe234aca40a45a882c25006d9108b331","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 11:09:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f9ccc3b303d42725f455edd18a63de1f565fe23","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-22 12:04:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16af83603e47ebd69155206a04339f0d77a18e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:04:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"209a7588f1a1a021fa72c146cd30ec27320d02b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:35:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a86ae1b06fee4dc8b70fec8f25fd13dd : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c596b9756df44bcbb65ebfade03508a : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cb6c1946df54931ba506341366a73a1 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e15990ec92479e9b2b55aa002f2f0b : SUCCESS in 10m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b9f8e7747064342ac507f775117f6a7 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a5bd0588ce84d7bb976713fe6bf3c62 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ddf5138f1c5b430e97f26fb6a67b3677 : FAILURE in 9m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57e046552f3643028ae5d0a96c78bf35 : FAILURE in 20m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48e8c374b30a4a46a373c6abc52b0fad : FAILURE in 16m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6da28b3a380c45949a0ee86fb27c4f7a : FAILURE in 21m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6da52be05ec9421d9420f54907e907a2 : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f460ed98134d92a7da41038bc6cff1 : SUCCESS in 17m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54466095b9674b3284ac2cb87ae674f4 : FAILURE in 24m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e902273afc0048ff9871bf172d638a1f : FAILURE in 17m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0062cac1d51740609f8eeec214ff3b50 : FAILURE in 17m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9e26bcece7943ad97f7ed6aee6af039 : FAILURE in 14m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/47024eccd50548a2aa92ba48cb436070 : FAILURE in 16m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c387a92d9f54ef4b9d7d07954d2c3ed : FAILURE in 12m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1b858cfe4a5945ecbe2bbddd38e30c87 : SUCCESS in 19m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acd31bdf610e4b1280d10c8e7633ff4a : SUCCESS in 13m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/05fecb28b929465a958b1ec6dbd8f2b7 : SUCCESS in 22m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb06823789e43c58698be699bd39a85 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1b7751ccae9bca82e4510aa4c27d115507e633fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 09:59:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba9ba39e7bdece5af3d28846245dd8ec43e5f8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 11:57:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f4811fe7f1348c08d8dedd3fc41282b : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f40610a223e45ef9095e490dc6052cb : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f6bdfe26458423b8f76499025ea1943 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d6d86397e924aa18d6d8b9c48b3d9ec : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54d091f93a814e85ae3dd62d370abb81 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a8c9b85420946da9e71eb717cb12787 : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c5f761bbac9415b93f02d5cd4737033 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58e79d12f93c41498eb4a98a1d0a3034 : SUCCESS in 12m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac783a59a9944571a92b3b6876d03403 : SUCCESS in 28m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6b1229728ec54eafbb7a39eb74d153ec : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d626bf5a4544bf5a688f5e23245d509 : SUCCESS in 51m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d41c225af04f463d925c816eef402353 : SUCCESS in 1h 19m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5761a37d7eb94c629b5df1fa3c71711e : SUCCESS in 26m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/532a5b7c53cc47e58e4514b7434e29ab : SUCCESS in 11m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d300534b66a4cf59e541f4e1f755a3b : SUCCESS in 1h 51m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d5b0bc596104475b373f4f1601263c1 : FAILURE in 46m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/01dc5cc5ecc0401aa64b11982ea52d99 : SUCCESS in 35m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba00764aa6124ec38a0607c235dbeb04 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16a804c453ab4086a64a52fb7f6e9b47 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fad5f01ad11b40e9bbe3d998d3d51f3c : SUCCESS in 22m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/81f5c635ce054a86b14444afa7164043 : SUCCESS in 17m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70be237263154e5392f0887c824e77c5 : SUCCESS in 12m 44s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2f3943163d74ce7a67734570cc57f10 : SUCCESS in 14m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c301059e3254b7692831a52241a7566 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9661b566ebf8427f2f8ea9b27c25624bbbcc1704","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-07 09:08:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9d498bbdc50f6e8413d216b78415c47027fbb82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 11:38:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51063ba4902b48e3b1afd970890c87c4 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1c81745cb84ca886cb1bb15a7fee7a : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4304ed6bcf5440f90fae4854c80ced9 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4549968298b6436c94b44ef9a40ba939 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/930fef49d7e14c2e81b516128c3c9d30 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476fe4bd43104637aa903dc304ba18d4 : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7331394415949c796867c36b6289a71 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49f838b040ba4bfdb09da9afe232c015 : SUCCESS in 13m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d91e2d372c24fe0870695cb9669ffab : SUCCESS in 34m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/16404d54361044f4bd774741435bdde6 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f2d2c609a247ac9f0a21a4482fc24e : SUCCESS in 45m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02926e0c67c54b84b3118cd1ae3870a7 : SUCCESS in 1h 13m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e87237296cc54a049fe5deeb1c581e47 : SUCCESS in 23m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bdf2bf7d42b4cdd87ffa0e7249cc7bf : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e6dfaf6513746fb8aee7a2b58567962 : SUCCESS in 2h 20m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c04aae2095544ae4b98a46f780373fcc : SUCCESS in 1h 11m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c49577a4e664f2c8022643a6747e156 : FAILURE in 19m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5dd33d4692b45fdaa3c7c7bb9c7fd90 : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc1f43465185464cad092a737d264412 : SUCCESS in 29m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8742cebed8f4016846d468d2163d8f3 : SUCCESS in 28m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/019997e9104548b8a9ab2dbb2fe4e52c : SUCCESS in 18m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21b4bac931064b1286e384749f04a5eb : SUCCESS in 12m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/37390bc4748b4b1eb9fe58431a21d056 : SUCCESS in 20m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adb354fcd01648a588b3151e6b4c99c8 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2f6fe0133b0b912588c1a7e0ee01e2b9188e4104","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-07 12:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e6ae19dd27e82afd818cf39ee16b9ca9cc5a724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:28:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acfc90ba6d8a4dd6979184c87ea319b2 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa8fa87269c4b08b67e6ae2d1b334c9 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5060675bc83b4b8296dea3e682476f9b : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342ca13159c14342935c4e88d0a01d3f : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d009f9ec3d03482db0149e8ddc0a397f : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f60c2eae594425b9d29d61d147470c6 : SUCCESS in 8m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fb8a1ca437b4411bb8df336432807e2 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/374e11e1c1424d6cbe7045c3a19b651b : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e2092ca2dda4bb9a9a8cdf29cdfc3d4 : SUCCESS in 33m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7d79bf591c7b4f7087395d6435fc9447 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cadd2f31a434fc7ae47ab5e2331bb3a : SUCCESS in 48m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1109ec979a764a319139661a49709707 : SUCCESS in 1h 23m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8ecd1e474c0c454ca67606d469a93361 : SUCCESS in 23m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/398df497257243ce8a29a27f60e40f96 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8b6b7aa6c6940419bfaa7aa7ab2e599 : SUCCESS in 1h 57m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdbbb058de9f48cdbf0725aa00cea0ab : SUCCESS in 1h 14m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0088b5be29a44b07b95b3c42ea8198b3 : SUCCESS in 36m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d586287df9624d8591c6bfbc7920ded2 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dcd562c31a92470da8e6021c164d7207 : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2d25fc847a34799abc445acdc92dc43 : SUCCESS in 24m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c44a864cbece4341bdfef9c8681f90ea : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bc4b5af830d4fa595f948b29e583637 : SUCCESS in 14m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/23e321b1c8d84e4bb9814774e661b6a4 : SUCCESS in 18m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b7e37288bd40a284a978f3b5239035 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c1c6a4ee38cc01ab492a3a470bfcd8c598df098e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eb72e4f90affed6acc714cc02632413431095bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:28:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f41a18fff04b349a65804971a50c32 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5707d68b12b141ed8cd39e2592efe9e8 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d725d06eb84883b264b7f842d59ce2 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d116de43bd7545418ba61ad430ba8ffb : SUCCESS in 8m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d6bd2974a8742c997e8380b630421c1 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ef493b2f9b44c88a13e0c8ba66ffb65 : SUCCESS in 14m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b6eb7cb66654060b5adfadcbb5047d3 : SUCCESS in 35m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6d01e0b2bc3a4d4d9baa06e4017213b2 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8700e07aee8e437d88607a659161c1d9 : SUCCESS in 53m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2494f951c82e4646b7c53cda6364c38c : SUCCESS in 1h 11m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d534e415934c468ba2876e73d859bfaa : SUCCESS in 25m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e2947fc8c6495db89ba6ff73628f93 : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb31b5e7ea0a42e396242e0281f56cc1 : SUCCESS in 1h 57m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/262f1a8585d64d569cfa0db17b1a94d8 : SUCCESS in 1h 11m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0672186f30884c708edf84029538ba5b : SUCCESS in 36m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/703843af44444849a594f9d5b9f7b748 : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4368303cd28c4fe380cc43c6363aafca : SUCCESS in 31m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aed19915224441b85e8a48b486b310b : SUCCESS in 24m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1dfba0b1bddb46528a362e35663d4bad : SUCCESS in 19m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9352a1237e324aefb9dc7ce875d44b94 : SUCCESS in 15m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0fdd96b831134e8d8281ca4d54597e95 : SUCCESS in 20m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dd9c4cc3193486e9d9f320837311b2d : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3cc96dc5c3f831f55045c97312da3bbca0c9cbd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"965496600cf9a0c79eac4c27cbf6885465217089","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:30:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9ec4e008d34df886e5a89c84f07eaf : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d9b86e29a42663feac68e13a56b4d51e96a0b82","revisions":{"7d9b86e29a42663feac68e13a56b4d51e96a0b82":{"kind":"REWORK","_number":1,"created":"2021-11-30 08:44:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/81/819681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/819681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/819681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/819681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/819681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/819681/1"}}},"commit":{"parents":[{"commit":"8e6a66bf18a389b56d725eb2d1062e95e0ae70e1","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6a66bf18a389b56d725eb2d1062e95e0ae70e1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-19 08:38:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-30 08:44:43.000000000","tz":0},"subject":"Document that extra kernel parameters are important for inspection","message":"Document that extra kernel parameters are important for inspection\n\nWhen customising kolla_bifrost_inspector_extra_kernel_options, for\nexample to define which console to use, operators should note that\nKayobe defines extra kernel parameters that are important for\ninspection, such as ipa-collect-lldp\u003d1.\n\nChange-Id: Id789dede2d1886ef5ec66ebc86968d6cf62fa2de\n(cherry picked from commit 7ed91dd06613b36889597df0cd4a8792b43f1b61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d9b86e29a42663feac68e13a56b4d51e96a0b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d9b86e29a42663feac68e13a56b4d51e96a0b82"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
