)]}'
{"id":"openstack%2Fkayobe~819683","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","project":"openstack/kayobe","branch":"stable/victoria","topic":"inspector-options-doc-stable/victoria","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-03 21:57:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id789dede2d1886ef5ec66ebc86968d6cf62fa2de","subject":"Document that extra kernel parameters are important for inspection","status":"MERGED","created":"2021-11-30 08:45:06.000000000","updated":"2022-01-03 21:58:25.000000000","submitted":"2022-01-03 21:57:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819683-inspector-options-doc-stable/victoria","cherry_pick_of_change":818491,"cherry_pick_of_patch_set":2,"meta_rev_id":"2998e519b4fcfa5ce8001e007134aa33188fcde6","_number":819683,"virtual_id_number":819683,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 21:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-22 12:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-20 11:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-12-22 12:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 13:08:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 11:09:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-22 12:04:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e69abf2855d60a4f22bd4ab8579997238d0e2af7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-30 08:45:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2fd5635794d000dac50a925b9f03fed936555008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 13:08:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d110ece310904ec99e7fa2153852d75d : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87b401e69054ec98e6274e22e87bcb6 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c377d60ee8f943a88fc6f8753851ab2a : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e006564217b746fd82705aee91cbc0b5 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7443db3cb4a459f8462de7026e438f6 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a30605f9f66143b0819d30bf0ea27169 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7f38c3782574e1a8fccd907f7a26fd4 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd721e48485d4049ae45eb0693ff5253 : SUCCESS in 25m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3cf8f59b244a43b0ae375361f4dc130e : SUCCESS in 54m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/97d4771650014eab8f6282933ce91ac2 : SUCCESS in 55m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c573edab55dd4b87a8e02d51ba3cd424 : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6e5c90b96d4d4ab2b61efe74156f7eec : SUCCESS in 58m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/27dc5d4dd8624434ac44a714802566a4 : SUCCESS in 12m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/28acb70120c14ed68322efd28a3622dc : SUCCESS in 16m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/30db30d5f70548d49d3ddddc09f65340 : RETRY_LIMIT in 1h 19m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0a96e8a904e486994f11337a85342d2 : RETRY_LIMIT in 1h 13m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/abab08f28c484e21a8f33e6c6d06e46f : SUCCESS in 27m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4907b85ea8cc442ab1441168c7a1b001 : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dbff2122dde4cb98de539cc344cdc0c : SUCCESS in 30m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8d6a8f2a0bf94840a9d2ec9111de1def : SUCCESS in 25m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3ab617332c664811a4a973397aacd8e3 : SUCCESS in 8m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/18e3e84e4b8e41a28024c103055d4601 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"916633819141fa47e5a73772d10cc7fa78190da8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-11-30 14:13:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce679891b40359b20bba4d998f05b2baba4b181f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:17:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53a37d6a5714a59a12902d222b2b2ad : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc7b2a379d8b4cf093e9e424b63a2af1 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9efc5631a54964ae0bb9ed748f760e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c050e926eb04f7b98c10047d8cfcf1b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825ffb6af7ee4079bc68fe98314a6ad4 : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d18f7eb34a624da582b0564d3fb06867 : SUCCESS in 8m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76c36cf12b0f4c2094dd7525d1adbd5c : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b2978ccf6e2486d8c201e8e3a181972 : SUCCESS in 17m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/effacac531e84092a6ad7df34a927bdf : SUCCESS in 58m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/911e996f99b4401d9815f1053e7d96ad : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/70eb26620302492cb0edfdf70f1ccd79 : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c869aab488e249b98793ab9f1dc078ac : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/963cca4fa2954243b03f1f7f398642b0 : SUCCESS in 16m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/418a1350435948539b68e0ff1220faac : SUCCESS in 19m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3536287d9b624b2d9fa893863ce443cb : SUCCESS in 2h 38m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7aaeb3f5d1a34b82b0e97bf12a1bda02 : SUCCESS in 2h 08m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f0a4198aa83a4a28a36a05fcfe2fb288 : SUCCESS in 28m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/daf7d08af6f342edb7efd2b214140abe : SUCCESS in 30m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e0c9bab244e04cf38606fd900c66ad25 : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d4014515cca040cead8dc692bf4e1b6f : SUCCESS in 29m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/be30e34312ea4e1ebdfece4a45f9b9c2 : SUCCESS in 13m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/94f3efd60c824a9387c6495db8229953 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b75eeb72caef76ba5b5a6ab4b6de8f7a0f9bcf8a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 11:09:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"211950e1974f9b45948e6c14622438987b03ff47","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-22 12:04:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94ca1642f373738bedf18a0141bb27d70dc9f952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 12:05:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8576ec5d7ef4d321c418ad5e592866f18fd9c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 13:05:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da906dee5c940db8e62b92b58483870 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3825c902f06b41e3b941e6cd715b3630 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b3add9b6b9544deb34494d723012d2c : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a314a7286574282956723d6109e995c : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc509180475d4b80952c6a568a37ecb3 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/392b60b8305a4eba8acff5a41ad98870 : FAILURE in 4m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f95a334977034eb0bc4fe934204aa05b : SUCCESS in 27m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ea1616d2804e41fca408edd6e726108b : FAILURE in 17m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5e5234dd1aa24aa69cc6c7a8013e122a : FAILURE in 14m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/71f3a6e3cc2841ff8932bd57b1aae432 : FAILURE in 15m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5eb8fb4683064c86a796699fc409816a : FAILURE in 15m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4c5ef1db1d5e4cd582daf94fca009ed7 : SUCCESS in 15m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2b093ae409eb4d3da1df9df95035b2da : SUCCESS in 18m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7222b4021f2d4418b8ae559fa0eeb96d : FAILURE in 16m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ef0bd257248a4252b3ccc1fda0edd9e8 : FAILURE in 16m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e6770eee4f0045849e577e2d9e48df86 : FAILURE in 12m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/50eeac62baf1401d818175eb1d1a345b : FAILURE in 15m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b26fd08506b64d629a7e6d9196ae3cb9 : FAILURE in 14m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bca4c2567c2649f2bbeb79d10481e228 : FAILURE in 14m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/251d99e9939949b99341fc70d172352b : SUCCESS in 13m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c51c90c231e74319991740f15c03c80a : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2936a892a8eda3d04e4a6028afe6a7e4e2914672","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 09:59:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a66877eda3d61ef97c75431975f9bd53dcfb6bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:24:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e016cafc1047a282c3ea94c599d5e4 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdb67a7932b4619adece1418c9788af : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0394ac614424f338d8bf2e2cb4d565e : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f67d06122034d14add5119fd83b03a1 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309253b2c9a447648a791ba39361ed56 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74234bc8c6b24fb5973376ab101cc338 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1058051e951e4430b65f6677d7114a33 : FAILURE in 3m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc900e496c874280bfeee1264d2c22f9 : SUCCESS in 19m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f7004d80fdf343c3b5113e68bf9c2f74 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e96160a9b7854555831e0dad06025922 : SUCCESS in 47m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fa12dd8d1cd246dc9008a11c48ecbb0b : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64217e22e8ef466ab49ec6c58253fc9d : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d3c88a764e514eebb09a452b7b466b68 : SUCCESS in 14m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e741ab9919b4a61b1d888bd6e9347ec : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0c68eb43119c46d9a6737332c856569e : SUCCESS in 2h 22m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5ac010c73e4b41cb8f6f7afe48a22e28 : SUCCESS in 1h 58m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c586f8940cc645fd801c1a1ac0be6918 : SUCCESS in 28m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/30cd070a13c34d2ca8147ce5e9c1b9fe : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4ae60288f0344a1280e92c69f78de61b : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/defbf28c557a4d41925ae310a006f29b : SUCCESS in 25m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/de9e70265fa144dc8d01703210cee1fe : SUCCESS in 11m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b1986cf18a824682b40e582e97d4bb51 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"876ba656419b248322f95d57c26b762db15d3a1a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-03 17:17:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d40deb69194e5b2a2f4ffa9579d0d783fd3c2195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:41:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74052b0b364b47e89be0948ce8c91593 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c92d7c28c264428bf083e6f93b959eb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b1b52f2f4c349248bfa579c2a28a667 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed34c78e223c4b09b56063483a930184 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b843e1e59cd49a3a70d831a2a498bfd : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4603f0c8a8af43499f0cc51e85d7ca87 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25415f0646b54146bb8c5801c729d278 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25aa43d3d1b44c10ae72f28e047b6af2 : SUCCESS in 16m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ce64dd610914153914c83cfca420fc3 : SUCCESS in 54m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/84ba31db3d7d446c9a87e285695434e2 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f1f85acb05894f1095740487dd2890ec : SUCCESS in 1h 17m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3f97edb592b74d299b59324b20a19198 : SUCCESS in 57m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b39772fb7c4d4338a06ef3074d655dbe : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6406d5825d724bdf9fb961685749b562 : SUCCESS in 16m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f38e92be64214305b652a141c1c2f356 : SUCCESS in 1h 38m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2264bdde241d40d9899189fc23385753 : SUCCESS in 2h 22m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ae78a4df7e9a4ec8b012aeb9b0430a5a : SUCCESS in 27m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a306ecd84f244cee8bbb657e3bf3a376 : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a74f529e940941a290b76fa19a2999f6 : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b543c28b1cf64d6e99f743b57fdb195d : SUCCESS in 24m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b5122be68d39430ebcdced1bc2c8dbe6 : SUCCESS in 12m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/db832bc8160648ff9cb51db8ea039330 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"09ab3ad9d300ddd0984f73510a42271d341ac58c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:41:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"414df2b166212ca02f4d1c3a3fb049eab2d20eb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 21:57:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b92d2cf98f4f8bba689f54b827458a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d159ac448b4c16a7760445405a7a51 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe44df83b5042cdbcdf099fa43c5714 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34b88829ed7d465ca98a8927242e3e29 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/895a8d52e9c944f0b9942a44f67ed01c : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4056fa8e1a647998e9fae05e894c815 : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ece87b771a244b7c8a7cac005f3a560b : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a9abef78142b4214b61d9e86feb4e235 : SUCCESS in 57m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eeb275407668411ebc00c98f74510488 : SUCCESS in 59m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a3cff20ddc9e4ae9b5bc594d7d566cbe : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9756a31039cc4c38a41bf29802cf517c : SUCCESS in 59m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/33278ee74beb43e692c0d950bce8c2ab : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f723b7b645584749b86ca96ff474854c : SUCCESS in 17m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d45165bd330c44f89f5edc37d071509a : SUCCESS in 2h 13m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ce0fae8f1834a958227f415fabd43a4 : SUCCESS in 2h 10m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/855d31a8bda645f489292b242df4b19b : SUCCESS in 27m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c47fe05222e14cb49bc9f07a8c993101 : SUCCESS in 32m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3853f56fb53b460a8c0b9ae009adc981 : SUCCESS in 26m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a78ab92dce3c49229b1101e942e9df67 : SUCCESS in 22m 41s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/629670a9a5e846ce85c7749e1deb89a8 : SUCCESS in 10m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/586cb68dc4dd4ee0af21977b40e4892b : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6af35aaf2f2c0c04bcd602b6cc3cb5a0675f00f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 21:57:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2998e519b4fcfa5ce8001e007134aa33188fcde6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 21:58:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac69dc6078744838b8008fb60070dc8f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be1d6adbb105c2995888497181264ba44b89c394","revisions":{"be1d6adbb105c2995888497181264ba44b89c394":{"kind":"REWORK","_number":1,"created":"2021-11-30 08:45:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/819683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/819683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/819683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/819683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/819683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/819683/1"}}},"commit":{"parents":[{"commit":"7bade28f73043d5f19a87724199aaf73ad3a79c9","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bade28f73043d5f19a87724199aaf73ad3a79c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-19 08:38:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-30 08:45:06.000000000","tz":0},"subject":"Document that extra kernel parameters are important for inspection","message":"Document that extra kernel parameters are important for inspection\n\nWhen customising kolla_bifrost_inspector_extra_kernel_options, for\nexample to define which console to use, operators should note that\nKayobe defines extra kernel parameters that are important for\ninspection, such as ipa-collect-lldp\u003d1.\n\nChange-Id: Id789dede2d1886ef5ec66ebc86968d6cf62fa2de\n(cherry picked from commit 7ed91dd06613b36889597df0cd4a8792b43f1b61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be1d6adbb105c2995888497181264ba44b89c394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be1d6adbb105c2995888497181264ba44b89c394"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
