)]}'
{"id":"openstack%2Fkayobe~820211","triplet_id":"openstack%2Fkayobe~master~Ie9b15db6563546ede9ce9735292ec566d540432a","project":"openstack/kayobe","branch":"master","topic":"story/2009721","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2021-12-03 10:55:26.000000000","reason":"Will Szumski replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-06 11:04:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9b15db6563546ede9ce9735292ec566d540432a","subject":"Fix installation prefix detection","status":"MERGED","created":"2021-12-02 17:32:48.000000000","updated":"2021-12-06 11:06:34.000000000","submitted":"2021-12-06 11:04:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820211-story/2009721","meta_rev_id":"c6f92f8c00194513d9c7a24873d522cf37f2b106","_number":820211,"virtual_id_number":820211,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-06 11:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-12-03 10:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-12-06 09:25:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-12-06 09:25:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 17:33:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-12-02 19:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-06 09:25:39.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"d46208372e0cf6a52f40ddc7bc521b628db94963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-02 17:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a98d9dd9868dc3ba6fc732314302cdd31d6b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:53:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25a7a2e49a6943229f1cebeb7669b74d : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19868777ff77400081dfb7e9bbb9d74e : FAILURE in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9797e4fb4bba4422beb4bca6b5b528a7 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fabe915e421c4cfcbb1bbb29e760bd39 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc464ce69b9045bd810c84b1d1f88a88 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aab76b66b094f279f697574d6b943de : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7225e10ad1704570bfc6e925511a8e6e : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce52c57aa3af4aab803b028d27565f15 : SUCCESS in 15m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/396745a9ca17493199fe2770e67fbba5 : SUCCESS in 26m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4eec6ce2ffb04da18f4e9bbc357a8ee0 : SUCCESS in 57m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6c2dd1d3b774c4d8db42b0db7f7e1fb : SUCCESS in 49m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0e7012baadd04028ae0ed88302ada793 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/faaa15bd1b3c46f8bf9fd104c953f71d : SUCCESS in 24m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83ed69bd8fa4b0a831e09f997f6694c : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/391e03b13aac4bfbafcae27826e347a4 : SUCCESS in 2h 13m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42bbd5fa48540118245524ebd23f978 : SUCCESS in 1h 03m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/58afe7619d654bf7a4732a66131cfca0 : SUCCESS in 52m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc26b214ac924442a2c2ec0c0f85d3fe : SUCCESS in 31m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56c7d95163834ff1811179cb73245ac9 : SUCCESS in 34m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c5156ec33e541a892aa5aba136013b9 : SUCCESS in 23m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/319a45009f714949bc9cafd94f72697c : SUCCESS in 19m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e29bf3247e20474abe5a272d8b4e9909 : SUCCESS in 13m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a16a695c8f9c4e60a8c3b7c64bd344a1 : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860f8fb19ca94d42afc14ae1066bfa9e : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"75cb48ea548961ea6b41a9ac9638eb14a69d1aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-02 21:44:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0929c66c0c6a0443ead8c3c9360ff39ff0433e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 00:09:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f295afc031154f299bc7b8b6f6fa7e0f : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683af8030c1f4bfeb7741876d09d0a6c : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb92644b95f4b6e9c428becb8b3c5a9 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09246103e60e4404ad89e7ad9a73e80a : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812d1abb55f146f9add6771b7e5ff473 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8962fb5b03fe4afeba2b3e4286e308ba : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/44dacb3f7f3441c9a8b03daa4f90c384 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/828e65a4be654cb5bfab17a804c1fd38 : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0e27d0b3a074cc2adcbc43f649bf265 : SUCCESS in 34m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/000477643be7442c861becbee55ecbdc : SUCCESS in 1h 16m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f751533138774cb7971b7cedf52fc874 : SUCCESS in 46m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a958114e72324b13987ddb2191925464 : SUCCESS in 1h 24m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b2791f89999f479da8c7c1de61f95add : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae583dff99144ae840a3ce6185ca78b : SUCCESS in 15m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe811ce36d564390826117c50abbc124 : SUCCESS in 2h 24m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c464ac46e046b4aa0ef6801879b9ae : SUCCESS in 1h 16m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/291b4c276ddb4e7b8845290ea0b85ef3 : SUCCESS in 47m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf8341d9aec4c5dba74419ad4d7b0c9 : SUCCESS in 44m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c5b72e6750b745a5bed905d8afd26204 : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d08430b4d78642d58cc5cdb396e5eea1 : SUCCESS in 24m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a318ea15d954eb3b0b36cd044812c1e : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f75cc08675f463c8665602af880732b : SUCCESS in 12m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/495a27c20eb04bf09e80f7215e2bc225 : SUCCESS in 19m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1232087230664101bb20343322d8d81c : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":2},{"id":"62309c9879c8e8363dd74c69d8a8f9ed4eeab382","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-12-03 10:55:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5099060218ec954ae5ed5f1580fa85eda2247829","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-12-06 09:25:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59a0f76be404f961a0d776f5fcc1f298ae7234aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 09:26:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c47a443bbaa45db902b05845a400a499a172bca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:04:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23621c8e93294788abea483bb65c18c2 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c1fef84fb2344b6a4ff4de7d3c77242 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d50c5b900044153a1fa1e0d563ce42a : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c67caac3084821b853e3ebc7c4261c : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d70d0275abf4a5f85cf2c447ca3db62 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e3439789d8345f486ed569dabad7a4d : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3117803adf8147c39c9ff6d103158aeb : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14d81e5a565848b6be3fbc68898b1f19 : SUCCESS in 26m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3472f004cca49b38126879d4603898b : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35d067341e374c828f78cbbdf9c32712 : SUCCESS in 40m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/801c7738824d4c78af0205a73265def2 : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/75715a6cdbfe4c2a8c50b479fd1adbb2 : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685e53f26e4e4de2a41234c5fef19448 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18d53cea3ab849c48f099fa22d69e6f9 : SUCCESS in 1h 38m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/968f7382eddc4639aeadfe69d44831c8 : SUCCESS in 1h 12m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bed39c6f24e14d8a81d9705462ce3cef : SUCCESS in 51m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f41ba37a395b47d9a6e9f2b571f3198e : SUCCESS in 44m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0deb01f5f07541d9a5424508b7f6f6e8 : SUCCESS in 25m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6bea7853be4fd99be479c164928536 : SUCCESS in 23m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5be059bbe78545afa4b740406a7bd896 : SUCCESS in 16m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/accf357c04f34cd8b2778a0361bc795b : SUCCESS in 9m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f4d72e4d91b4da591d25d7569fdd2c9 : SUCCESS in 18m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddfe5a97d1a04598bf2f44a5a8252398 : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f2196b25568c56dae2f61c912cf5575a8083875e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:04:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c6f92f8c00194513d9c7a24873d522cf37f2b106","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:06:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca8299949e24d0591c4057dbabf6ae9 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ae251cf1211444f8f1e5c3cad552c48 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e438189a198a3bc656a99be9d82793ddf6077c53","revisions":{"1355f62f90201ef50d486b3597f9a8c598b4927f":{"kind":"REWORK","_number":1,"created":"2021-12-02 17:32:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/820211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/820211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/820211/1"}}},"commit":{"parents":[{"commit":"8897ea6d35034501930e268145f0d823f369c543","subject":"Update CentOS Stream base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8897ea6d35034501930e268145f0d823f369c543"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 17:31:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 17:31:29.000000000","tz":60},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1355f62f90201ef50d486b3597f9a8c598b4927f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1355f62f90201ef50d486b3597f9a8c598b4927f"}]},"branch":"refs/heads/master"},"e438189a198a3bc656a99be9d82793ddf6077c53":{"kind":"REWORK","_number":2,"created":"2021-12-02 21:44:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/820211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/820211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/820211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/820211/2"}}},"commit":{"parents":[{"commit":"8897ea6d35034501930e268145f0d823f369c543","subject":"Update CentOS Stream base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8897ea6d35034501930e268145f0d823f369c543"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e438189a198a3bc656a99be9d82793ddf6077c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e438189a198a3bc656a99be9d82793ddf6077c53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
