)]}'
{"id":"openstack%2Fkayobe~820221","triplet_id":"openstack%2Fkayobe~master~Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b","project":"openstack/kayobe","branch":"master","topic":"error-handling-squashed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-22 11:49:25.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-21 16:37:41.000000000","reason":"Radosław Piliszek replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-22 13:04:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b","subject":"Install Ansible collections for kolla-ansible","status":"MERGED","created":"2021-12-02 18:15:49.000000000","updated":"2022-02-22 13:13:42.000000000","submitted":"2022-02-22 13:04:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820221-error-handling-squashed","meta_rev_id":"fe3fe25530753eea12d2c0aad4bb175a154a230f","_number":820221,"virtual_id_number":820221,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 13:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-22 10:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-22 10:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-22 10:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 19:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 12:02:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-22 08:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-02-22 10:28:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-22 10:46:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ddfa6837ffdf81056334c234238b93569e12cc53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-02 18:15:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b81684f09667884fbf12252988b75bcd50be4fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:31:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87e29d9bad004460b164fde81f0d65ab : SUCCESS in 13m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3536ca91d4e44459bf0e43b09252d589 : FAILURE in 14m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9926887c79214b1bb92307883912c588 : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c391af1136374713bf49b3fc546a8b4c : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e9418bbe8aa4ca146e90850aaed1f7635ef9525a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-02 20:35:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"681916ac5ca4c63f193961654e0ad69f34bb971e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 21:45:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8811a3a3da2f483c9990a4db1e3cb326 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6260272877104a9d88d977034eb4ff53 : FAILURE in 10m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9684aed6c94748e0bf0bd0369dcce588 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6326e4a7b6d4506bbcd59f8ff05bbe0 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3a9d218754c95cc210324d4e2dd6963d3628c6af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-17 17:23:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e88d6cd8cad83c26b29953d0e78f5a24a62ea03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 17:25:32.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"742c932eab10b571493243c14539a67faa0f9c19","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-15 15:30:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"061cbe7b04aa1d2c080b95fcc8a01e4f7c4ff0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 17:32:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9667d85546ab4d5ab52653eefdf469f1 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a73be20c1e14d1fbe75db131afc25b7 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f12b6c30444cb38e377c2c4dfffc08 : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7b722ffdb646f0b9a88ab0aac53c04 : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646642c665bd46b4a1888774267f489a : SUCCESS in 11m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/080e9d2ee46c41de99e6674783b10f47 : SUCCESS in 8m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98ae35f0ec114a3ebc9067aea7f9bd55 : FAILURE in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1acbeaffd2934e7eb4329300f0a66834 : SUCCESS in 34m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8610bc5212ee4286afd993bd5be97952 : RETRY_LIMIT in 5m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad8579b8dc347a0b4b87f0103677e4b : RETRY_LIMIT in 5m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3859c95e3e924bd18ed008b1fa21385d : RETRY_LIMIT in 5m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/11a495be6aea48b8b3cf1738f6ff6f1a : RETRY_LIMIT in 5m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53672527926c4108beebc5d2bfc8d763 : RETRY_LIMIT in 7m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f8cc7bccdec243248493c65831eaa234 : RETRY_LIMIT in 6m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85fdce848b9c4b9b863ff0a5e4cced20 : RETRY_LIMIT in 4m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c5066d426d0a44afb688d73a951022b7 : RETRY_LIMIT in 7m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20ffe4563f61425dadd04b101c91f281 : RETRY_LIMIT in 5m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a4461d45c47453caaf03e86df300317 : RETRY_LIMIT in 4m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34b87214f7404f4e95ba09b4bc6a62f0 : RETRY_LIMIT in 4m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3d6cb7784cc3495bbdd26c85db18783d : RETRY_LIMIT in 7m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0660d421614ba08be4955274cc61a7 : RETRY_LIMIT in 4m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6d04f853368f4a0fa4a795d4b0ebeb33 : RETRY_LIMIT in 4m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2235f26cb43c45428e1170f39cb553d6 : RETRY_LIMIT in 6m 25s","accounts_in_message":[],"_revision_number":4},{"id":"65d1483f5908856bdb899ff1de0fcb05334cd73b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-15 17:33:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"51717b705c50d24c7cf0ec00a7dc2a628298acbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:24:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2021d3ccecb342fdac5c5436af217574 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163bc9aa035e4a16872ca8665220eaf0 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d986872ccc84880b8e760852c98d105 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf37f589f1642cb8da28ece45276d0e : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebd3a8870144cc39316a11ac107ba8d : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3092b88b7c4c4662bf362a5dd462d82d : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/357c942070704f8585d9fd23829a0a50 : FAILURE in 13m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aef07e1153d44f119286d552e9f3c0f4 : SUCCESS in 35m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70b6fe6533e44b7eb0e1e09b31daca3b : SUCCESS in 48m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c869d35f267440aaf006b6e91a7e8be : SUCCESS in 57m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b2c2625980d349e8a5ec5648cc83cf6b : SUCCESS in 55m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5782669b65ee4730be8bce03738bd071 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7133cd877be45eab4d14fe2916699ef : SUCCESS in 12m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a040546adb0420f9590d1033c167970 : SUCCESS in 1h 46m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a494e3c312f743bba8ceb4a3349b1691 : SUCCESS in 55m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1aeda071b9d643f39da266943c1c2f6e : FAILURE in 21m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/599e64dbdc8b4bac89bf91c62461b8bf : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4f48e1d54c54306991abf39f6105625 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73975f69ae2b4ef69b83f4ce599669c8 : SUCCESS in 24m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f530ec7c8e9e4f50862d48cfe52e7c94 : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff23abc8605a4faa839818eb994fe7e7 : SUCCESS in 12m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a69b0613d4840a592b6d48b6a163d0e : SUCCESS in 21m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ccd1382ec33486098192dc714cc3520 : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6005601345e9db39ac2a51a66246fe638d1d04c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-16 11:07:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"17b4ab0c10b62d18693e31022692917c26a42a6c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 12:02:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fdc830aaecdf433f48f0fb9662de6888704ac2e5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:55:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"d6e1782f606e82a592315cd73986e3a2f1332cbf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:56:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dc260869b8da40f0a8ac972a801c5a3c269d3d2b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 16:25:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0141fe277abc29a5cca6fb4773d5b601a1131da9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 16:27:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4b1045a823be8e185646d8eb4fb1986bba6afa9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 16:37:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2e5e84067ba8e69a0862bac3d5f2659dc91c39ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 08:33:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"623ceeaf5a3756116277faf5f9240e4e45c6f5f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 09:23:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e2f46ee29ca83a8009a792be108d210c04fab243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:07:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc7e3f91023465da058c12cd2a6723d : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ed5eac365c4875918d3f57536362c8 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfefe8d4ec2341d18de0c342cef74ee5 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ecd20f12a9f4837bb163ec070e7131d : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491e6ae2d1b14110b5fbdb8a4ba4880e : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04a679258363448ab80ce9f472589ffc : SUCCESS in 11m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db3138e1cdb34214bb9b571543a184dc : SUCCESS in 16m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83f348fbad9d42df9f379dea61b07523 : SUCCESS in 39m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f6d4948140b64a03ae1ed58442ffe9a8 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fc2512dcab44c2b8ffeec7363f96197 : SUCCESS in 16m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2d504e51fbf44e4cb3c4384197eddcce : SUCCESS in 22m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56751eabc2c349f0af45876fd331db1e : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":6},{"id":"2962b094cdd36cd1ce6853e11ffc9316a4be81bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 10:28:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7800f29b49cd9b3add616c376f2b2e580ba8ebc1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-22 10:46:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e68c66bcb36a2255669b8b0a09d8137840f6f137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:47:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"32b2ae1ffd54fab879adf4e91ba2467c290e5801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:37:22.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"deccbf44466372dc959a9af51e2174eebcf5e07f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 11:49:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0daf51694480880f6cbae74eb7403de49acffad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:32:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f44ea0250ae44aaab80abb3199979b7c : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e114f8484a7043ff8eedec3e549f1066 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/517b0375badc4130b63f5b405aaac9ac : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfa30cf859564c008e8b7ac4b4fba1b1 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba17a9af78084e6d991222f679ea5d97 : SUCCESS in 9m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8db402936d8c46e5b6067c99e29f9b07 : SUCCESS in 8m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd58eefd73fc4d70bcc0ceae5e6c29aa : SUCCESS in 14m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81ec822c99144bd59f1c510aef26ef90 : SUCCESS in 34m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba256f8efe874688b9d451967b57a6c2 : SUCCESS in 20m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e5142a1de540479e381c173e5b595b : SUCCESS in 14m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c8bb1d4832104cdbb813f92b89577fd7 : SUCCESS in 23m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d070198260e450d812443f7e699f6e2 : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":6},{"id":"521e78dc7cea3e0821238f808d9d1ef3577482fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:33:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b8304cb359e988df514ce4fffb115a0805d0b7e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:04:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ac6a3541a14f31a6c5f44c5c205d22 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ccaa21c57714f1bb226f9c285ee5252 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4505fb4ce1834f90909164da9b9f6d69 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4799922b85374ad283cc880ddf47931e : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f236c711c954e39b2412204e84437ce : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1be3d8e530434faf965a01f8f5c1e7fa : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4484ae1ff97f4171b56d4a60ba3f7b27 : SUCCESS in 30m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36b7ece02d0e4437b9c01f98239fdad7 : SUCCESS in 17m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025d4ab485464ba4ba734fdaa7cba46b : SUCCESS in 13m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/75b6fc19cbf848cdbde7d686a14ab351 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaeaa3cc01404d019e8048a4cf326d2a : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":6},{"id":"36c4bdd360d54902ab63a4e5d25d2fdbaddea50b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:04:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fe3fe25530753eea12d2c0aad4bb175a154a230f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:13:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ce17d173f9464f95d748afa7db0176 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3efc551e7ef97a351350ac54791a17aa938fe9aa","revisions":{"b713841b3f52f61e0b615dc7ab21a4ff5bb8fe10":{"kind":"REWORK","_number":1,"created":"2021-12-02 18:15:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/820221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/1"}}},"commit":{"parents":[{"commit":"8897ea6d35034501930e268145f0d823f369c543","subject":"Update CentOS Stream base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8897ea6d35034501930e268145f0d823f369c543"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-02 18:15:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-02 18:15:40.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b713841b3f52f61e0b615dc7ab21a4ff5bb8fe10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b713841b3f52f61e0b615dc7ab21a4ff5bb8fe10"}]},"branch":"refs/heads/master"},"f8e87eae4ce8d0a04d8e7ad764ae1b0075786017":{"kind":"REWORK","_number":2,"created":"2021-12-02 20:35:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/820221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/2"}}},"commit":{"parents":[{"commit":"8897ea6d35034501930e268145f0d823f369c543","subject":"Update CentOS Stream base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8897ea6d35034501930e268145f0d823f369c543"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-02 18:15:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-02 20:33:42.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e87eae4ce8d0a04d8e7ad764ae1b0075786017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e87eae4ce8d0a04d8e7ad764ae1b0075786017"}]},"branch":"refs/heads/master"},"42ecdd2e74c427e2c5fcd9bce3733e7d1b49e84a":{"kind":"REWORK","_number":3,"created":"2022-01-17 17:23:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/820221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/3"}}},"commit":{"parents":[{"commit":"e2e0df58053c269dfffe67043119fa6f4be58bbc","subject":"Install openstack.kolla collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2e0df58053c269dfffe67043119fa6f4be58bbc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-17 16:44:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-17 17:22:58.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nThis change uses the new Galaxy requirements file in Kolla Ansible to\ninstall the openstack.kolla collection.\n\nCross-project dependencies on ansible-collection-kolla are supported.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ecdd2e74c427e2c5fcd9bce3733e7d1b49e84a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42ecdd2e74c427e2c5fcd9bce3733e7d1b49e84a"}]},"branch":"refs/heads/master"},"1b15b13c7fd578aa611d64803b4dc322be9fd2bf":{"kind":"REWORK","_number":4,"created":"2022-02-15 15:30:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/820221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/4"}}},"commit":{"parents":[{"commit":"2daf23f3be12a29c1d0c3ef29b116639dd80c2de","subject":"Install openstack.kolla collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2daf23f3be12a29c1d0c3ef29b116639dd80c2de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-17 16:44:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 15:29:31.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nThis change uses the new Galaxy requirements file in Kolla Ansible to\ninstall the openstack.kolla collection.\n\nCross-project dependencies on ansible-collection-kolla are supported.\n\nStory: 2009854\nTask: 44504\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b15b13c7fd578aa611d64803b4dc322be9fd2bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b15b13c7fd578aa611d64803b4dc322be9fd2bf"}]},"branch":"refs/heads/master"},"690fa8be995d375ec88c5a4163db315a28e16e6a":{"kind":"REWORK","_number":5,"created":"2022-02-15 17:33:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/820221/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/5"}}},"commit":{"parents":[{"commit":"2daf23f3be12a29c1d0c3ef29b116639dd80c2de","subject":"Install openstack.kolla collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2daf23f3be12a29c1d0c3ef29b116639dd80c2de"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-17 16:44:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 17:33:19.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nThis change uses the new Galaxy requirements file in Kolla Ansible to\ninstall the openstack.kolla collection.\n\nCross-project dependencies on ansible-collection-kolla are supported.\n\nStory: 2009854\nTask: 44504\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/690fa8be995d375ec88c5a4163db315a28e16e6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/690fa8be995d375ec88c5a4163db315a28e16e6a"}]},"branch":"refs/heads/master"},"3efc551e7ef97a351350ac54791a17aa938fe9aa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-22 09:23:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/21/820221/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/820221/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/820221/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/820221/6"}}},"commit":{"parents":[{"commit":"06c619179933103f2e701d3e6d33857c5078071e","subject":"Install openstack.kolla collection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c619179933103f2e701d3e6d33857c5078071e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-17 16:44:43.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-22 09:23:35.000000000","tz":0},"subject":"Install Ansible collections for kolla-ansible","message":"Install Ansible collections for kolla-ansible\n\nThis change uses the new Galaxy requirements file in Kolla Ansible to\ninstall the openstack.kolla collection.\n\nCross-project dependencies on ansible-collection-kolla are supported.\n\nStory: 2009854\nTask: 44504\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/819430\nChange-Id: Iac185dd2bbbca128c6cf71b2734e94b3e1c6133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3efc551e7ef97a351350ac54791a17aa938fe9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3efc551e7ef97a351350ac54791a17aa938fe9aa"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
