)]}'
{"id":"openstack%2Fkayobe~820450","triplet_id":"openstack%2Fkayobe~master~I0a2ff5caa01d54b1532d30d501b55ef23a6deff8","project":"openstack/kayobe","branch":"master","topic":"seed-hypervisor-snat","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-01-07 12:45:44.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-10 07:59:14.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I0a2ff5caa01d54b1532d30d501b55ef23a6deff8","subject":"Allow enable SNAT service on the seed hypervisor","status":"MERGED","created":"2021-12-04 23:11:07.000000000","updated":"2022-01-07 12:47:34.000000000","submitted":"2022-01-07 12:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820450-seed-hypervisor-snat","meta_rev_id":"fdb3b53036098da3f1ca689e2410714c4e4a0cc6","_number":820450,"virtual_id_number":820450,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 12:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-12-20 12:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-06 13:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-06 13:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-05 00:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-06 09:46:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-12-20 12:03:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6b74bc0ec4e56afc2b563667a10de6b2a27fbc13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-04 23:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c7f34e7d35ddc0c43fa1d838b88e212bc768834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-05 00:50:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55b311afefcd44edbb0421697fba2627 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804ec6a61c62450f903a840bf2fdbb34 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/006eed89de1a4fcdbc17335065ebf094 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b701ce99491244c9bf8df0d3afc03f73 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc168f65cc04159a55b67f2ee50ba2f : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3c7ebad02b24d2295d4438b0080b556 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/969e9c1882154c10b13458c1c53279cb : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc492275a17d4bbdb5d8ab33927ac038 : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/138799b36c3e40bda540957856748e90 : SUCCESS in 22m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5d6221cb26c34d778a7e487124ac463e : SUCCESS in 59m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2186dd3386fc4ca2a796e702b1bd9556 : SUCCESS in 39m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b4cd72bdf7ca498da53daa07ca7e802e : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4abc3a6b743e4ec18d63d1c5119b29bf : SUCCESS in 20m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea33167f235548ef8ea969af792446c5 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95e6c457d59645f9bf806fac2902f68c : SUCCESS in 1h 35m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3bd8f29733a4ceda189eaa0f8fe2bdd : SUCCESS in 1h 08m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/852465afb3554469aacf39b6302eba93 : SUCCESS in 53m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894e777dc7c7400984419d0ca25df3dd : SUCCESS in 40m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/455f7c74cc5d4c879ba7969368b02ad5 : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/110cc76ef7fc4b44b18a67a0ce74c720 : SUCCESS in 16m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/145b1c876d48425d9be99c3c901f9643 : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a341ea092ccd447babb79092650110ae : SUCCESS in 9m 26s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ead7ff6a79cb42bdb1d0f8bd9ae08a5d : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b034535061043238bf90c74af7eacf6 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"da5150174f6d4755929fe66f48ee5038410512ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 09:46:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d5805022ae88ec53019976a8bb694902264858c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 09:47:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d455e5192bb6094dcbca8152f91c8f576e46713","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-06 09:59:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9689d70a8ff1a5a1af4faee168b2b01069f5a1ca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-07 06:04:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db157cbce0d9da16f357b11d9ffbdea97fe1c481","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-07 07:51:48.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"287f63da074713704a3237872370484a935da0da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-09 23:16:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3939b3f4f0d4066681e776f9442b0b418434ef76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 02:50:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d5625047f234391ab53a8914ca3c085 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887d6903a55f415582c818acd3671f0a : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f75cb0e350f74f708c4b1e68538044ad : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04a7aca42da84197837cf0dfee4419a2 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9f92bde51b49adbc0b67bac52c4276 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86f6b5dad86543fbba6e07f6dbfb020f : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d715e0bb8d284e1abbc895b5f2fac1b4 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df3490c6e63d455f840461f33450f146 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c395571b6bb410daa90dcbeadb8bf29 : SUCCESS in 26m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1fa2ab20d6e94ff4976614e01cdbd718 : TIMED_OUT in 2h 06m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68ec3b7a58e64618ba79d188a9beef9f : SUCCESS in 42m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c85fdc80abda414bb2436a5af9dd9401 : TIMED_OUT in 2h 05m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f82abfe613054f2da505f23682c717be : SUCCESS in 19m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0703351b7e4c998c2639819fbcac2c : SUCCESS in 16m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d6ab3792fdb4319916527b82ebdc930 : TIMED_OUT in 2h 36m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ceff1cf1c704ce0a4ff787d0ef68538 : SUCCESS in 1h 17m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5e902303531346c8b104554da1743324 : SUCCESS in 50m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3957fa6eff248d5a52bea4d61add5e9 : SUCCESS in 38m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/470f4e975d1f476fa3df42ce58b6d3a2 : SUCCESS in 33m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5f17226b69346fda9f4bdc51473e404 : SUCCESS in 23m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/18022114ec56416bb9014373bcc5a36f : SUCCESS in 15m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/669910f631924f4292fdfa62086c37e0 : SUCCESS in 12m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca8f45d6e179493ca02f87294e06efb2 : SUCCESS in 16m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98bffdbe8b604242a43a834ff5512ab4 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9435c5956ac48f952da77a28cfbd5684c87a8d69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-10 04:36:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1771919e4ddebd8fcfdb8cc8ac398ef58872c303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-10 04:47:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d145c552a9d41d96765ae3734245712c9bec8fa2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-10 04:49:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5063e40e1120f986525a25807c96695f02d215c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-10 04:58:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d6cb7afe96dd7223bc20ba7dfc46bbfb838d383","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-10 06:55:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ae02049cc78d6912bcdaeb7a8e07e9105dfa459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 07:25:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23179696edd84e8db2c18d3d19a3ad0a : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2e8ae51a784294b6aa34e6c9361549 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8132baa17c4446a7b3dc767b0e0684a4 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7c5a727134426fb20bca3ff392ee70 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84c718b4499490db59bac204ab0e98f : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9db02db1271a468abcf8598be75e721b : SUCCESS in 5m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d2e3346eaea4d0cba279b980db18c34 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07325e9e7d5d42d79ffcd4c2e9285955 : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4341339ef194eb085187cec131611b1 : SUCCESS in 29m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7b6c80e4ee9442cd94c83883cb67aee9 : FAILURE in 2h 03m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c44e1bef84d47a29ecd5377cce308e9 : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ad5ee5b2d00d416ba0092de998d090cf : TIMED_OUT in 2h 03m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9da43d811d504f53b12d15613d5c8a2e : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5601222179548ea8e642c813482600b : SUCCESS in 12m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dbf943d099f4478a93473dabd5646786 : TIMED_OUT in 2h 33m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c476e57e6f8488eba8f788751eead37 : SUCCESS in 1h 17m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6dcf2f586cc348d8881c079ae53917be : SUCCESS in 47m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f89915966f1446f5b22752fa5293c25c : SUCCESS in 32m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/59cfa6e2dbf54ce097db2c1e3107f7a9 : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/949654b1156343de9332998e0b8af545 : SUCCESS in 23m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/328915ac48e84e1cb1f28d542204646c : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/145e9a50d65145c494ba72020449d90a : SUCCESS in 11m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8662819a7d94442ebf8c2c78bb325048 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8d0b790f66d4a248029f695545dba89 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a9f01332d0ecc32ddbc28e1cb6810594f9f09a54","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-10 07:44:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a56d62b80895826ec187024034551de12b00fe96","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-10 07:59:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce1da5b7dce3267c696ab783288c63af7701f039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 10:13:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc29af1d2370492face8a0f6ea51abf6 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca229a8f38f04681a5fb952bd1b9cc17 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ba3b7db33e4fb597d1dbb19e8b49c1 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fa315409cca435a93192f6cd48e5ccf : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc51ef7f19a40f9b9da6c33db392aa9 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b559ba9ac7e4437e997cb29974d88b03 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b917e5ba80054f06ad369e33b4f200ad : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e25835f0973b447b9733c1199c324f3b : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae43fdac1c324b3a892ba341b0505ae3 : SUCCESS in 32m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7a0b05cb5574758a2a7d0fff6669d8d : FAILURE in 1h 55m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31be4ed1f3fb47b6b781200e0675a962 : SUCCESS in 46m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a219e62c97a647fa851327999a439bc7 : FAILURE in 1h 57m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/27662e89fa264404bd2ce62c9a26f0b3 : SUCCESS in 25m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0327fe061cb04c729b23d6792fd22711 : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1e7d9949b6eb418dad635c0da2694cd4 : FAILURE in 2h 27m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a05b010412b450c8cf01657cb20af0b : SUCCESS in 1h 01m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/719aa9ff4fde4fef824d2a2c78ebd492 : SUCCESS in 38m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffceaff11eb346039842f0ebd9bf1ffe : SUCCESS in 38m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4deb5e7ec2ce43ed82ab5d320f87c2c8 : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbfd5c5613c64699b2d1682d7ed532b3 : SUCCESS in 19m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9600ebcbd38e4e4ab1819b0c71cab186 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35765fa113da45b79e2e756775996cee : SUCCESS in 11m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/35e700c70cb24956a986baa681137b7c : SUCCESS in 17m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd2d2567d413431398bc1de677c38f73 : SUCCESS in 11m 06s","accounts_in_message":[],"_revision_number":3},{"id":"aedd57f90a1a1c4ddb2d43053ee0aa1b0b0db8d4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-10 11:01:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72e8aa9114f4ce5cf5c82c6146815b77266e636a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 12:52:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4d0039c3cb145a89e50a8df5203b99f : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71af980305a648ae978f9e1660a5a1d2 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70e1ccc53c64809a3439e72a09eafba : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf28eeee50464199a84b22d51bdcb83d : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964e78913d554adf8ad7af4bf5a4378d : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7839143f34504b359e9545397eefb7dd : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8cf30e1dd3e4ff784dcb4a2b834a5c2 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/574719fef02d4e1dbbcb57bd122ba55e : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a2c6108043640258d1bba58baff6471 : SUCCESS in 32m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3c9f5369d626471eaf73498cf3a56ae1 : SUCCESS in 1h 14m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b0f001e95644e1fa1bd0f28c16724c2 : SUCCESS in 55m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6a4a9ee67c1449caa495d4054d8217c8 : SUCCESS in 1h 13m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a43d1f5a97246ffae69654f65a257bc : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663e02ad52504706a6ff9a1d5ae295df : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c5d91732fff41f0894be5d9ca7a4321 : SUCCESS in 1h 47m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebe9ff660a154ba69619ab4eacb76040 : SUCCESS in 1h 16m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/93dc15a89be54fda96590024c156bff7 : SUCCESS in 47m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57fcfb36900c43e3936f037a85aa1ac0 : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e055e707079748258c3f297a88577c16 : SUCCESS in 27m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/336b8cb96c1b40ffb24213a2d7940935 : SUCCESS in 22m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36097ee342d8426daa7a72fc81ac7341 : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c9c670e490425e867b703f3e378270 : SUCCESS in 11m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/976becb2a290415c9b0752d0191581ba : SUCCESS in 17m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/537d785684d84186a016fa9b59128e63 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7449ab5ff097e5d0e57247b3b611362940c40ab5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 12:03:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6823cd2d12bb3f02fc84c94a18aa43b22f9534a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 13:44:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ecdc03f93d9ea15bced79f6361ec2e4db5fe5fb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:45:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48acb35c32ab6fdd9bfe5de6e3770cd95f551319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 16:17:24.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c0912f43224a8dbda56f693cbf062d : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd9ab2bdb9142ac9340375c13f584d0 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/587e2502cbaf4030ac5cfcf5f152b4b7 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20d04ebf0dc4604b11aab7cab3ec90a : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c6a6e7791e4e0a9dfb8d33eb3c77c0 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b96e704d70043978c86f3cae0c58b51 : SUCCESS in 8m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa37595c73be49a6a2fc282f023e58e6 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0871594d64194c3d82007b2282b92a2a : SUCCESS in 30m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b7db0af1a7da40a89b7108f713d992cf : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07bf3f7a24b1485f8796318229833a44 : FAILURE in 40m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c7405c8f58804c7ba3a9c7c47fb0fd58 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2bd7f3424ed84f5082e21b79602d00cb : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6626ba72ce3b46bcab9760be5addf1e7 : SUCCESS in 19m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db47f30b1b144560b5d566a585bf38e3 : SUCCESS in 1h 53m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b398e035bb04002afc51d90616b3fd1 : SUCCESS in 1h 13m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/44142dd56bc14e23a0fb811db21f013a : SUCCESS in 51m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7118bb7db6c41128fa0709a6d56774a : SUCCESS in 42m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6f4dc96966a547ab8f905f2353087942 : SUCCESS in 32m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09d4b622a3b448ce8bb36cbb7cc5481d : SUCCESS in 22m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5622ee522b284503a865e3186fe54b9b : SUCCESS in 21m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d3223227f244cb8e4fabd29174f5f8 : SUCCESS in 14m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d7571023dc2141bcb6c3fe99f5912faa : SUCCESS in 22m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7574f5f11960413a879481605b7bb6bd : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":3},{"id":"24129a7ff581691ae61d3f74c403d5e355e2e781","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-06 16:29:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ca40a9dab149c3ba6cba1cd211cc9f5b47e99ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 18:21:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c047a940acc64ea2a5020ae1b1336624 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f91bd0424c4397be1ecbd1fc0abd83 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ded49415f9994c678950feefd8e2ae08 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a29fb31248a04e9794e4e6b5c39dbe21 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98873b0b5f18444eaf99b5027c24efc2 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e5769ac3ca44ba8bf9ceb6d24873644 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f586e923f95740798f4343c04c1a2202 : SUCCESS in 8m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d10ef83e384e4ed28405abf66185d084 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1c3c4d4e53b404cacb4ca0164023c81 : SUCCESS in 31m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/69b7c0d4097f45d58d722b4ac304d542 : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d32d0bc472473a8e64c3b6f1162d94 : SUCCESS in 50m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7c838b103abe42f08afd31edcea04e52 : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e91b92db23044921a29d07451b417462 : SUCCESS in 25m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/092f7492bfe44ce5a4e73ef0fe93ef60 : SUCCESS in 18m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a3f48166c6e04506b912ac1f049072be : SUCCESS in 1h 49m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e69f5d16e174017a0abccd9abea9d93 : SUCCESS in 1h 17m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8ba662d59d6d45628cfc9a513d57256e : SUCCESS in 54m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2739b1e7d57940b6b585fb630b75f542 : SUCCESS in 43m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/21126edf9de4476f88500edf20391f53 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f637d59aa4bf4da5837427b3cb909be2 : SUCCESS in 23m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/05ebf1e729de4214b707883e90d3a879 : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49200399447d4fd195d7c8f40cdb7aba : SUCCESS in 13m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4b3432ce6fdc4c4f949a5c8e621999f8 : SUCCESS in 18m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b8f18c5036c4994a6e5e72b00a23fa2 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":3},{"id":"97b91a544891d1d2874e3db1638adf1d60ea681e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 18:22:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21e15fdf64c32a2cb168bd72aa10f8bcec5534c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 23:04:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ddd588ec9142f89b29164d8160c130 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f18973b532d1443886f69af271646123 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f93940e9d940f48726e92f8d69d7fb : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414b65fedbfd43f7954b4176d12967f0 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7189ca339b21491db8983dd8e8346942 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdcd1adc0a754d878df2c6f55a5ccb12 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41fd2ed8055144f3ac2eb75ab5f93caf : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ffef6a27dbd9449895f4a8efb16df3d8 : SUCCESS in 34m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/83094f50a4cf40fdb8f2190a971ff68d : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0a4d26606948beb98e65baeb2da040 : SUCCESS in 42m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0f6fa81d76cd4b1fbaf16b0eb63a5613 : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0698e13d86c44e0eb723673dee4d7616 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/754e5ae76cd7471ab883438bfb852391 : SUCCESS in 17m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86b8e69423e04c1e9d05d31bb9ac2fb8 : SUCCESS in 1h 48m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec39ecd48dd7408380d847beb8d3e87f : FAILURE in 40m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aa749f0c8dc24a01a4c7aa6be4c0ed0f : SUCCESS in 49m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f60f77898e06467fb459636485920158 : SUCCESS in 38m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56d2fadc0ceb405fa24dc045ea24b8c8 : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d115f6d31d844620bc85f2861a924801 : SUCCESS in 23m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a56826af6304b899b209b85299b0701 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ff8afef29564f6f9e5c654b37085bde : SUCCESS in 12m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/42a3987b302b47859d7834535c85ccd2 : SUCCESS in 18m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ca960c9866a45e6b961b79766e9f045 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":3},{"id":"657ed6a2dfdbdcb6b54e2fb80ddbbdd2efd8d935","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-07 08:52:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6fe02d2b7dda826923f4c918bd64cc564b85778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:46:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/031eaa8057824d5680a7cb2c5688b209 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c504473e234011a5ff519972a9fae2 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e66fb21e466498bbb5f9a8f2de87276 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06bb61e8c68c4ad98e63b9cbd88f204b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6f1b1e22a147b9a142a9b774e26377 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a56b597d3ac24400a8a257c601b8ae9d : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/137f5518d9fd43bfa1fd521fa41e8dc6 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/205654402128461497e4d51515ee4712 : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f02f0a633e0f4c00a4925e3a9915982a : SUCCESS in 31m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f45c4aa02edb4fa5aae9b00fbffee501 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feba2e4d572c4b72bd383ea8c2574915 : SUCCESS in 50m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4489fbd17a0a41f5af9f701a36bbd7d6 : SUCCESS in 1h 12m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9289065eeea04249b856916d171e95a3 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72aef06b0524d7db992d489fc5091d5 : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6dd50356fab47688bb989c880c4c069 : SUCCESS in 1h 49m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e3ea8034bb249c2bade18eb2e6e186b : SUCCESS in 1h 18m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a86864023f12405c9425611125ddb452 : SUCCESS in 52m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/163ee70c027a49409cb12e19388814e7 : SUCCESS in 38m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/361403a6fb7449e68f96476530d0d50e : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02891c26982648a787f5058564c1ea0d : SUCCESS in 21m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47731e2c9b584df5b7cba93ec9374e94 : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8d64b1efb2043f2a0c6927a56758515 : SUCCESS in 13m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/80d0e478064e43d7b68ec991089649bf : SUCCESS in 23m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d86241f152c43648359506f2f4574e3 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":3},{"id":"af53d3cd99b6442747777abc0b565a81e2c28ee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:47:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a0415746bc05f807dc2356a9832a047df7c5c58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:45:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1f1612c71e4f62b54c52231d31b637 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b157b8bf0140b58bf49b3e8d9f1752 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e175fe478fa4492c89af1669702ba1b6 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a096b94c9e4f008700610b1fda31df : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cab8ec9ff964e23a569037d04eb25c5 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c7047984a974147a250c80253749c05 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34a86dbaea3e4880a306dc1267516e9a : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26b2ce567c664d89bc668782cefa1186 : SUCCESS in 32m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4a6a8fa1dd44480495aef5e82a109e5f : SUCCESS in 1h 15m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f71414c6d7448f9bf9977a42f0f07f : SUCCESS in 55m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/29cc70e762dc43c084bd40883feae4bd : SUCCESS in 1h 15m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7cf6e442da2c49cf81b64ec546c5d31e : SUCCESS in 24m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0116fac9b1234df0ada6b8fdd6a14bb2 : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d730acedd8414181a18d685f0248fe67 : SUCCESS in 1h 56m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e223a52ee34842098a2adb9d09d0ae3e : SUCCESS in 1h 15m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/29b47ac329e041968b171f2586b15684 : SUCCESS in 55m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5686ccdc7cc04b6f94bdb215d89e974d : SUCCESS in 38m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ae4ef780fc049c2b0794291ae18330f : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d334fdbc7e04744808bf699ba78e6a3 : SUCCESS in 20m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6a8335760f854987bb44aa5fc4a98753 : SUCCESS in 17m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a68e9d6de1fe4281bb6d0a98884d7c08 : SUCCESS in 11m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a7536d69be714249afaa3b88b2df67b5 : SUCCESS in 22m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f570288eaad3478bb1b10e15727d41bf : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":3},{"id":"e1922b862c38ab024e1f10a12a6e70a16e45747f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:45:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fdb3b53036098da3f1ca689e2410714c4e4a0cc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:47:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/966b6569b77b450981379c43e3ca6e43 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed461a7a445442a9d8b909f763793d0 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e10b52c588a6bbfe2dfe01d903e6080b3d50d693","revisions":{"6eae476eb36548309b272e6ca86589c3850caf19":{"kind":"REWORK","_number":1,"created":"2021-12-04 23:11:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/50/820450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/820450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/820450/1"}}},"commit":{"parents":[{"commit":"8897ea6d35034501930e268145f0d823f369c543","subject":"Update CentOS Stream base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8897ea6d35034501930e268145f0d823f369c543"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-04 23:01:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-04 23:09:57.000000000","tz":180},"subject":"Allow enable SNAT service on the seed hypervisor","message":"Allow enable SNAT service on the seed hypervisor\n\nAdds an ability to enable SNAT service on the seed hypervisor.\n\nChange-Id: I0a2ff5caa01d54b1532d30d501b55ef23a6deff8\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6eae476eb36548309b272e6ca86589c3850caf19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6eae476eb36548309b272e6ca86589c3850caf19"}]},"branch":"refs/heads/master"},"39aaa105257cd6708f55758d6f3a6cd09b7261cb":{"kind":"REWORK","_number":2,"created":"2021-12-09 23:16:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/50/820450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/820450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/820450/2"}}},"commit":{"parents":[{"commit":"f88f49ec676258edd4a84d3e58f024757c0a70e6","subject":"Document seed_enable_snat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88f49ec676258edd4a84d3e58f024757c0a70e6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-04 23:01:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-09 23:15:58.000000000","tz":180},"subject":"Allow enable SNAT service on the seed hypervisor","message":"Allow enable SNAT service on the seed hypervisor\n\nAdds an ability to enable SNAT service on the seed hypervisor.\n\nDepends-On: Ie42ab7a0dc9dd1ed1925b3a17134b3770ae8ba98\nChange-Id: I0a2ff5caa01d54b1532d30d501b55ef23a6deff8\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39aaa105257cd6708f55758d6f3a6cd09b7261cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39aaa105257cd6708f55758d6f3a6cd09b7261cb"}]},"branch":"refs/heads/master"},"e10b52c588a6bbfe2dfe01d903e6080b3d50d693":{"kind":"REWORK","_number":3,"created":"2021-12-10 04:47:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/50/820450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/820450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/820450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/820450/3"}}},"commit":{"parents":[{"commit":"f88f49ec676258edd4a84d3e58f024757c0a70e6","subject":"Document seed_enable_snat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88f49ec676258edd4a84d3e58f024757c0a70e6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-04 23:01:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-10 04:47:30.000000000","tz":180},"subject":"Allow enable SNAT service on the seed hypervisor","message":"Allow enable SNAT service on the seed hypervisor\n\nAdds an ability to enable SNAT service on the seed hypervisor.\n\nDepends-On: Ie42ab7a0dc9dd1ed1925b3a17134b3770ae8ba98\nChange-Id: I0a2ff5caa01d54b1532d30d501b55ef23a6deff8\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10b52c588a6bbfe2dfe01d903e6080b3d50d693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10b52c588a6bbfe2dfe01d903e6080b3d50d693"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
