)]}'
{"id":"openstack%2Fkayobe~820500","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ie9b15db6563546ede9ce9735292ec566d540432a","project":"openstack/kayobe","branch":"stable/wallaby","topic":"story/2009721-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-15 18:41:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9b15db6563546ede9ce9735292ec566d540432a","subject":"Fix installation prefix detection","status":"MERGED","created":"2021-12-06 12:04:23.000000000","updated":"2021-12-15 18:43:24.000000000","submitted":"2021-12-15 18:41:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820500-story/2009721-stable/wallaby","cherry_pick_of_change":820211,"cherry_pick_of_patch_set":2,"meta_rev_id":"4160cf42b43a0d23a57511a613f32caee0dfc950","_number":820500,"virtual_id_number":820500,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 18:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-10 15:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-15 16:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-15 16:51:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 14:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 15:06:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-15 16:51:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"de7da124897dc52909f9552b688586cefed091ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 12:04:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"38d1c6fc65b99243ae1eb8acb224c0621cce33f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 14:08:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c87536f08be40f9a2d52183d9555ecc : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496ab51759ff4b2db4658a2e94eb04a7 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52d77fdb872d41c5b0e26c9c8efc9f8e : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/518af163dc394e4b9f44594681e700fc : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62487e06043e498981143b9bab68941b : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fb23d9602c40efa25dfca94d59827d : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b170fc5e7ab460cbbf00768bf1fbba3 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53d8f18d279c4cc1935b8713c76131e7 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/537c965de80c423084c05dda599b0717 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08c4c5354238486cac07af75ddd07d8d : SUCCESS in 19m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/657f2d135de9463790531784337e2079 : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9f2b43875164c9aab3691c1d9af47d7 : SUCCESS in 50m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/322c2ff6b33a4aafbb9cd06169a8b4de : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd6780afe2a7476e95bec32cc7eb26ca : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01dfaeeeebaf4e13a3ee6de7bcc39d33 : SUCCESS in 10m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12a0742a17a74783956c61edf1def241 : SUCCESS in 1h 58m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/19c836cfdc5643509218e9fcd05b097d : SUCCESS in 35m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e1d57b840cc462d87d57ee783d5275a : SUCCESS in 1h 13m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f67a6873b22d406fa3f2800c44edf8f2 : SUCCESS in 26m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/de1510f5311e48ab95d6f99aafcd1cc7 : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/393c4791841744f8804e69c58092e2e7 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ee7c0b5dd0f1c70407edf5bc8c4988862852ebdb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-10 15:06:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"537837833314457fb87e3e4e9dda85c3334efa96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-15 16:51:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fcfbc4a68db831141568995147639a6fafe611a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 16:52:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce06f9ffb3d5a4f8aeb8e401aa2a62179b219f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:41:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a1928d96124140bf0adb91e36b8239 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67af9c6871854f8782e38f43d72e2af0 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec60f48115e64616a3a3c417b19bd2df : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ee85fd9dfa40c6b83abeeb798e7de7 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a50acf526594ffdb519f3b839960b7c : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89e961e3904e4318ab5cfc43f580d308 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32f8266d2e2c49f194fd595f11ff4753 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12bf5f7e05254441b4bd3197baf92e8f : SUCCESS in 18m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/33d5b835970b4ea4a8919d9c6cd40307 : SUCCESS in 56m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdec1f919d8e42748007ee4d9917bca4 : SUCCESS in 41m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cf5663034afe407c89f57df82e064aed : SUCCESS in 59m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e1c3f8c67873423b8e804a5b71670d03 : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70aca35d5f346f68d5b116c15c68026 : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f0f7acf39a443378ed12b1bb93a3fd3 : SUCCESS in 1h 38m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9b68f282877941629567aec28f852ee7 : SUCCESS in 30m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03036280ac724c7eae2e3ae219f4d049 : SUCCESS in 23m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5133588d775b46b0aa8f20ffd00c0e1f : SUCCESS in 25m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/32581477e13c4ee993218e056c0b6c1b : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f813d9a6a17416fb3e59cd661d4ce74 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"97e3f403e6b1c3ca7cae5d68a5791b4237442f3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:41:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4160cf42b43a0d23a57511a613f32caee0dfc950","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:43:24.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5e6cfe3f96464d8f72dd602010d996 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fae72e9a011f4251a88b02012eec3613 : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf7ef2aef3cc001927db5ce1b3983f29362f4be1","revisions":{"bf7ef2aef3cc001927db5ce1b3983f29362f4be1":{"kind":"REWORK","_number":1,"created":"2021-12-06 12:04:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/820500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/820500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/820500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/820500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/820500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/820500/1"}}},"commit":{"parents":[{"commit":"4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 12:04:23.000000000","tz":0},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n(cherry picked from commit e438189a198a3bc656a99be9d82793ddf6077c53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf7ef2aef3cc001927db5ce1b3983f29362f4be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf7ef2aef3cc001927db5ce1b3983f29362f4be1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
