)]}'
{"id":"openstack%2Fkayobe~820501","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ie9b15db6563546ede9ce9735292ec566d540432a","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2009721-stable/victoria-stable/victoria","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-15 21:51:25.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ie9b15db6563546ede9ce9735292ec566d540432a","subject":"Fix installation prefix detection","status":"MERGED","created":"2021-12-06 12:04:40.000000000","updated":"2021-12-16 04:54:20.000000000","submitted":"2021-12-16 04:52:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820501-story/2009721-stable/victoria-stable/victoria","cherry_pick_of_change":820501,"cherry_pick_of_patch_set":1,"meta_rev_id":"c35f8e6ac0842452b6f81dd6cf068a2e3a4dea23","_number":820501,"virtual_id_number":820501,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 04:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-12-10 15:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-15 16:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-15 16:51:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 14:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 15:06:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-12-15 16:51:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8a87396aebc58af88389cde978f06003d757296d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 12:04:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"788fa5a0a5c013dc39e3a50ce0bf5d331a8bf036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 12:04:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b456708b782642731c4c7910cc0dd05da212466b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 14:28:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a10156648d140a2995ce216291c3bbe : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ebfe84b38d450196bd57344242719e : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2a45ade55824f2a95e749720b383b4d : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2089d588b7e54b94aadc4e9a08f9d356 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b26aae134e45f18b151e19f66b6798 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46de18f31c23425e8eaac2f656965ee6 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58ddacad53b04d0da5fbbe2ad4b61a6f : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9da727b7f26c417199823b5dd7a585bb : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73281d3da1694f979efd708562c3c3af : SUCCESS in 18m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7abcd0c56a914dc39d64651e1d674d56 : SUCCESS in 59m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5f756ed6cb904fa9a43031885adfb1b3 : SUCCESS in 58m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d68ff098c2a546a289ae48ad55f9631c : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/267664ecc8db4317bd33557cb7e7056a : SUCCESS in 57m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/be4dd5f351b04da5b46d778ea711f9db : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bf5b21dee113449181a2133e48fbf373 : SUCCESS in 17m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e6078690b1af4bd39b053c8de80841dc : SUCCESS in 2h 16m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f939ae3cf6a64982bf1f2bd8eb34e746 : SUCCESS in 2h 21m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1386bf6364ca46c1aacc913dfb3511fc : SUCCESS in 28m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d473ffdec93d46efa57923f27a84f44e : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/666344fd0080494492cef0fddf985b2f : SUCCESS in 25m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/74e04bdbe7314a1aa7a23fe275d49688 : SUCCESS in 27m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2db9b4fa947f476fb6681d518978f90c : SUCCESS in 11m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c64c2017f74e4f33b20f0ce6c9252d51 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"051dba02b97f30fba3169678ba504b9a4a8e5beb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-10 15:06:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba6e66d5dd71414c3b6bf15bc40fa4140e60e05b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-15 16:51:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84172eb7e1307fc72e618324791df25a4fdc6909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 16:52:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6ed1cea9eb5fffd8855e98e68f0bca6f12887ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 20:20:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a31cf6f06c4f109e6ed8c55999db83 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd159d66fc94a04834c72400acc9768 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a0ba42e526f49ef9fa518aea43bd204 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168b30f9d90a48b5b08e956d9e06c623 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66e487f55165491da6afbec96dfea0ba : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2529e6b40a784b3588f82bfac51d0ef8 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1d1ecb99f7c4cbbbae9413bc28a2c19 : SUCCESS in 7m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bb45993c64e405ebf325677c5eadcef : SUCCESS in 16m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/34348485bfb64fa191a1a60fd267f967 : SUCCESS in 52m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d48b0cb5a5d84da0a0d74955216c7cad : SUCCESS in 48m 51s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b1e66824ba10473fa49bfe1e03fcfab6 : SUCCESS in 52m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bcc1bd694be548f09ba038bc331e740c : SUCCESS in 56m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9b58414786b145889205f06528920c80 : SUCCESS in 15m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/91022b80df73425fa0994ffcb46cb094 : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/90f6989de67743b6acef1bb7b2648215 : SUCCESS in 1h 36m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ae5acf695d0544f08f65e3cf3f3c6451 : SUCCESS in 2h 06m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd3e8cfa2b88414bbae460df8591c104 : SUCCESS in 27m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2dd597fdb7cd42739c32664a68d09ef4 : FAILURE in 19m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a0783bb145c4471f94af90d54e5d4961 : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/831a033f44634fe98dd7df7fa2c066df : SUCCESS in 21m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2be0993d3167474db52b6be447847f31 : SUCCESS in 11m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/09bfa0bfdf724058966d056f2bfa7d8b : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f7a03906b44d9d36f54dfb71c80eed7b50ade3a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-15 21:51:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"825281843ffc9db9ca4f323ea15390e4169b5423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 00:08:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c01f1219c5b0490ebaaaeea37a0c9961 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab20cb0e0e19450c968b8c25b82c5fcb : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eef43bb9cb945c19ae79a81ac06468c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9179fff4e6de4787bb411b7fe30e4e7f : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f9a1a62a404c358505645bf5695af2 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73cd041fecd7468597de48f6e9bb699c : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ec576c0cb6d4e4a80db57ff55699ef0 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fd2a1cb63464dfeb4afe2ec9780a4e6 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a87fd97abbb84458bc5dcf9800eb4082 : SUCCESS in 14m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7370e7ca0eb9427daac61ecb4a59c32a : SUCCESS in 49m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/489fffaf7dc14ac5a901af5948a7063f : SUCCESS in 48m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/006b72238ca741f191e50ffcaa272f11 : SUCCESS in 59m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f01d9985e6cf4a4992aa52b6fe2f0dd8 : SUCCESS in 56m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b25ef8bd80e48abaef0053ff1365461 : SUCCESS in 13m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c795ea2d6d27413480083c572e90ae13 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aabf3459d4484e2994879e15c32a0889 : SUCCESS in 2h 14m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/90d785b580024226ac19692e2e9b12b0 : SUCCESS in 1h 46m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/729c5780c8174d4db12297b71843064b : SUCCESS in 20m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f1603eec46e34b2e9075bb64f5a78530 : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c87717d154984dfc9a50ff1f23449531 : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ca9f3fc6e324ce38ca559e7f6813783 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/662209b451b448b49986a66ed143b550 : SUCCESS in 10m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7adb260835e549dd9d69e3c97c7f9847 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3c5f0f35c27911963c7bd5b589302feb0007059c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 00:08:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad53bfdf579e677e9f9397f5822be96181e97e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 04:52:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51291f67a0849fba3fd265c729c1d45 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba0b6c4810d345ff9e243884263a1cc3 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a9e26b0452b4683a2d54d8f2f087f60 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1d91cdfc2b4be5a3f2f4fc569af646 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74dbe99d0bce4fec918f359c1ae930db : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/837ba4a5058f416dbfb7185c43723eb7 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb97978073b6415f8e91ce4b5c59f12e : SUCCESS in 7m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c975e65e2b8f45b18a1646feb42069d0 : SUCCESS in 16m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9c25b008595944f88cba5e4be199552c : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/89a75b6345c84b0489c39b213d90afa8 : SUCCESS in 56m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/57c8b13e58a9416597bd5532ca9f75ab : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/25ddbd8b2e4847ca9ecec4653937657b : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bad5dabbb17342a3bcb7e6e5e111fb67 : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3405cecf0334c7b9776c26ed25b3b46 : SUCCESS in 16m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/952d85743d0d426ebea88af1a96580b8 : SUCCESS in 2h 02m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/086864f4e6144f53a6254cf650fb90dd : SUCCESS in 1h 59m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/05b6e50d8c6d4b7487a575d49a88afc5 : SUCCESS in 32m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8b11741c650f4bab9b63e510bbe7ea8f : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6b4b3787d3fa4fae91d99a00bf0a1fed : SUCCESS in 24m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a335c0957948451c81ac203a76d9be29 : SUCCESS in 24m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/097a7e4b65794fc7b009274bc816f856 : SUCCESS in 13m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2655bf7479fd4cef99d3871cf8b552d1 : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2d93daf0649277f715cb6400b2d283475ccd7969","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 04:52:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c35f8e6ac0842452b6f81dd6cf068a2e3a4dea23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 04:54:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3e44f5d8a74573ae97e79868b63b6d : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794ca4933dab4be2939e8a9232971e9b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0073e7eafb9080e078b061ecdccbab3c359e1f8","revisions":{"d7f447d945b1a7646fc9f0b9750ac8a5c6a250b4":{"kind":"REWORK","_number":1,"created":"2021-12-06 12:04:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/820501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/820501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/820501/1"}}},"commit":{"parents":[{"commit":"7bade28f73043d5f19a87724199aaf73ad3a79c9","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bade28f73043d5f19a87724199aaf73ad3a79c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 12:04:40.000000000","tz":0},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n(cherry picked from commit e438189a198a3bc656a99be9d82793ddf6077c53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7f447d945b1a7646fc9f0b9750ac8a5c6a250b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7f447d945b1a7646fc9f0b9750ac8a5c6a250b4"}]},"branch":"refs/heads/stable/victoria"},"c0073e7eafb9080e078b061ecdccbab3c359e1f8":{"kind":"NO_CHANGE","_number":2,"created":"2021-12-06 12:04:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/820501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/820501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/820501/2"}}},"commit":{"parents":[{"commit":"7bade28f73043d5f19a87724199aaf73ad3a79c9","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bade28f73043d5f19a87724199aaf73ad3a79c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 12:04:57.000000000","tz":0},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n(cherry picked from commit e438189a198a3bc656a99be9d82793ddf6077c53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0073e7eafb9080e078b061ecdccbab3c359e1f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0073e7eafb9080e078b061ecdccbab3c359e1f8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
