)]}'
{"id":"openstack%2Fkayobe~820503","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie9b15db6563546ede9ce9735292ec566d540432a","project":"openstack/kayobe","branch":"stable/train","topic":"story/2009721-stable/train","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-07 15:55:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie9b15db6563546ede9ce9735292ec566d540432a","subject":"Fix installation prefix detection","status":"ABANDONED","created":"2021-12-06 12:06:53.000000000","updated":"2022-02-07 15:55:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":820211,"cherry_pick_of_patch_set":2,"meta_rev_id":"8038b242121190d9c23a78d8a0108f6a4af89db6","_number":820503,"virtual_id_number":820503,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-06 13:23:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 13:23:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93cb998e80c6861b1a340ec93af05d07f5b45840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 12:06:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"310f1d192bcd2fa7c352373b9c50769de118339f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 13:23:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d93404befd43bf92c10fdfee27d7a6 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8b9883ec0dff46f0a49da2a57edb32f3 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48acb7287c234990897614419a5de050 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0825a02f52f4382a63a1463547e3b1f : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/339d9dc0f86c4b4b806d0847f634d81d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7594244ec847bcbbf9a224b815f27c : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca7c8aa5e0b54c69ae1299304c6ee050 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1f323a7ceb74b7eb0207bfaffdebd98 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a15ab4a7dfea4cd1982a12e61f1b70b8 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60d5fdc1ba20477dadd4e1839640f7cb : SUCCESS in 15m 32s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/661d13e568ff42b9ab00a2f949c0cca8 : FAILURE in 43m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e4ebdca2a3d24f8ebc5c052bf2e1298b : SUCCESS in 1h 13m 24s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/88c056c3c8ad4865b1b446164c6f564b : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/51fe19c99aa54dd38b0ea5cd1190d125 : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/aa6f94acbd094d92839a71be2a636949 : FAILURE in 8m 45s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/aeee5af3400c4c3f818533d3ff8de994 : FAILURE in 16m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e57bc728618e4df5a980212e7cb891d6 : SUCCESS in 33m 14s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b81594fcbfe445cd85cb92f0859dfbf6 : FAILURE in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8038b242121190d9c23a78d8a0108f6a4af89db6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-07 15:55:46.000000000","message":"Abandoned\n\nTrain is broken.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e42baedeed513c6e2147a2e5b07c252de8481356","revisions":{"e42baedeed513c6e2147a2e5b07c252de8481356":{"kind":"REWORK","_number":1,"created":"2021-12-06 12:06:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/820503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/820503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/820503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/820503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/820503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/820503/1"}}},"commit":{"parents":[{"commit":"4c156c3413474e5b7a0090a2d94a54d388acc444","subject":"Fix wrong filename in bifrost docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c156c3413474e5b7a0090a2d94a54d388acc444"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-02 21:41:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 12:06:53.000000000","tz":0},"subject":"Fix installation prefix detection","message":"Fix installation prefix detection\n\nWhen multiple \u0027lib\u0027 components are present in the installation path,\nKayobe fails to detect the right installation prefix which causes\nvarious commands to raise an error.\n\nRewrite logic to detect the last \u0027lib\u0027 component in path instead. This\nshould work as long as this function is not stored under a lib directory\ninside the Kayobe source tree.\n\nChange-Id: Ie9b15db6563546ede9ce9735292ec566d540432a\nStory: 2009721\nTask: 44109\n(cherry picked from commit e438189a198a3bc656a99be9d82793ddf6077c53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e42baedeed513c6e2147a2e5b07c252de8481356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e42baedeed513c6e2147a2e5b07c252de8481356"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
