)]}'
{"id":"openstack%2Fkayobe~820601","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I6acd7bc29ec002c68eb7a3e3883cdc8e38502e84","project":"openstack/kayobe","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-03 11:08:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6acd7bc29ec002c68eb7a3e3883cdc8e38502e84","subject":"Disable failing Python 2 jobs","status":"MERGED","created":"2021-12-06 21:57:12.000000000","updated":"2022-01-03 11:13:29.000000000","submitted":"2022-01-03 11:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820601","meta_rev_id":"ea02697026476e61b2d557706536ae9ba27be48d","_number":820601,"virtual_id_number":820601,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 11:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-03 10:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-10 16:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-01-03 10:00:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 23:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 16:08:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-03 10:00:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c0f19b45b1afdf9a472ce0eab4512b7c78004a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-06 21:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcf5b84a70258fe767281dca5e1e04157f9ee583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 23:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5667c75a7954ef984d322e7a832e3be : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/017fa6c8936542f3a35270a119101eed : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd4bb0ae7bb24439bfa520cc6bf270f5 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8530ba7fcc44aa5a82cacca41e7ed0e : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a150f5c92d544b7cb0572a3135034084 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6068b7b893442e79cb0f7780927717a : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1dd3735323ad4b18a7b98683faf56527 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e28cc0fe2e8489cbc75c388749f5926 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/632d7d134b7d4003936621f83d62520a : SUCCESS in 14m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a47001bffc6b493b801f8d36b3ed3348 : FAILURE in 45m 53s (non-voting)\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/911168f6431347febe6b63a2c208c701 : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/b6c82029a87045e185d0d0a103ff1de5 : SUCCESS in 12m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ec6b662d717f4c4da82adae8f68880d4 : SUCCESS in 21m 53s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8757c17e8d0840f3b89ca5f51dcf0a01 : FAILURE in 7m 42s (non-voting)\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/e15b3b637102475987b110a55de5b80b : FAILURE in 20m 07s (non-voting)\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c8c9c9e221a3406c8a0174e3b7add882 : SUCCESS in 33m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c5b321a05640458fb7a464cd2c0e7561 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e3fde9807e9442dbf05ce6f96f7bd7410c3a3ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-07 08:37:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c429f7e5ad1494512fbe520a3dc1fe203189309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 09:34:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50e8e7bc3ae240a8916c9952e2f00329 : SUCCESS in 7m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d0346fd5ca554662bec4ebecd9dc351c : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d7e2d6acd845d9bfd08b3acafde2fc : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52b2b8674304795a4266f3c3dfb78f0 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22bdfc333e304363992fced48d5811b6 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d42781956c41fca918027ed67a7b86 : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03b0c09780364750b0f35cfc3e99f92e : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd077b35084346a9a2cc4820cc4feaf7 : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/569c1d5a9d7e45f7809132e0ea7a6137 : SUCCESS in 18m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/040fd7230fc843439054b530acd077be : SUCCESS in 53m 54s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/efcc4a39d5b94aa1923010aa8f3969ef : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e91c1b6c5bbe48ffb6d41c4a03d5b6c1 : SUCCESS in 14m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a4b3c0b00e4447f5a9e9900623285229 : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":2},{"id":"24c721ecab6a89c9cedfddc8e4e967c36f99f3ce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-10 16:08:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a53b357d0b05b90412964f389b69024412bfbc7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-03 10:00:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0aec03c02c351d1b6e7b1beb17e3d8af05c1c688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 10:01:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8873d38a6c1fdb3c0816f36a8c3db2e8c2eb6f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:08:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31388ce1a9bf43d78da9b85fcf33e7ef : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c6c83afdf3d4cacb3a8f0ea98791b01 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986bee5575a94bb280ef0a91245cbf47 : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd1e55229f224f81bf0f95e7283ffedd : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f11e30088a490abd4bcba7793e2a7b : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/442fa19e28254ff0b6d52883661d747b : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6345530809a440388455c31e037f691c : SUCCESS in 7m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/231bf6d4be7f4b88a8cfe3843d514124 : SUCCESS in 14m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/afc3079b3797491583ed8cc5a2c93c2f : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/98dde5b26f2e41dbadddcf0bb10fca52 : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4542c920ddf45e3912b959824ed8485 : SUCCESS in 15m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0664b69b2a2640c7b926655b8369a8ca : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b8bf1d5e32f7cf88d4c15c58bb2f1a679bf16c2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:08:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ea02697026476e61b2d557706536ae9ba27be48d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:13:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c0b5fcafa64ffcadaf70df6855b914 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d8f66ea0791c1bdb903a46b78a38378b6eed99b4","revisions":{"71d504d19b578b626fb4732d4f8cb6525797814f":{"kind":"REWORK","_number":1,"created":"2021-12-06 21:57:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/820601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/820601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/820601/1"}}},"commit":{"parents":[{"commit":"4c156c3413474e5b7a0090a2d94a54d388acc444","subject":"Fix wrong filename in bifrost docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c156c3413474e5b7a0090a2d94a54d388acc444"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 21:56:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-06 21:56:10.000000000","tz":60},"subject":"Make broken Python 2 jobs non-voting","message":"Make broken Python 2 jobs non-voting\n\nChange-Id: I6acd7bc29ec002c68eb7a3e3883cdc8e38502e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71d504d19b578b626fb4732d4f8cb6525797814f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71d504d19b578b626fb4732d4f8cb6525797814f"}]},"branch":"refs/heads/stable/train"},"d8f66ea0791c1bdb903a46b78a38378b6eed99b4":{"kind":"REWORK","_number":2,"created":"2021-12-07 08:37:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/820601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/820601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/820601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/820601/2"}}},"commit":{"parents":[{"commit":"4c156c3413474e5b7a0090a2d94a54d388acc444","subject":"Fix wrong filename in bifrost docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c156c3413474e5b7a0090a2d94a54d388acc444"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-07 08:37:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-07 08:37:17.000000000","tz":60},"subject":"Disable failing Python 2 jobs","message":"Disable failing Python 2 jobs\n\nChange-Id: I6acd7bc29ec002c68eb7a3e3883cdc8e38502e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8f66ea0791c1bdb903a46b78a38378b6eed99b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8f66ea0791c1bdb903a46b78a38378b6eed99b4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
