)]}'
{"id":"openstack%2Fkayobe~820862","triplet_id":"openstack%2Fkayobe~master~Ic8c212111301d6b144e3aa0fe124b022b0296ec6","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2021-12-09 00:37:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8c212111301d6b144e3aa0fe124b022b0296ec6","subject":"CI: Use correct TD agent repository version","status":"MERGED","created":"2021-12-07 13:35:31.000000000","updated":"2021-12-09 00:39:11.000000000","submitted":"2021-12-09 00:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820862","meta_rev_id":"86b1729025304f352b81e59e8f20f187e45d464c","_number":820862,"virtual_id_number":820862,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 00:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-12-08 15:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-08 22:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-08 22:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 18:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 15:21:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-08 22:38:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0cf8881bfd82cafe6163735a83c2e4163f6a2ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-12-07 13:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0216cb085ab71ce0b6df2e3e80a6782705255bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 18:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d127dd5a1c04bcb9ee760fb96801f51 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d9aa7b0c5d48058a362dbba64a3a9a : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c442bb774e7b456188f30e091fb5ad4c : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0652737241ed448bac8f89770754d5a2 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87d03b018354ae1addaa0a434b23bc9 : SUCCESS in 11m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c1830d878e8494aaee53d554b3156cf : SUCCESS in 8m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a33f162a4ee54879bd28cc161e98e565 : SUCCESS in 15m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba92635724074fceb8abfdefffdf375e : SUCCESS in 34m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c556f325fbe04c249568c80b7a327640 : SUCCESS in 1h 14m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/109ba0ca9ee34089b71fa01918b1aabe : SUCCESS in 52m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9497014a8e60449596cd50e0e5d0e918 : SUCCESS in 1h 31m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7e3fbe2593834348823e6592cea9706d : SUCCESS in 23m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd1212eedbaa4956bce5e130aecfc20f : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bee1f53c92224910889d9418b6f1a93d : SUCCESS in 1h 50m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4ed8c62213d4d9d9116010f012c8205 : SUCCESS in 1h 16m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5525cace8e9442b58dd6777a21f9e2a2 : SUCCESS in 54m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c5b2b24a89f44c3a57f2d4d0bbcf0b5 : SUCCESS in 43m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d054336acfd343d98860a2814cad4393 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cd1fb71143d457f8068cee88401b55c : SUCCESS in 30m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7fdad7c6f474424fa53ec5bf24e3fcd4 : SUCCESS in 22m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a17b496b5aa24dc89b97f41c6c792566 : SUCCESS in 15m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fce69f9d5ff04b5390814c074931459b : SUCCESS in 20m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8457267947f54141957815c52d3a5d7f : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"01dfa396304d82a26e701ec8b3419fded21a7730","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-08 15:21:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93a480cc09e320f0d0171ab959de0d24e444eb06","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-08 22:38:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49b617ea386ab5af7b942a3b77dac775a87af277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 22:39:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1eee6a5a09e72c660565ee77925825fcb6cf0a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:37:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d0b892df3441508e9328d90c4a5a14 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd32d4caa9974ff0b732e7854fcd3b6a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32bbcb45e6444466bbbcd69e0ebf75f8 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48083801cdeb42529eb26cb186649363 : SUCCESS in 9m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/887a60202b9242b18012c3e53608f5fa : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ecf932a2bee4271a300c9ec58f455ab : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1155afddea6437f8e96232deee29b2d : SUCCESS in 29m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9047ce08548443d1ae4aebef9a7b54d5 : SUCCESS in 56m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/192a3ee27b21434b92b5517ccde7b116 : SUCCESS in 37m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/90be3c6225514605992f2991b17f73fb : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9edd4109d3b8470facebe1886e8c90b6 : SUCCESS in 21m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/869ec4e2c55241f1a9800e20d134cc66 : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1e903c4f578f40f8aab9d068daa4baf6 : SUCCESS in 1h 53m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc039bd69c749daac472e1ae9e30ec9 : SUCCESS in 1h 14m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6f3ea13bfecd4cecbde4b5670294621d : SUCCESS in 50m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/decadccb906b4952be28133570e21181 : SUCCESS in 37m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f3ad7a44d0e544c1acb337ae72377a77 : SUCCESS in 30m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30eded446c084ce084c05bd27266792e : SUCCESS in 23m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fad41ab0785643c7b7e0a6d56d72ff3f : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a007c4061f1f46cd8227fe31bdddd388 : SUCCESS in 11m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdf84ccd90b948e3b3858dde3fb2190c : SUCCESS in 19m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60a931a4fc8c464fb984ae6e8812c039 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e57e71f166c025481636ad48bf8beac72ff0155e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:37:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86b1729025304f352b81e59e8f20f187e45d464c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:39:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625c0b8c9126493684d35fad907b24a1 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5f0a17acf57e953f8a71f95747675597c22ec70","revisions":{"b5f0a17acf57e953f8a71f95747675597c22ec70":{"kind":"REWORK","_number":1,"created":"2021-12-07 13:35:31.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/62/820862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/820862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/820862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/820862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/820862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/820862/1"}}},"commit":{"parents":[{"commit":"56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f","subject":"Merge \"Add dependencies for EFI and LVM based overcloud images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 13:35:18.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 13:35:18.000000000","tz":60},"subject":"CI: Use correct TD agent repository version","message":"CI: Use correct TD agent repository version\n\nChange-Id: Ic8c212111301d6b144e3aa0fe124b022b0296ec6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5f0a17acf57e953f8a71f95747675597c22ec70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5f0a17acf57e953f8a71f95747675597c22ec70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
