)]}'
{"id":"openstack%2Fkayobe~821129","triplet_id":"openstack%2Fkayobe~master~I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258","project":"openstack/kayobe","branch":"master","topic":"ip-allocation-skip","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-18 14:32:19.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-17 12:11:43.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-18 14:32:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258","subject":"Skip IP address allocation and configuration if needed","status":"MERGED","created":"2021-12-08 21:01:15.000000000","updated":"2022-03-18 14:34:06.000000000","submitted":"2022-03-18 14:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821129-ip-allocation-skip","meta_rev_id":"85939754c3fa7d197ded65f4e64852b887dfa274","_number":821129,"virtual_id_number":821129,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 14:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-18 11:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-18 11:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-03-18 11:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 22:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-09 08:12:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-12-09 14:03:20.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2022-03-04 09:56:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"1576203811c3f5af6d5190d21b9ac194680df3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-08 21:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4aed287b9d7ea6ce0f808867f2410d6532368266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-08 21:03:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"edf63b2c1c083f76f1bb274ccbd5058ca1aefef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 22:59:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d662a4b3442b406c8b1f484dbfc13b2c : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5a21f9d8844e52a5f937e912b489e3 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e9ab9f97b7f4880b5b3a064c203672e : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fd63eff580746b19371d2098946f71b : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528930ceb4d445e9945ddf963edd96cc : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31590c086959433e8c521342c7926723 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca18756f7fca43b8afa12e72fb3a242b : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1dc34ac96db84a7ab5f5b98070d7afa3 : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/915ffd4a042b4e56a7f6e59479a128a1 : SUCCESS in 25m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/899bdd55f5104deaa27417b270d8fb7c : SUCCESS in 1h 07m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c70297cdc2b948d8952e788393748850 : SUCCESS in 56m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3ec9d15c552a408c8015d2a287ac71be : SUCCESS in 1h 15m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b59f9475656b4660abbc21122d819eb6 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d923aefe85bb4533866f84d5a525aef9 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24658b01af704c6f80917a09a016e7c3 : SUCCESS in 1h 54m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18bb668127824a2b8004f3d7cc056e74 : SUCCESS in 1h 16m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84cc9ad8893d42a59458f5a9fc80af8d : SUCCESS in 51m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3ec41f8c224c45b8f918b058b7bc5b : SUCCESS in 40m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca604eca475a43ccb153b20363ce9108 : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68043fcbc64149d0a65a6d8fddba34ad : SUCCESS in 23m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9654d4477f7e44148df7df3dbac22de5 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/432d7630e0a645dc9d49229031765fd3 : SUCCESS in 13m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/497b64fe35744dbe98d7354d728a1e53 : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9288cf6c6f54e6283960e8f3bf82742 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b3754bb5f80c2c81280273e83fb8c1a35753db4a","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-12-09 14:03:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f5f0deff72d1688e7cf2450533f12b8232848b4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-21 12:06:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c718db875e46583b491416896a9d0eee203834b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-22 11:35:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5df0c9bb107f3754f1a92b40520fe4fae45c230b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-22 11:51:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a82ac3fdb6f544941a3d90bcc0e110f371bde793","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-22 12:17:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ed4420e0d7c2741bc90ffef5f2502babaebd2c8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-12-22 12:17:18.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9080e1a47e8eab9552a83fccfbdd8c35127b7d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-12 11:58:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c733f7d5052095b94039bce806986cf1b230d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:07:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/860a4a73831949f4acc938d7990059d5 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f695219ce0f348f2af823a5fd951e0e3 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f886c713fc4944fc95758057f586f757 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa15c4e2c9e546868bb2438dfe8cef5a : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350e284efc7a427fbf783ec99a8c9fc7 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2df19a936b94623b3ebfb1d5f112959 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ba4911c890f45ab888d20c71c8acbe8 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d29f77d23ca49ce81e092bff538f4aa : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97477ea427774b39bc5fda88157761d2 : SUCCESS in 29m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fa0cb82898de454cbf2c0e6da56392ce : SUCCESS in 55m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/293b680da700438d9d868e6d7f18fd47 : SUCCESS in 44m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9e6c0ead4d6d4b48a7f378f70dce1622 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/949b91810e764cdfb5c2651538abe549 : SUCCESS in 25m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca472ebc671346658e98296c2a59d41d : SUCCESS in 19m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66e51239b4784bba820a671615dbce7b : SUCCESS in 2h 02m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05f3385b777440689ec0d4ce5553b959 : SUCCESS in 1h 12m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aa97d49317584c39baa64ce674e1f9b9 : SUCCESS in 52m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cd41d69675e4989bf842e0babe12aac : SUCCESS in 45m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f90b0e3b4eb54679830b65b58f360adf : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41166cc14986439584452db17054d5cf : SUCCESS in 25m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9524b20ca01142178fe3082c7ec512c2 : SUCCESS in 18m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0af7cafe9664b65b0578d5ff963e09a : SUCCESS in 16m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/989cfdef2cb44f32bc3955c48d206698 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f877ddc4218543a69d2f142ae15559b3 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":3},{"id":"5dda276b8a1af31e682bbeb83c377bdca1c53ac8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-20 08:32:35.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"59d8529ed71529d46041269e5972f424716625bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-20 13:38:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0c5c7837b0eec565a5ac5cba853a1f549635ae7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 15:34:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc61be106c0c49e589efd9a61cebed4a : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb84899fe8d54939b6c9c18cfa31c505 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13668b8b59e841089e7da4ca97f83719 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d9b992bebe4f46972e3ec58219837d : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d45c80840c6460ca6c32e52bbac02b9 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab0cc0abd7bc44b78fcda11fb3f2a076 : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73beb166a90b4564b4ed555b639f2f0d : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9729c712d314e37b1afcd574a2d4dac : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00631eb8a90e4683b4f1f9e72e309748 : SUCCESS in 34m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9c69d3af0d242338f4ef7d6fd1f6a35 : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/596814e25e15486bada401dc5a1d374e : SUCCESS in 51m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9915db71532a408a882bfde0242fd500 : SUCCESS in 1h 14m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6d4ed67b18ae488ea6007aca5f018828 : SUCCESS in 24m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c7efe7ca7254ec48e17e57b55a2ec88 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2ed812572c047459e755779e9884543 : SUCCESS in 1h 53m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9423a5ab281940e5996ff867a2b8c171 : SUCCESS in 1h 17m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/956e50e5270d407c91e63b98bd594327 : SUCCESS in 52m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a2e7320e3504413ba1a788073e3d3c5 : SUCCESS in 36m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/02502cce7ee748cf9ba83c5533d11e0e : SUCCESS in 26m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f2a7460dae14560bddcd905abe0ac81 : SUCCESS in 21m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d5e998af18d4fe0b8627c6a94a4bd38 : SUCCESS in 17m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1599e08864134708a9ff393cd73b35b2 : SUCCESS in 11m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fb8e17a28d64551867813e859005d33 : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42caf7d397ac4b5cb4a7b199a8de04af : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":4},{"id":"8ef3fd34e33e5e2ad5e9404a350751e3412641fb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 17:20:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ece6c985b47d8b3ff7f5e46cbdd92d503063c20","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-02 17:55:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d8c5b56cbe9f2293830286f537d2c9fe2c05564","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-03 09:44:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9780ffddae70471216ae9798f0360795dfefc5b8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-04 09:56:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8c1492e5aad559cd64823261ef58c5c56cc828e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 10:50:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3d2e43c4a2a6580bc5354bc0203414a2917dec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 11:49:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"23fdc6e8df6f445166ad75098fa0f2d5eb253b5b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 11:59:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a408fa90558b6b9d9129340dcc99934a7e50f86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 12:01:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"21e78c7982f0c989ed486d1680d7170fb514a728","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 12:48:41.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"f4eda71d6433bb895e11b16a7a068a0376a876ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 13:29:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d58d5c0d9424b7c44cc42544167c7096d2e3e84b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 14:11:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b0a9e89f72de9d6d404ac2ccd873dc02dc06f186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 16:01:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b80b81070954db18b2a3d0af93268b1 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a088afe62c4a9aa8117b4c9637ac71 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f614507bd6dd4c59aeb96c461839fc78 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/544a0d7a19c24b409b4a30c25a0ee7cb : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998dc6741ae8482fbab40f6695b568dd : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f502cafbdef40859ab4f6b7ab8f0217 : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07aa5a84598d44dcbe542244b2dbba69 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19cc4ab861414665af196dda02929b2f : SUCCESS in 15m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a47a4450a4744bd7bc12758006c5c684 : SUCCESS in 32m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a09b14122f524e27a5934c9c91041ce5 : SUCCESS in 1h 12m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76a401ae81474b83b6cd45b6ffd8d060 : SUCCESS in 56m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5173b9cca1974dbba26c8398b975542c : SUCCESS in 25m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f4676cc15b48f78e1d2704c58b1fd5 : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2688179695b247ddaa52162344c4634b : SUCCESS in 1h 46m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea403c47f644c8fb0e555e98bd6aff8 : SUCCESS in 1h 19m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/60bc38e1307c48e7a61f974f1c1a8e31 : SUCCESS in 56m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5760f69623e3409192b80643152fe2d2 : SUCCESS in 38m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/944916b4e15941e2b8b15d2e22a4b68e : SUCCESS in 46m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/429c0a3c339f4efb9ca3d8808a66297a : SUCCESS in 33m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b6cb02b0c996406a9eb6d9d135404c61 : SUCCESS in 18m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d312bb6ab642a6a5bf2a9032576ba3 : SUCCESS in 15m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/96e5ac10940d4890aab331a4a7cbb49c : SUCCESS in 18m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a565f77fe4874753bc465221464fb7a0 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":8},{"id":"795e9c169e5afcb72e3f167340edde2cdf2beed3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 09:46:46.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e1e45dd752173692f3afd4ab09f2a0a28cffcf46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-17 10:02:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"95a88c819927833801c01ddcfa392d699ba83215","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-17 10:02:54.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d8249fddac64240d9c802fc6dc24b5cf62407770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 12:06:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0bc0c9e5a364a278fc1117b4433da75 : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e23bed9b70483eb859f92c260e24f3 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aabcc070e1e54fab83f4bf4bd9ce9d0d : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db1d76cb56b14314ad8b65489cb0379a : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e405f598e284abb949cdf7616543a55 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faece72b43854fe8b68ae772eee6b4d7 : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6305fc5658d540cfbd566544d09f2522 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/014dff44a50b4ec689e147c3a78df0f9 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/533d24f5cc2440f099c80264b2a0b159 : SUCCESS in 32m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e6627608d661484ba5250ad35678b7dc : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/906350b7bf9d4f708ff5d64047632373 : SUCCESS in 56m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/97547f72c3004350b89411d7410ecb67 : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c08a5bdd1c644b2d904c8de06f647b12 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d4e115e649d41509e5035482495a52e : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44ba2dc724b444b181b398dcc99db9b7 : SUCCESS in 1h 57m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e9416a84d4745349e45f4c2facaf3c9 : SUCCESS in 1h 16m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e453c946437c4933901da16fd80f22e0 : SUCCESS in 51m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a4f5c1ba6ae4372ae50ac95dd05f4e3 : SUCCESS in 37m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e68c845d08345e2ae112c41bd9871a1 : SUCCESS in 38m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/050e78a3683d4b94868a94cc3d4be113 : SUCCESS in 33m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d1d8e53a28944868a9a300dce948a6ca : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f8611cfdad145ab957c227611e6dcc9 : SUCCESS in 12m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb120ca7248540389f536465ed82acd6 : SUCCESS in 19m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75559a4f58ea4a908cd3f08235e8fc96 : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":9},{"id":"61f25b2c5125c73acc4242ea4ca8c0600ae46ae4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 12:11:43.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ead8ae6d5860eb3dc12b4dff74210a8feb5bb3c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 10:40:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a2e4122402ac230112cca970cf758a024d8fe73e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 10:57:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f71be64e6d60f2fb1c973dd6473a384b77c67d87","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 11:25:00.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"762524eeb33eaf95af22111ce7c90e4b850e32aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 11:32:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5507b50197f38c80195572027dbdf994ed2dfcab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:26:30.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61115e03392c452fab8cd2d09c24485b : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e0f3f395cc4d6691cdff7b6133cd4b : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f84089c595bb4bb08fe8d67a611ea455 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f400ebbf82a487e84056420c2ae82f1 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65e41f40c614a839d9171534a1f3f8c : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd22e64279f94d5bacb848574d2fc78e : SUCCESS in 5m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a0c154ff1524d89b5c6f2f7fa6777b5 : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4433ab96aa294324b0ede1686f2d3cfc : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/368b9ae85798415b8519e20e0aa844ac : SUCCESS in 30m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c6a3dfe430164a80ba7a020c1bc8c440 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cea390bfe9e9438698726946eea78c1e : SUCCESS in 50m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c3cf59ce3cb248ab9adbe43d1e185f7c : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ccb60ea45f3b445491f2825baeef65ef : SUCCESS in 24m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d76dbee23074207ab750bfeb73dc25d : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5088dbba4bb94d859284061a49ce9c0a : SUCCESS in 1h 41m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80eaf49a426542fb96d02d29b2db5ff6 : SUCCESS in 1h 10m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8c0f2385592c40e79e5a123fb7bd6e23 : SUCCESS in 51m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fd70af9c5b4f54835f2cbb64cc5b26 : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e0a4bf8370f44d3b226fbc2512c8d51 : SUCCESS in 43m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/673a1da3adb74ca3b398ddc6c7719396 : SUCCESS in 32m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/64740cd2cf814570a6e56e8158790b5e : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4fac65d22f1432d905dc3cf326f613f : SUCCESS in 12m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fe0f0e26998f4967ad47cc104a74f11c : SUCCESS in 28m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b0a0bd28019431896bd35811c7497ba : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":10},{"id":"0c4c6a43acaa87c9703117fae1315a62e349a849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:27:03.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b0aa793a4dd19899387634b795accafcd59ce1b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 14:32:18.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7223c98a24564be2b0ce46119b3038c7 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b3ef0124b52494590e578b19122f3de : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67911c8e91e54cc79569f4b5e6197089 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37db36ef4ff495db7802122e04cc522 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47ea30e265d24febaa90aefe450781b7 : SUCCESS in 5m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb3c9def3b8c46248a96fcf3f73660d3 : SUCCESS in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc8b6ec5b32d4014b789ad0e38dcfcc3 : SUCCESS in 15m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f5e1b0fddd842bd9ffb7e0b0ea7dfd0 : SUCCESS in 34m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/76110b3a21c046c19092d67e3554d90b : SUCCESS in 1h 16m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb836a9e0ff14849bf68035663eecfbd : SUCCESS in 55m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/599c686a3052412c9164b23502409634 : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/673214da7d084c44bde8dd6d863a4770 : SUCCESS in 27m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41f223776c08410e8631973e52c6abd1 : SUCCESS in 15m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/688a6e3596154264b924f7f5abf2340e : SUCCESS in 2h 00m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a781c8f86d1e4690bb08b0780c8f7498 : SUCCESS in 1h 25m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fdc74f1b5ccb4beb80a7e4be1cc43db3 : SUCCESS in 55m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58ffac6ae80e427cb7dceacb8629de24 : SUCCESS in 42m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d465875c5bf546b189f16a274e12e2e8 : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d4c24d0525e4c4381599570c543d01e : SUCCESS in 39m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e687459c750d4776ab490e492e66e51c : SUCCESS in 21m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d25c70c6b92490a89e9a6407d98cfaa : SUCCESS in 15m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2b79afc2c92a4a85a13316279a286256 : SUCCESS in 20m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/113b3976ac074b9db6b7545ae501ca28 : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":10},{"id":"740bd0b05e49904a7067b435b387f12452d93a9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 14:32:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"85939754c3fa7d197ded65f4e64852b887dfa274","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 14:34:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6d258aba404966a963c21971f9d081 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f391e7272887473ca9be15118ef2f1f6 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1de4f2a4a3f6ef16399d812a544c6da622609e24","revisions":{"32d438666e391032927bd5e6a121666f3d27f9cd":{"kind":"REWORK","_number":1,"created":"2021-12-08 21:01:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/1"}}},"commit":{"parents":[{"commit":"56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f","subject":"Merge \"Add dependencies for EFI and LVM based overcloud images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:56:54.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the check of the value of the IP address set.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32d438666e391032927bd5e6a121666f3d27f9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32d438666e391032927bd5e6a121666f3d27f9cd"}]},"branch":"refs/heads/master"},"332473cec7c56cbf9ac126e88cb1d05f3eab293d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-08 21:03:04.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/2"}}},"commit":{"parents":[{"commit":"56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f","subject":"Merge \"Add dependencies for EFI and LVM based overcloud images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56a01a41781c1ca0b704a4f9c12e4f5bb0dd302f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 21:02:47.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the check of the value of the IP address set.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/332473cec7c56cbf9ac126e88cb1d05f3eab293d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/332473cec7c56cbf9ac126e88cb1d05f3eab293d"}]},"branch":"refs/heads/master"},"921e05b4d76ee96ea1cf5f3c91ccc23ca966cb52":{"kind":"REWORK","_number":3,"created":"2022-01-12 11:58:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/3"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 11:58:31.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the check of the value of the IP address set.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/921e05b4d76ee96ea1cf5f3c91ccc23ca966cb52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/921e05b4d76ee96ea1cf5f3c91ccc23ca966cb52"}]},"branch":"refs/heads/master"},"dd4ea874473664cbbde56bffbf976019c1ad8ad7":{"kind":"REWORK","_number":4,"created":"2022-01-20 13:38:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/4"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-20 13:37:58.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the check of the value of the IP address set.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd4ea874473664cbbde56bffbf976019c1ad8ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd4ea874473664cbbde56bffbf976019c1ad8ad7"}]},"branch":"refs/heads/master"},"06b283064f5424d95154222bdfefed92608f7825":{"kind":"REWORK","_number":5,"created":"2022-03-16 11:49:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/5"}}},"commit":{"parents":[{"commit":"5fdf643807b679e17ef4db7f990e682403522d6b","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fdf643807b679e17ef4db7f990e682403522d6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-16 11:48:42.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the new attribute \u0027no_ip\u0027 for the interfaces\nconfiguration. Also the change contain the test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06b283064f5424d95154222bdfefed92608f7825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06b283064f5424d95154222bdfefed92608f7825"}]},"branch":"refs/heads/master"},"4f2f709fc4b378f5654535067371b5036495cfe0":{"kind":"REWORK","_number":6,"created":"2022-03-16 12:01:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/6"}}},"commit":{"parents":[{"commit":"5fdf643807b679e17ef4db7f990e682403522d6b","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fdf643807b679e17ef4db7f990e682403522d6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-16 12:01:01.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the new attribute \u0027no_ip\u0027 for the interfaces\nconfiguration. Also the change contain the test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f2f709fc4b378f5654535067371b5036495cfe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f2f709fc4b378f5654535067371b5036495cfe0"}]},"branch":"refs/heads/master"},"745ef2849098b9156de5effbf05fa756d27206e5":{"kind":"REWORK","_number":7,"created":"2022-03-16 13:29:51.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/7"}}},"commit":{"parents":[{"commit":"5fdf643807b679e17ef4db7f990e682403522d6b","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fdf643807b679e17ef4db7f990e682403522d6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-16 13:29:28.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the new attribute \u0027no_ip\u0027 for the interfaces\nconfiguration. Also the change contain the test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/745ef2849098b9156de5effbf05fa756d27206e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/745ef2849098b9156de5effbf05fa756d27206e5"}]},"branch":"refs/heads/master"},"a46962b7977423abf43f474034e48da7897637bb":{"kind":"REWORK","_number":8,"created":"2022-03-16 14:11:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/8"}}},"commit":{"parents":[{"commit":"5fdf643807b679e17ef4db7f990e682403522d6b","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fdf643807b679e17ef4db7f990e682403522d6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-16 14:11:22.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the new attribute \u0027no_ip\u0027 for the interfaces\nconfiguration. Also the change contain the test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a46962b7977423abf43f474034e48da7897637bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a46962b7977423abf43f474034e48da7897637bb"}]},"branch":"refs/heads/master"},"346d31b1b43739623ab1ce34c526405c9d71d6c7":{"kind":"REWORK","_number":9,"created":"2022-03-17 10:02:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/821129/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/9"}}},"commit":{"parents":[{"commit":"5fdf643807b679e17ef4db7f990e682403522d6b","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fdf643807b679e17ef4db7f990e682403522d6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-17 10:02:08.000000000","tz":180},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without\nany IP address set (e.g. bridged interface) and to achieve that this\nchange adds the new attribute \u0027no_ip\u0027 for the interfaces\nconfiguration. Also the change contain the test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/346d31b1b43739623ab1ce34c526405c9d71d6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/346d31b1b43739623ab1ce34c526405c9d71d6c7"}]},"branch":"refs/heads/master"},"1de4f2a4a3f6ef16399d812a544c6da622609e24":{"kind":"REWORK","_number":10,"created":"2022-03-18 10:40:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/821129/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/821129/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/821129/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/821129/10"}}},"commit":{"parents":[{"commit":"a1ed4a05bb51e89abf7841715dee8251f9fbf53a","subject":"Merge \"Use naming convention to infer VLAN tagging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1ed4a05bb51e89abf7841715dee8251f9fbf53a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-18 10:39:40.000000000","tz":60},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without any\nIP address set (e.g. bridged interface) and to achieve that this change\nadds the new attribute \u0027no_ip\u0027 for the network configuration. Also the\nchange contain a test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1de4f2a4a3f6ef16399d812a544c6da622609e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1de4f2a4a3f6ef16399d812a544c6da622609e24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
