)]}'
{"id":"openstack%2Fkayobe~821176","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ic8c212111301d6b144e3aa0fe124b022b0296ec6","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2021-12-20 12:52:45.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2021-12-20 12:52:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8c212111301d6b144e3aa0fe124b022b0296ec6","subject":"CI: Use correct TD agent repository version","status":"MERGED","created":"2021-12-09 08:56:56.000000000","updated":"2021-12-20 12:54:47.000000000","submitted":"2021-12-20 12:52:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821176","cherry_pick_of_change":820862,"cherry_pick_of_patch_set":1,"meta_rev_id":"92e02be24afabbfc1bed1c31483f42bc27d0fa23","_number":821176,"virtual_id_number":821176,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 12:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-12-20 10:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-09 10:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-12-20 10:50:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 08:56:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-12-09 10:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 10:50:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c0c35bebac3e2c7f27b2107184594330fdaf323d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-09 08:56:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"451aa5e7952557d7c3124bdb121d2f0959df2a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 10:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b28cd7cfa8d475292d08642c1be5278 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef91cea1b2cf4a0995f4206a24205b1f : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec5819c3d8894f04b89533f29fdc86e9 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b639849b44074ecfa6b9a7f26dec4ea3 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dacc302aefe94c2893fc174e4555bea0 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69385864c49b4b1b89d1c83df9add067 : SUCCESS in 11m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca35ee58e44e46228708f9de25523d08 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/197794df04834bf08e105f4e0f054701 : SUCCESS in 14m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe1df39a71b74770a37deb27712fe921 : SUCCESS in 27m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/85d23a072ebc4b2a92688c8e8ca883fa : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c4c0aa2e302492080855c453df7bb6e : SUCCESS in 1h 24m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/18e9b8284d95479fb36cf2d3dec9ebf5 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ddc7b1efa3464535867c172bbe466cc5 : SUCCESS in 25m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a58da136b29d4e77ab0a359e201e51d1 : SUCCESS in 16m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2fa9bf61ad6a403f90c7cb6cfa244cb4 : SUCCESS in 1h 40m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89235c61d874417b9852ca28bd842dd9 : SUCCESS in 1h 13m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3917e1e5f45648eb9a47ddfa0bc5d533 : SUCCESS in 45m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13d2b9d8a27d4548b097d055d293f85b : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/87f84cf4f53a4c138a20458005e191c4 : SUCCESS in 33m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9e040f9b75d4f21b89eda3e96ef9947 : SUCCESS in 33m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5dc3c54dfe1a4ae183d22dabb341c248 : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c2e5fbf3a064895aae61f783b890960 : SUCCESS in 12m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/530ae3489e39417a9578067b94c7d4ba : SUCCESS in 17m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129e7d683b294d89b7f8a217c2172c4f : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"87d8040c55afe0178222bc316d24a7d86a6dcf75","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-09 10:46:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1152ba3038dd53347affadf5f08a6f0c0ade4ad8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 10:50:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7f07b58802a0b813657bb989c114aad61779289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:51:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02ee394f5104072b294b719c2f8cc2401dde4d3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 12:52:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3abf36ac2144d928fabe7f3fc9acd85 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0a5943b8fa4e89bfc0618cb7993533 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7572db9b298c4faca16630a6a4c60156 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9d4c2685f14100965c4985925fa4e1 : SUCCESS in 9m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/078e53124d9a41a68c002c412e3da913 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5008214301cc4381b9ab136fd615e0f8 : SUCCESS in 17m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2915c9d25464b7d99966d112593a26e : SUCCESS in 30m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5cb2cb5cb1144064ac468d8234afb4f4 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c89fe73d65c444d993dce72d90aa419 : SUCCESS in 45m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fbfdbc61b0474d02b312d2b830203095 : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c1d403bf021e4c409a9759508514a0db : SUCCESS in 24m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332b2771b4e44316b907e0ed3b930e52 : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b39fe178f2ea40b89ffc5fc7c34420aa : SUCCESS in 1h 41m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6568061438b541bf9d2f670f81416260 : SUCCESS in 1h 20m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cb589a4490b94501bf6961225cafc3ba : SUCCESS in 1h 02m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d350f87458b4239b31d6fddb3aa36b6 : SUCCESS in 31m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1a8e88c79b474443b803aba911943bcc : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf4611af7444566bbc23a4fe2c3b65b : SUCCESS in 23m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4612f0953b44674a7b47681d045d022 : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8030d6bbe2c240c08fcb1a2aec4bf7d6 : SUCCESS in 12m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d11bfa30a5740c5bcbe3384ef762bf0 : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0525a012df7466dbc3e6e71d31a70cd : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c93ffef6d55030a03d8986e2fe76948ef015d603","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 12:52:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92e02be24afabbfc1bed1c31483f42bc27d0fa23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 12:54:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be689d5de3ba4ceb941fb25262c661df : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fde9b65ce0868a9c45cb1970aa036864a5943ac3","revisions":{"fde9b65ce0868a9c45cb1970aa036864a5943ac3":{"kind":"REWORK","_number":1,"created":"2021-12-09 08:56:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/821176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/821176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/821176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/821176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/821176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/821176/1"}}},"commit":{"parents":[{"commit":"8e6a66bf18a389b56d725eb2d1062e95e0ae70e1","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e6a66bf18a389b56d725eb2d1062e95e0ae70e1"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 13:35:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-09 08:56:56.000000000","tz":0},"subject":"CI: Use correct TD agent repository version","message":"CI: Use correct TD agent repository version\n\nChange-Id: Ic8c212111301d6b144e3aa0fe124b022b0296ec6\n(cherry picked from commit b5f0a17acf57e953f8a71f95747675597c22ec70)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fde9b65ce0868a9c45cb1970aa036864a5943ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fde9b65ce0868a9c45cb1970aa036864a5943ac3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
