)]}'
{"id":"openstack%2Fkayobe~821177","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ic8c212111301d6b144e3aa0fe124b022b0296ec6","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2022-01-04 11:59:14.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-04 08:38:21.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ic8c212111301d6b144e3aa0fe124b022b0296ec6","subject":"CI: Use correct TD agent repository version","status":"MERGED","created":"2021-12-09 08:57:05.000000000","updated":"2022-01-04 12:01:24.000000000","submitted":"2022-01-04 11:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821177","cherry_pick_of_change":820862,"cherry_pick_of_patch_set":1,"meta_rev_id":"da787ef0481109cd3b3fee9bbc4382d0f77883ee","_number":821177,"virtual_id_number":821177,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 11:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-12-20 10:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-09 11:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-12-20 10:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 08:57:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-12-09 11:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 10:49:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"886b20173ee102fd0090be21617b00bcb8b11611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-09 08:57:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a12f6c1b9345f444bb7cfd57c433126ead1c81eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 11:03:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e10846a10b74e8997dc514a11e946d7 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c7848e8be3437eab27541f8b19491f : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d065bc3e5439464ab23b69ea12b3db21 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505ade0fadf44c14830d6d6270bfd11c : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e93c4aeab87b4a2c9ddbbfd41ec2505b : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1bebe46442437c91837f67f0476f27 : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4755337e8184bc6816f9ce4538e11f3 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d891bc383e6c4823bd1657057ba5c452 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aaf0692fdf24477fb9ea46e8e70b49ba : SUCCESS in 22m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1a45f605fae94f7e8d582986e84fa723 : SUCCESS in 1h 10m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d72aa9e0172a45d7a90ba71d6b36ba0a : SUCCESS in 48m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fabb2ccf71e34c2b9e6a9c65358547d6 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/19d0c215b04745309e79bc3039056efa : SUCCESS in 16m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1d6193e1d243fca7a972f2e03b9d4e : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e1cdfe3abd94470bbd81e4cba0d7d30 : SUCCESS in 2h 02m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/759a540a205f45d6a129dd8084e4f461 : SUCCESS in 44m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b576410c0d648f6ba0a1e56ccf4941e : SUCCESS in 26m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f87491231d5c4f1787188de78f693734 : SUCCESS in 31m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f4dbb9acadc64035bc4af6d7bc836aec : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8912398448d74eba93a109cea625890e : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ac32a103f8216f2185b8ccccc359b1c449e74801","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-09 11:41:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a85b5ab8175aca9f2d103af4f443cc102a6372d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 10:49:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37fe91d78e7d2fd868422cdb70a2fa9079f6ba33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 10:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb2e27011ea1484e2cbaa95fd08a6f4712e25765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 13:17:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee600b9410c4fbc82bef8a44214494b : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/803d82e66ab6438fb1d5513679818bae : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f284b1f033a24856917be7a36a6a47f3 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7a74fc0b0c4d27bb8f75258ca627e3 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fca9de4fd90c415495f8faeb2de60ea9 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42b14b2764ca41d1a4555531fda5f60b : FAILURE in 3m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ef9bb9146f24c4f9fccce2b81c25c9f : SUCCESS in 21m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0dce0200a69b4ed6b0d6e7d9ddeeb4e5 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ce61d14a1cc4bbeaecb13d7a4f0b6f7 : SUCCESS in 44m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c33ded125dc745f9a00f4a87ecbe776b : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3f9f6a154e14a5f86eaf5cd08642206 : SUCCESS in 19m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883e9d868dbe45e2b465af62a5b8765d : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/42db3d664d134b7498de82fe815a0748 : FAILURE in 1h 11m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b17219d580874b0bb08bb94b022c408b : SUCCESS in 1h 25m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92f7c8c2db64415a49746af937b116c : SUCCESS in 28m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/545f475591384c64a1e9ec0afe2888d3 : SUCCESS in 27m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/37d800ac67db47e5a8e6f167d66d8d3f : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8cdd75f1a8d4495b4934176c06bb6c2 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dd1b97e9fa32cfe28748f6fee7cfc5af3fc7c885","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-20 13:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bcc08a39d7def7e09cb70b6e48a8b3a18c16b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 15:14:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2776728b19a4795aec4dffc5899622f : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec287403d4d4e669f9ea2cf7e350bfe : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2113d4cc7144bbab43a2ef836c46f24 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f429d6ead7ec4b439e2e3baa416e8ae4 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ff6aa4666304b59a6db6bf7e8ed0e0a : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a41b62852624cadb370a4e051524a46 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8e33644f356435abd49a3b9e6d5d9e1 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/164d56d2c5e74cf683889de6ac98bba8 : FAILURE in 3m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2511095e6a38494cb28c1c2cb08bab6d : SUCCESS in 18m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c8b3114cdf344fc6b2016bee09ff1980 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7f3fd93024c41d8a3c229e1a699bcd2 : SUCCESS in 45m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1bd397ab3f554a478ef6a44ad24fa77e : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/acc0a1e9620e45b5bb7c21f1816888d1 : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8169c6d058c4f8d97a144e74aaaf14f : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fd0a9e9f4c24c1ba26138f57792b512 : SUCCESS in 1h 47m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6cd6bc16e270433793e6995ef73345c1 : SUCCESS in 36m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/512c6bf0e9a84c44974f678a38e96312 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c85420439629477c84771a1a8b87a16a : SUCCESS in 25m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c82132997d55432198cf57548c62bbbe : SUCCESS in 14m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd722a4ab434ad992c37b3138d48bb4 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4a308617fbbfa66cbe46d74afb73d5b5acf93fcb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-04 08:38:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88af4ba52148f57399131fb2ef8e99f5734f600a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 10:14:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63154d80dd6141d4bbd93cd4d2d1e2bd : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a8b40b0671e4f13a4cd44e7b35e5f1d : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cee55ec7cf84591bd372d31863118eb : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7de96d8255b542cf9b412318d7b9e0d0 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0291114185904045b1fd0359b7f45179 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd294504b354c91be1c9f5c1aef37a7 : SUCCESS in 8m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c11bdadd4414a84a79500ea7a5e3fff : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d2464ee0854448e9a41e57c3e650d92 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6d350414e0e4c4282ebc2184b19d0db : SUCCESS in 16m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e702cfeb9be412193b6ebfc9295b00a : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/700e8dfddc1d401d8627ee5996b88424 : SUCCESS in 44m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/37ca44ac4cb24fcfa6939443b1fed266 : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/764349382af24b7c97dfd804d1ce2d48 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce83ea62b724dc7beb6fec4d605024a : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c36c36dbc5e540ce9235abcd1933604e : SUCCESS in 1h 34m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/af4e56777f1a4a85a54b26db78104db5 : SUCCESS in 35m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7acb7df3679843f69df797428b4d051a : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/726f4bb573374df5bb28a136ad4c59a5 : SUCCESS in 25m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/41a7170081eb493c869530bd230b3df6 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0556a52038944b49bfd4bb5af8989abe : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6c175023f19f63993f4b550a90ffd568508a1573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 10:15:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06bf87d0f0acce0ade8723615cced0025eea7421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 11:59:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2731fa806c5343428152b1b3837e1f71 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7261e5fc384bcfb05f273d4e4763ca : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d3716edaf8463e8c26c0702d07d312 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e353e23c30584a3eaba63254df1ddc8e : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d56a71634b464a05aacd469eac8041d1 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83360f1a61cd4d46aef5d363bbbe93e5 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/299d1d9a46fa4aaba08c960daa3c28d0 : SUCCESS in 17m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70253c0529c84794b0303b4ebf712666 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d6e4e955c344935bd2f94445f721e86 : SUCCESS in 49m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cee0086da5124dde84e52d4f6ff2d0c2 : SUCCESS in 1h 15m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a5cddf43967c4015a0440a530ad188a1 : SUCCESS in 22m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db941ade598148fca24629490d233c2c : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7ee31ff910a4a6281dc7ca233ba5e35 : SUCCESS in 1h 40m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e0fc05ea35aa43c2944a46965f85bd18 : SUCCESS in 39m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5202276a586d4fe08a213e843241bb03 : SUCCESS in 35m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/38ee4d0dd3fc462e86556dd1f34e1ad1 : SUCCESS in 28m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e8d0faa82bb6439f98e40ede183e6c24 : SUCCESS in 16m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c84b3c9bd341939d81af6e9733253a : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2d9043d9cf8c0d4e5cba4c32b287e25974ab443e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 11:59:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da787ef0481109cd3b3fee9bbc4382d0f77883ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 12:01:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07464f9ff9d3437689ec3d6d7c87aff1 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43451c49d23f04d33c932db916419feca1544bf4","revisions":{"43451c49d23f04d33c932db916419feca1544bf4":{"kind":"REWORK","_number":1,"created":"2021-12-09 08:57:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/821177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/821177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/821177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/821177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/821177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/821177/1"}}},"commit":{"parents":[{"commit":"4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59","subject":"CI: always return host configure test results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a76d8000f12c4ab63f7cb22b8bb1c6ff0529b59"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 13:35:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-12-09 08:57:05.000000000","tz":0},"subject":"CI: Use correct TD agent repository version","message":"CI: Use correct TD agent repository version\n\nChange-Id: Ic8c212111301d6b144e3aa0fe124b022b0296ec6\n(cherry picked from commit b5f0a17acf57e953f8a71f95747675597c22ec70)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43451c49d23f04d33c932db916419feca1544bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43451c49d23f04d33c932db916419feca1544bf4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
