)]}'
{"id":"openstack%2Fkayobe~822758","triplet_id":"openstack%2Fkayobe~master~I96eb4685f997e85ad2ee5318640d58d0287a016d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-09 16:45:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96eb4685f997e85ad2ee5318640d58d0287a016d","subject":"Disable EPEL by default","status":"MERGED","created":"2021-12-22 18:08:56.000000000","updated":"2022-02-09 18:11:03.000000000","submitted":"2022-02-09 16:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822758","meta_rev_id":"739d91502a1b07623b798d4cee41f9c3d71ccd0a","_number":822758,"virtual_id_number":822758,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 16:45:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-09 13:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-28 21:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-09 13:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 20:02:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-28 21:49:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-09 13:52:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ef18240efc581a820a9963f107569bcb233a128b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-22 18:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd478d4488e8fb74c0d7756b5bfcfe3b2e6c3111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 20:02:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a1a46613da40e4a1064471cbaec863 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b978cec2650d4000aacaa339b70a6fc5 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b926b79e46640a0a86f2d17d30e4076 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcf7ecef35e44adebd94e1a8a638ccd9 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e551700e0d32498ba2337870630d7afb : SUCCESS in 9m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b24af958323f4f29980d2aaf1414902d : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71a0395482f44db99a349800fecb49f0 : SUCCESS in 12m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/226c48def7da48f9bd50272da32cce03 : SUCCESS in 27m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99cf30d4c8f64e0c988d42e5000ad014 : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebf21a6ea3fa4ab8a3448a2da2c1b007 : FAILURE in 43m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fae873634e3149daa772538185f0185d : SUCCESS in 1h 13m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fe8ead4d99284bd8a83f0e0d129ffab6 : FAILURE in 23m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a45b9722009c4d7c8a802119271e6f2e : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8366d39c7dad4d51b40c0c2fdd5561ac : SUCCESS in 1h 49m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23f15e4eaa7c4c4ca4677529664ec4e0 : SUCCESS in 1h 17m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/19ff6c8bf2cf42fdad5ebba8686cef8d : SUCCESS in 52m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/892fb53767ed498ca0678d30cc8b32a6 : SUCCESS in 44m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40470e788f5f47d29b0b6fe80abffdb6 : SUCCESS in 28m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00a8a9723eb947b5bcf597ac38403756 : SUCCESS in 24m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e52a0bb6cdbe4d338466790d8a32ce3d : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ad84f9dc2042e1a203cef97f9fab70 : SUCCESS in 13m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/671a18e0db734284be7e0f0abb52a85b : SUCCESS in 17m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1931c8a0198448d87f213fcda11a816 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"86c5ea32f5534a43b88b827eb84f65b4a01608c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 10:48:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03b76e845bdd828fec7d9dfcb53bcef0fa5e7dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:38:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0106416ff654cb89258b1a178c3b405 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e624a68fba574e95a74eaaad3c69d93b : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4a482b8addd426284e526fd6a76b3fa : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f87db80d1b344eed87bdf7535deb40fd : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0eb3e5cdf542fba3b0479fe0af3559 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a0bfad27d6e4e9b9a277487aa247cf1 : FAILURE in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1973e69a532c4d8f9c10965b446432cd : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb46c93af8a348d788547427dc4da8bf : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2f0483ba088401d83c61334fd96b276 : SUCCESS in 29m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c78a0bed66204abcafec46fda7208903 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad091f52b85943ecb076a1b893af1aa9 : SUCCESS in 44m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b5acec0f48e845018cfa60012ff8daff : SUCCESS in 1h 11m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/423e9013417d4564ac6d117f1e129368 : SUCCESS in 20m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fa25e3ac53c4bb49eba9ad162068664 : SUCCESS in 14m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b86532fc68c44894a5a300d472577520 : SUCCESS in 1h 44m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14c1ef2a6ae5482e8a2a167ede186873 : SUCCESS in 1h 09m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e3fbd8862fb4e1489125f019fcb2232 : SUCCESS in 47m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68aa45105e224309a62f94bc219583f8 : SUCCESS in 37m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8edd15bc9226485f95db2144c6569994 : SUCCESS in 23m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6089e30926c945418914adf66d7e1cf2 : SUCCESS in 17m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b3710bb20007451fbf4c8c300d36a2ad : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3241484418404abdba808bfe2604a877 : SUCCESS in 9m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c16aa7eebcc14070afb2d918de3ef80d : SUCCESS in 16m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e96d9122c1b4c59b9237badc4bbb32c : SUCCESS in 9m 58s","accounts_in_message":[],"_revision_number":2},{"id":"995d51990e4f1c1c46482bfd45e69a1daef47b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-12-23 13:23:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48780ea3c96dcb47a65178ce9a0bcd488c4423c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:21:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6378f97b8e714e78b9edba42e6826bf0 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57d0900d492f4112b3ce89fe6b317e90 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/287f6fdaf2aa47c3bb7fb49e6bbeaab4 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd09cb923ed948298e1a462db4237343 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a70ff1cf1254ceb90a8fbc5c01a2f65 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c32e44e08c04adc976f8dad6c04c406 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62a3e35c007841ef82d7e5f303ce362a : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35e9e49496164c8e9fdb82ea88c226d8 : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb0189e8034e47abb59a55a404cd7c8d : SUCCESS in 30m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e0f6d5c7cbd94c06a954b1bc5539e8da : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cbb2e8e38ff48e6993781d51ed8f2c0 : SUCCESS in 49m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/723967fe820a46f691b617caf05bdc9f : SUCCESS in 1h 35m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/486da7f6c06140cd84efc98096dcafc2 : SUCCESS in 27m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0dae627302a433e9cdde05c32542d3a : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2feb1b6e91c2406ea7a98a009cb91623 : SUCCESS in 1h 46m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/666e00f8ed8448998680a6a3c1e18d65 : SUCCESS in 1h 17m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/88beaf2d1091418d8b3f88cf0ec90ae4 : SUCCESS in 47m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/351451eeaee445fd932154859cc70395 : SUCCESS in 39m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33edf8a32a9d415f99be4d1eb404e5f3 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b206ef67c5954ae4a39e6fd8933bf44c : SUCCESS in 22m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/39c27ecef4d04eb69cccc6aab2c4515b : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860505f7ffad44b7aefd7a94bb9abff2 : SUCCESS in 11m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/60de2ed381314e9c939b3fd0d3cfb555 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39359fa8cca0422caa30dfe61135308e : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":3},{"id":"7644fc4b24a83db922575b699380051f155a3e8c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-28 21:49:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"22415722f5f6cf10a697e1c884b065c3d16e3aa2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 13:52:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5f861adb35ebba7689594141eb5c427887cb9d47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 13:58:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6d623fac1a2817e7ce8537cc2a95ecd6de8304ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:45:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbe675138574310bd194b132c013732 : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7349acd1687a4e4d8e3990ae87c18168 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/810216c735464e79bbb254b2eadf1b9d : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242c2fda61a14e0fad76d217014a837a : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b5ac7f21964aaf98fdbf9eebb8ff88 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1bb34ab55444b08bb1121b658dbc407 : SUCCESS in 8m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a714a418fd043d587e67783414400e9 : SUCCESS in 14m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5742d47876fe4137bf3c774216f1105a : SUCCESS in 37m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/971cbd3e3b404e1cb2baf769c547b7d7 : SUCCESS in 1h 12m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d9f5447fde4ec8b2c417296760dd19 : SUCCESS in 51m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/954a68e66d594ea8a4fbf57813e4cbd3 : SUCCESS in 1h 15m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/61551f902a5b4e1397e172f22b6ab640 : SUCCESS in 26m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f2c65ac67444ed8857e27c45dd7a7fd : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7bbdb027dd4943aa85691d4fb1bfa5b6 : SUCCESS in 1h 57m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/194b7d7a8c9a4949b40e53aeab19786b : SUCCESS in 1h 20m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/055af9c6d0c74566a2ad182fa00c8307 : SUCCESS in 52m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb664e6595fd43819a30890eb8365e02 : SUCCESS in 45m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6cd9dba30ebc4972882bf4ed4dd6173e : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd4c52fcce854f6e915923ea21d41b23 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4da0716f685e4fef967ff0110baa5460 : SUCCESS in 20m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b047a73fcaa4246b0d35d541ac7c49e : SUCCESS in 15m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4a1aa2b271e49978de7941c5afb05e0 : SUCCESS in 22m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08054191acdc46eab16d8cafef714431 : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":3},{"id":"533b469b9ff4a6a0e8f23270b85e3ea863bd6d67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:45:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"739d91502a1b07623b798d4cee41f9c3d71ccd0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:11:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8726271306f641c0ac0c3cc5efc5d0eb : SUCCESS in 2m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af50ca19a4f44038bd4920c63625d87a : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0dec91689aaa867d407f02264dced5677d84f84","revisions":{"a44ea84ae8986a7d46e406a935ac03675a285fcf":{"kind":"REWORK","_number":1,"created":"2021-12-22 18:08:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/822758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/822758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/822758/1"}}},"commit":{"parents":[{"commit":"719968f694ab65249cdfc67a5c216cddb1d2a70f","subject":"Merge \"docs: Improve all-in-one scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/719968f694ab65249cdfc67a5c216cddb1d2a70f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-22 18:08:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-22 18:08:42.000000000","tz":0},"subject":"WIP: disable EPEL by default","message":"WIP: disable EPEL by default\n\nChange-Id: I96eb4685f997e85ad2ee5318640d58d0287a016d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a44ea84ae8986a7d46e406a935ac03675a285fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a44ea84ae8986a7d46e406a935ac03675a285fcf"}]},"branch":"refs/heads/master"},"6f1f766003dd7c402b8a3001010ae686464944f1":{"kind":"REWORK","_number":2,"created":"2021-12-23 10:48:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/822758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/822758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/822758/2"}}},"commit":{"parents":[{"commit":"719968f694ab65249cdfc67a5c216cddb1d2a70f","subject":"Merge \"docs: Improve all-in-one scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/719968f694ab65249cdfc67a5c216cddb1d2a70f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-22 18:08:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-23 10:48:27.000000000","tz":0},"subject":"Disable EPEL by default","message":"Disable EPEL by default\n\nEPEL is no longer required for a default installation. Let\u0027s disable it.\n\nAlso clean up the install_epel variable from Kolla Ansible globals.yml\ntemplate, since it never existed.\n\nStory: 2009757\nTask: 44227\n\nChange-Id: I96eb4685f997e85ad2ee5318640d58d0287a016d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f1f766003dd7c402b8a3001010ae686464944f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f1f766003dd7c402b8a3001010ae686464944f1"}]},"branch":"refs/heads/master"},"e0dec91689aaa867d407f02264dced5677d84f84":{"kind":"REWORK","_number":3,"created":"2021-12-23 13:23:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/822758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/822758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/822758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/822758/3"}}},"commit":{"parents":[{"commit":"719968f694ab65249cdfc67a5c216cddb1d2a70f","subject":"Merge \"docs: Improve all-in-one scenario\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/719968f694ab65249cdfc67a5c216cddb1d2a70f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-22 18:08:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-12-23 13:04:12.000000000","tz":0},"subject":"Disable EPEL by default","message":"Disable EPEL by default\n\nEPEL is no longer required for a default installation. Let\u0027s disable it.\n\nAlso clean up the install_epel variable from Kolla Ansible globals.yml\ntemplate, since it never existed.\n\nStory: 2009757\nTask: 44227\n\nChange-Id: I96eb4685f997e85ad2ee5318640d58d0287a016d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0dec91689aaa867d407f02264dced5677d84f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0dec91689aaa867d407f02264dced5677d84f84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
