)]}'
{"id":"openstack%2Fkayobe~823266","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Iff8c9c5d3b13207c25ae58fee4053426b13d62fb","project":"openstack/kayobe","branch":"stable/victoria","topic":"ci-stop-testing-non-stream-centos","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-01-05 16:06:29.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Iff8c9c5d3b13207c25ae58fee4053426b13d62fb","subject":"[CI] Stop testing non-stream CentOS","status":"MERGED","created":"2022-01-03 11:01:12.000000000","updated":"2022-01-05 20:33:35.000000000","submitted":"2022-01-05 20:33:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823266-ci-stop-testing-non-stream-centos","meta_rev_id":"003e0a2de3ac73f1398441ea5cdf62db7bac838d","_number":823266,"virtual_id_number":823266,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 20:33:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-05 15:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-04 08:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-05 15:38:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 11:01:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 08:36:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-05 15:38:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"935cf1a4e2a8ab47becbdba81444c7dae49cf48b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-03 11:01:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"464cb90e4a68708fdee667831043f7bc93c1b8b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:01:51.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe-config-dev on branch stable/victoria.  The error\nwas:\n\n  Job kayobe-overcloud-tls-centos8 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - openstack-tox-pep8\n          - kayobe-overcloud-centos8\n          - kayobe-overcloud-tls-centos8\n          - kayobe-overcloud-host-configure-centos8\n          - kayobe-overcloud-upgrade-centos8\n          - kayobe-seed-centos8\n          - kayobe-seed-upgrade-centos8\n          - kayobe-seed-vm-centos8\n  \n      gate:\n        queue: kayobe\n        jobs:\n          - openstack-tox-pep8\n          - kayobe-overcloud-centos8\n          - kayobe-overcloud-tls-centos8\n          - kayobe-overcloud-host-configure-centos8\n          - kayobe-overcloud-upgrade-centos8\n          - kayobe-seed-centos8\n          - kayobe-seed-upgrade-centos8\n          - kayobe-seed-vm-centos8\n\n  in \"openstack/kayobe-config-dev/zuul.d/project.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9bd782974181119cd0a23eef56f56f77a3ea4cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-03 11:24:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8b0a9fde2198a7da61663df54ba6a28cd0b71681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 15:44:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62a3cd8dc964a939db4c31dc4cad5fa : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbfac849f2f44f0ca59ba16e626e5f62 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/770ca5945c6f46beb19fb5a015724711 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5956e19ed51042439edd00c78b1edf1b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7c0b8b98214d61bdda15e3d3e0833a : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d56a5ff391842f1ae337e5f7fffb868 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2c4f2fc911e4c2d81224fc21abb1dd5 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d475e5e086c94be79d2670a57ef6f92c : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79531cff28264ff396a987b3f2883b96 : SUCCESS in 59m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/296d617bc8fa48f38753f73c670e47bf : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a2a2a69bc313419eb9cb04ac6724b824 : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2e96803f5ec4446db0b998f17f744545 : TIMED_OUT in 3h 01m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e2db0d030a914ef7adf12e032c317a24 : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ab4e8e3f42a430796e4e966822c0888 : SUCCESS in 25m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ce30a86f3f5a45f08733c911a69ad15c : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0512fe0dd8da518677dbc529a5f260400be267c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-03 16:36:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"944c43837467f7730b583cee9d0f9df82b7d47f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:57:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64df6867875b4f80be7a905063e2a524 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1980fac1b5fb41c4b48b2d5f4351cd14 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2350a49f44ff47efa4e3627efb972c16 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/808187a6d7e84513b06b1233d2046654 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac3883b7dac4895ac1804d8c621c518 : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/085fa92bed624f63917c8b17a33dcee7 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b0359748a654fe283df03a74420a148 : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5aa101358cf942599bfcdf7696c2d84b : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6655e39175404603837efad59f301b17 : SUCCESS in 58m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/38b5dce9fa2d41c2891cf260481f7c33 : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/84966fbcd3864737867f139d8f3b4fd0 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4af1d67124d14c6f987697f1f968b235 : SUCCESS in 2h 18m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7ae97e8f1caa4b26b3f3cdececfb1f90 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd793e399b6b44bea5a1c8b24e9e4364 : SUCCESS in 27m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6816b6c1b8f047d1b67edefd7af52ad1 : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":2},{"id":"dcacdfabc98dd1eed4fe4205ebd08555eeb0eb1e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-04 08:36:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4f0ea1c50715302a2355c18e8b10607d6029ec74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-05 15:38:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e511082207ded73b68cf6ee601e9485bd6d0654b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-05 16:06:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94c33292c983d23caf2700feb42db1814a4fb630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 16:47:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e03f006f34b0bf3f8d3dbb204f93bd89ff6a2220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:33:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56be434b1b2492f810008e48477e833 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424bd175b1bc45b39532fbb6e515133b : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee8a1be8bca4117b76843cfd3e710ed : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf23140acd3436ba42be2e40a28aac8 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2affa941e1dd4d77baf739a41313487a : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a1e98ada0e8462dba716c7a8d5b4b2d : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7d2672d88bd4bc5aaba18378614016b : SUCCESS in 16m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/11ec6f4fe4fb4d3fb73e1cc1137280ad : SUCCESS in 58m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d6fcca7fd43e40d589b773c08106f850 : SUCCESS in 57m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f73643b57ae8405b97ee4cba7cec5a1b : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2563d17ca2384ae7a17f94ae3d19e23d : SUCCESS in 2h 03m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e589c330bf5a4d819591f8d786ae94fd : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d4f0a7ca7c7e461eb6c80bd639891f81 : SUCCESS in 38m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/073781b4afbf476ca802ec38e26b8de5 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e4001956faf93dbb8270824e5c625550a8c567bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:33:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"003e0a2de3ac73f1398441ea5cdf62db7bac838d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:33:35.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe-config-dev on branch stable/victoria.  The error\nwas:\n\n  Job kayobe-overcloud-tls-centos8 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - openstack-tox-pep8\n          - kayobe-overcloud-centos8\n          - kayobe-overcloud-tls-centos8\n          - kayobe-overcloud-host-configure-centos8\n          - kayobe-overcloud-upgrade-centos8\n          - kayobe-seed-centos8\n          - kayobe-seed-upgrade-centos8\n          - kayobe-seed-vm-centos8\n  \n      gate:\n        queue: kayobe\n        jobs:\n          - openstack-tox-pep8\n          - kayobe-overcloud-centos8\n          - kayobe-overcloud-tls-centos8\n          - kayobe-overcloud-host-configure-centos8\n          - kayobe-overcloud-upgrade-centos8\n          - kayobe-seed-centos8\n          - kayobe-seed-upgrade-centos8\n          - kayobe-seed-vm-centos8\n\n  in \"openstack/kayobe-config-dev/zuul.d/project.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cd50e8c2778bfe83a207b10339eaf02d7a7af12","revisions":{"5c466c1601ebbe5abb362c349240b3fc979b9a5d":{"kind":"REWORK","_number":1,"created":"2022-01-03 11:01:12.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/66/823266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/823266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/823266/1"}}},"commit":{"parents":[{"commit":"295329e69d3a94d971e9d6dc1ace4c8555b4ca03","subject":"Merge \"Limit ip-routing and snat to seed hosts only\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/295329e69d3a94d971e9d6dc1ace4c8555b4ca03"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 10:53:14.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 10:53:14.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nChange-Id: Iff8c9c5d3b13207c25ae58fee4053426b13d62fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c466c1601ebbe5abb362c349240b3fc979b9a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c466c1601ebbe5abb362c349240b3fc979b9a5d"}]},"branch":"refs/heads/stable/victoria"},"0cd50e8c2778bfe83a207b10339eaf02d7a7af12":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-03 11:24:16.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/66/823266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/823266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/823266/2"}}},"commit":{"parents":[{"commit":"295329e69d3a94d971e9d6dc1ace4c8555b4ca03","subject":"Merge \"Limit ip-routing and snat to seed hosts only\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/295329e69d3a94d971e9d6dc1ace4c8555b4ca03"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 10:53:14.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-01-03 11:24:15.000000000","tz":0},"subject":"[CI] Stop testing non-stream CentOS","message":"[CI] Stop testing non-stream CentOS\n\nPer [1].\nIt was agreed on during a Kolla IRC meeting. [2]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-December/026408.html\n[2] https://meetings.opendev.org/meetings/kolla/2021/kolla.2021-12-22-15.01.log.html#l-53\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/823268\nChange-Id: Iff8c9c5d3b13207c25ae58fee4053426b13d62fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd50e8c2778bfe83a207b10339eaf02d7a7af12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd50e8c2778bfe83a207b10339eaf02d7a7af12"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
