)]}'
{"id":"openstack%2Fkayobe~823306","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I7b719aeb20cc63f41280e20063785d77b0334d64","project":"openstack/kayobe","branch":"stable/wallaby","topic":"grafana-monasca-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-01-05 11:51:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b719aeb20cc63f41280e20063785d77b0334d64","subject":"Deploy Grafana when Monasca is enabled","status":"MERGED","created":"2022-01-04 09:09:58.000000000","updated":"2022-01-05 11:53:01.000000000","submitted":"2022-01-05 11:51:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823306-grafana-monasca-stable/wallaby","cherry_pick_of_change":819799,"cherry_pick_of_patch_set":2,"meta_rev_id":"ff384e887c38170ae2bcc73625af1960862dac44","_number":823306,"virtual_id_number":823306,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 11:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-05 10:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-04 11:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2022-01-04 14:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-01-05 10:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 11:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 14:28:36.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-01-05 10:03:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5fbaa80c7f6e06e93ae724f38da5ad639592ae07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-04 09:09:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b82fc03ceb34f1b9c0ee1a3594346930e3fc8d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 11:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac0926b9a9944c5581fd49db92135761 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75c2c79196144f482de5a54b8bddea6 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/158e09a2027d44c3bd41606994731c3e : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221252d7ce704b72b2034a26b7548bce : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252d08d1c21c4ed1b1ed5088475f48c8 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cd8eb45bea431e863751d59a8d6300 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82362ea325f740399c85fe55224248d8 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a39932b5bfba438b92467f43c8b463b4 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfc2742f01d043d4bc83771ecd5ef2ba : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9845ae59ce8c467ab6967cb1d19ed419 : SUCCESS in 20m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/96b3860e51dd44ddb1dd59495e5181be : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1659dd295a66471a976f1327af81ec59 : SUCCESS in 47m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cfa292c218a1406b9cbe6ae2ebf9f90f : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bb259fc3bc594ef5abc4c234cc76dbec : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d74d92c46cd4339ada06cb544e0d839 : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e51fcd9375114a62872341f7291d1ad4 : SUCCESS in 1h 53m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb0f16dfe450449fb466cbeeb07e0a9d : SUCCESS in 36m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724f66a2fcfc401c91b3c51170b9db2a : SUCCESS in 35m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2d7baaa49a41415f84c477b87e7cb0d2 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5b8e759f45f047539dd51649fca74b09 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1fe076f6dee424f876266d37010621c : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"908e9d55979f0849c2504bbe19299fba3e71e8d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-04 11:25:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a5d32095ece4d2a24911946ceecc5c7247b84b0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-01-04 14:28:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d02d7913d3591b40b1370288e40efc31eec17d9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-05 10:03:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"497acb12e1cd3dc34141477c6b3eea792e0ff533","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 10:03:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a0506dcd0beddb5f73ea1dbe6a67c374eb27487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 11:51:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f4899ffbef40e9887bb776e284dec3 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bb81aadee6844808bd92fe3dbe8d661 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/377fa2666cd24b5f8a549c9ddac502e2 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7712564bc5024e499ea81ed7b4616efb : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d49ce3b7c3443f5a48165b1d99ad67f : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b659801e561541fab6bb05be188d5348 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0955d1bd505a4ebeab398693952dd06b : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f33f6c0e4c74396b9d2342d2f57cf4f : SUCCESS in 21m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c9800ae8dcb24841ac7e89ddfbfae151 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a8f2ab05c0b4da293a1b980fab28577 : SUCCESS in 50m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d02e9555ab0149b9819c52e73135ad2e : SUCCESS in 1h 45m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/45e7498e3b2f4b428b8cf8ec1e1618aa : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73c84fe519cd45b1b27b9bfccf0d7e73 : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e09c02b5e0b54d0ba6483559fdf0ca11 : SUCCESS in 1h 45m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0cf97dc2e34a4bacbef74daf942c2615 : SUCCESS in 37m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cfb2084a239459a9c0c854cb2abc4ee : SUCCESS in 31m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec9347d9de604acd93507d25a5118f43 : SUCCESS in 29m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca0e5fd832af49aa9974aab7c5fd10f4 : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44b6e0020dfd45a2a822a292b35cdea8 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f477c23e52e8fd95a766ad0d805a84e2ca90b17a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 11:51:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff384e887c38170ae2bcc73625af1960862dac44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 11:53:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a223273b33fe400188f8b11bb4939531 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa83e7e0f0a442bf8a37c495bed55ff7 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c68913d87dcabb7ceac0c99c10c7dbf9ae9153c3","revisions":{"c68913d87dcabb7ceac0c99c10c7dbf9ae9153c3":{"kind":"REWORK","_number":1,"created":"2022-01-04 09:09:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/823306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/823306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/823306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/823306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/823306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/823306/1"}}},"commit":{"parents":[{"commit":"3634c0eb7ab63c8935ce6b5b502fcd7742274df1","subject":"Merge \"Limit ip-routing and snat to seed hosts only\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3634c0eb7ab63c8935ce6b5b502fcd7742274df1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-11-30 10:21:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-01-04 09:09:58.000000000","tz":0},"subject":"Deploy Grafana when Monasca is enabled","message":"Deploy Grafana when Monasca is enabled\n\nThis is necessary to deploy grafana as a replacement for the retired\nmonasca-grafana image.\n\nChange-Id: I7b719aeb20cc63f41280e20063785d77b0334d64\nStory: 2009717\nTask: 44091\n(cherry picked from commit 8ab4530f2620a35bb6049640b45de3a70176269d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c68913d87dcabb7ceac0c99c10c7dbf9ae9153c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c68913d87dcabb7ceac0c99c10c7dbf9ae9153c3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
