)]}'
{"id":"openstack%2Fkayobe~823459","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ide13ad5d39388a3d484bb0285adf49e978239515","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-05 15:37:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide13ad5d39388a3d484bb0285adf49e978239515","subject":"Build neutron-sriov-agent image when enabled","status":"MERGED","created":"2022-01-05 11:46:44.000000000","updated":"2022-01-05 15:39:41.000000000","submitted":"2022-01-05 15:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823459","cherry_pick_of_change":812667,"cherry_pick_of_patch_set":1,"meta_rev_id":"29fdd8e956095c2bf60b055f00712101ada2c402","_number":823459,"virtual_id_number":823459,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 15:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-05 12:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-05 12:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-05 12:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 12:04:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-05 12:56:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-05 13:40:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0df2021224b65abb3a20970f271e7325d7fb3cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-05 11:46:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"063e906fe8984c1e3873f1d1f66b78f6870bc564","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-05 12:04:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57f461f5740a71411518cc1f87a4eef3b0d37df2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-05 12:56:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66cf85d1bc7378d108f49cc1074e2677294d2017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 13:40:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d6a75a499774da8bf4c386539deb313 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed362c5c56b2444dbf60c685b0f10c70 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e383780d6c414f5e8a95201ce16b88c7 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/902b5955eca94f0b9c2338f0270659d0 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/667aaf0bf3bc4d1da5ff0889c5f06de9 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd690015310e46c5b1a0c09774d1413a : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60ba0a6b347845cf891e0e009da44b14 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1768e458d4a747b9a5a17602195bfae1 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5825eea25cb646269f3314e47f8d6961 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b707189609d44a391401f9e82b91144 : SUCCESS in 31m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1b5ffee8784446fb8766b6fa92494dd9 : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3ac3a1f402b4a65afe82b48a01e1cce : SUCCESS in 45m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/070edd43f6634f88b024e0de6de032de : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/341e06fed47e4fac8a40f0b5e50b3f91 : SUCCESS in 24m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd57e411997844db998299a289a9d852 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56151d46fc0b495f8401b466d96edb50 : SUCCESS in 1h 51m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01f316b4382e43c59badcdfd445773e3 : SUCCESS in 1h 23m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b9475eb5c4654ad6a70a1a1e9373b711 : SUCCESS in 40m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e72173e77a2d446fb50197b8753e2b90 : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/45a313d68e01478ba1ccd6ccccb98089 : SUCCESS in 33m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/280e44ee5ed34701b32cf9758d7da022 : SUCCESS in 29m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5e2ec16c26074e439cdcd235428a8de0 : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7ed06e58f1042e184b28403c7e47954 : SUCCESS in 12m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab57555e4e444f8c882e84922a41400a : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a9ebf8689ed400e9033fdfe3db9006e : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5b874c1c0959bf4fde45c6863597224fa8e2aa99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 13:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffe0696b561e6991e1ea7a04bba6b01e3f0d3d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:37:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71b629251064bae8d1820788aa96056 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7019b744a3644d42b1f6a746a6dd0fa2 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5707427bc4475fa5770f49fc484962 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f51631f6a1d4a888cc32fb2dbf44484 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939db00912e34d10bdd7af7582f086b4 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e51ad87149d4964b799f9c184ca5278 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa274bec00be42ce8a8d0d0c6217f5be : SUCCESS in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0bf0eff40bd4171b3aa9e78e83c3e05 : SUCCESS in 30m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c26832f505a94e94832c8fbb41c96d83 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/435e9c64194c4079afd89911c4ffb9fe : SUCCESS in 46m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0cf675695b464fb097ecf55e60ca6312 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5c5373c6e91046e58e2443f5592f7dfc : SUCCESS in 26m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eb61ce20c23458aac0d97620166d8ee : SUCCESS in 23m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8b2909a6e3e545408368e8b3b3860164 : SUCCESS in 1h 51m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0711b72f1a5848d8865373c75640422a : SUCCESS in 1h 22m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18b1d4f9576d445793d617082832a137 : SUCCESS in 39m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3fb6c4b4bf94a249248b52cc1e1d2a5 : SUCCESS in 31m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0dee0782ec194f0088d58a53270ae654 : SUCCESS in 34m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ce30f5ab104b9ab760cd59007bb1c0 : SUCCESS in 25m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/112d9d4cb6a54f7885c3d2acf238e1f8 : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f1ee76a43e04e0d9bbb8555084a1cc5 : SUCCESS in 12m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/23b97eca50624f5a8ecfe81f1337953e : SUCCESS in 22m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8442e2fd01f54b38b13cc8f74d94de2b : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bb37657244463b4fa79f22d520a2e6c4bc2e1d82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:37:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"29fdd8e956095c2bf60b055f00712101ada2c402","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 15:39:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c9eb4d69bc433992d62357a8f32e20 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca490574cae44be79c96b6bd3b6a793f : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b89b7a7366998d5bbb91186ccabbaf67b1b6ff62","revisions":{"b89b7a7366998d5bbb91186ccabbaf67b1b6ff62":{"kind":"REWORK","_number":1,"created":"2022-01-05 11:46:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/823459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/823459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/823459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/823459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/823459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/823459/1"}}},"commit":{"parents":[{"commit":"a2f834182bf2d3a99f2b7ef83c552e0b68b9a9e8","subject":"Merge \"Limit ip-routing and snat to seed hosts only\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f834182bf2d3a99f2b7ef83c552e0b68b9a9e8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-06 11:40:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-05 11:46:44.000000000","tz":0},"subject":"Build neutron-sriov-agent image when enabled","message":"Build neutron-sriov-agent image when enabled\n\nSet kolla_enable_neutron_sriov, and the neutron-sriov-agent image does\nnot get built. This change fixes the issue.\n\nStory: 2009275\nTask: 43530\nChange-Id: Ide13ad5d39388a3d484bb0285adf49e978239515\n(cherry picked from commit a669ccfa26c430ef634efa1f2e85a1b68b534ccd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b89b7a7366998d5bbb91186ccabbaf67b1b6ff62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b89b7a7366998d5bbb91186ccabbaf67b1b6ff62"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
