)]}'
{"id":"openstack%2Fkayobe~823461","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ide13ad5d39388a3d484bb0285adf49e978239515","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-05 20:37:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide13ad5d39388a3d484bb0285adf49e978239515","subject":"Build neutron-sriov-agent image when enabled","status":"MERGED","created":"2022-01-05 12:00:59.000000000","updated":"2022-01-05 20:39:54.000000000","submitted":"2022-01-05 20:37:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823461","cherry_pick_of_change":812667,"cherry_pick_of_patch_set":1,"meta_rev_id":"9b56ed774e10ed2b6357b7183af81c1adf4dee41","_number":823461,"virtual_id_number":823461,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 20:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-05 12:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-05 12:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-05 12:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 12:04:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-05 12:57:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-05 17:03:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1193cf59a7a3e8ec21d92f9999f199343885977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-05 12:00:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e89b5f96354572b0221dfa7a695d4fe8af0d6b4a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-05 12:04:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24623093b64db252d24baff8eecdf19ed0417b39","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-05 12:57:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bae67bc3bf697cffdd69085104e78d17d0ded2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:03:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53a2cc27482491786e425b9a78e334f : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153211a72afd405d8468168a54acfe7d : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fccba9bbc604da1ad663b630bcd6ba3 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccf43e970c2f4b77bd6da547a88b15b2 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796c3440b10742878f65f20e31dac132 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aef1a6b92dd4dbf98941bf89cae5374 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b84ed1b7f9040e596bc13dfede72f3c : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59955117455c4dcba5fb93fd95f384c3 : SUCCESS in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f24078c12521413b85a01add76bdf0f7 : SUCCESS in 16m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b33a1960a7d140bdad8c25b9f9d269d5 : SUCCESS in 57m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/389816f9eddd4b56a75fabdce403c1c5 : SUCCESS in 58m 38s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7d46e1f45a1040b79c478bc1b9728b8d : SUCCESS in 58m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c8b43528018444e8ac673af35d79d589 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d64ba2798bbe4f04b428d4d759315574 : SUCCESS in 14m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bab3210f4ed74ee8ad0d6fc2aadd6591 : SUCCESS in 24m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/61200f86900c440f85871873f4920415 : SUCCESS in 2h 11m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/80f38d6b44354ca9a3a9efc1b7f7aae0 : SUCCESS in 2h 21m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/abaf66ad58c64031ba7f41b1eb00924a : SUCCESS in 29m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18d22eb834ff44eb9cf5cbcc3b6ea1a2 : SUCCESS in 34m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fdc4bb4dab7c447a940219f6f5a4d5c7 : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d650804354244bfba9d77b6f5398d23b : SUCCESS in 31m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/831a764255f24b50a0a9457dde0845bc : SUCCESS in 10m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf6c928f11894ba29252ed88ac458a5f : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2aefcfd5e1e615c89729f646fbac75269f38f343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 17:04:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a357d47c47f70bb648480b661a072ade7f5f3bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:37:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e42f628f704afaa07bc97053a6085d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8c1e42e89f4785a0dd4ee7c008e825 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d12d76aa45c422585f15aa23effdc64 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f39d0e1e02b42558b138e72d6a13ebf : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2a42b90955b46d891dc325edb98ea22 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3eecf4ef58a54cd9b5f05ca09e98df66 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5139fad1cda3488ebf162f2daad6bfe5 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bb01beae10f4ca696b6f4f7dc10c0d3 : SUCCESS in 16m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0580124032764ba2866755fdfa103248 : SUCCESS in 56m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5d7051cb9a264cefb5af1676384356b5 : SUCCESS in 59m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3a9f2b3f907f431fb5fb7f5f1e049972 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5de258c1f7eb45b5b759c799953dd44e : SUCCESS in 2h 12m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4f9411e9b20c4c7d9a1851dc1a1ca154 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec08cd6e4f514b4d9c59f9e25ea6f167 : SUCCESS in 31m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/519b0ee4ec16456f8e9abe99dad4b0bc : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"48e6b8a0b9da6e812ee17a6a04b4a97ff3778043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:37:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b56ed774e10ed2b6357b7183af81c1adf4dee41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 20:39:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a83b705380440cf8418db11711559d0 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/441e39f6b9074c849866bb939ea96819 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07fd6259debf39763ddb39d53bd785dac3da93a4","revisions":{"07fd6259debf39763ddb39d53bd785dac3da93a4":{"kind":"REWORK","_number":1,"created":"2022-01-05 12:00:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/823461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/823461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/823461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/823461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/823461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/823461/1"}}},"commit":{"parents":[{"commit":"a0698ae6c57fc511d6f90d2455eca4734b3198b9","subject":"Merge \"CI: Use correct TD agent repository version\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0698ae6c57fc511d6f90d2455eca4734b3198b9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-10-06 11:40:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-05 12:00:59.000000000","tz":0},"subject":"Build neutron-sriov-agent image when enabled","message":"Build neutron-sriov-agent image when enabled\n\nSet kolla_enable_neutron_sriov, and the neutron-sriov-agent image does\nnot get built. This change fixes the issue.\n\nStory: 2009275\nTask: 43530\nChange-Id: Ide13ad5d39388a3d484bb0285adf49e978239515\n(cherry picked from commit a669ccfa26c430ef634efa1f2e85a1b68b534ccd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07fd6259debf39763ddb39d53bd785dac3da93a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07fd6259debf39763ddb39d53bd785dac3da93a4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
