)]}'
{"id":"openstack%2Fkayobe~823763","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ief9adfcff28d269fc80a073cdde536e8e803ca37","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-01-10 12:23:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief9adfcff28d269fc80a073cdde536e8e803ca37","subject":"Generate kolla config when running database commands","status":"MERGED","created":"2022-01-07 13:34:48.000000000","updated":"2022-01-10 12:25:56.000000000","submitted":"2022-01-10 12:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823763","cherry_pick_of_change":815895,"cherry_pick_of_patch_set":7,"meta_rev_id":"896524a751f89fe34b204c6ad370d1af4d444177","_number":823763,"virtual_id_number":823763,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 12:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-10 09:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-10 10:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-10 10:17:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 13:34:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-01-07 15:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 10:17:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"8564b7450ea28bf189c9f6e68cdd7636c0fc300a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 13:34:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a5b357dc17202c1de5e70d11a2c9540b29547b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb50e249e82d4e798e311a7e27906f6a : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8aaa408a6545e7a790eb543b388487 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e43778a347054e4ab26662a42f11c5aa : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ab27315cae64bc58e0c895f8c3e6bad : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80c881a236d24159953221937d704496 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf65ac2b80ab4bdd924fade7dbca06fa : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a462b782e2944e458a74f35b57b9bf01 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7671d76072e1447090696053b8cd88f1 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3bac4d98159f4a8fb0603e28bf4ddabf : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b0a27aea645464d9f0344cbb77da9a6 : SUCCESS in 17m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/253c4f11879f418d8674b1bfccd8e8cd : SUCCESS in 1h 15m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/481d74b6792c4cd2996c19882eb91293 : SUCCESS in 47m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fc2a49d9d42944298c740164b7bc6e6f : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4c68ed8838d849e3ad1eab05d6e2490c : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8575d255019f4bf5be9bc66c9aee5aad : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3038dfc7c2d04caa976fe46048825ffa : SUCCESS in 1h 45m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/edebf73e53094d9e97d04fd899d883be : SUCCESS in 35m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/991d8f284b014af28ce0d09911d980b0 : SUCCESS in 29m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3a04e26ab6004d18828f25ac8029af41 : SUCCESS in 29m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/94a5ef4d6fdb4e268a1ec43bbde76d5f : SUCCESS in 16m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf4289ded98415c8d26b2e730391f6a : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a323c6ab40f3a7921f4a9f81842a05fa420fc10f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:21:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d53c86aa8db95623a4ccc90481a65c1c5fa3b2cd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-10 10:17:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d1a1b093fc4080f7788ef0a632cb378c9d3406e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:17:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15d89f63dd1573306522d9a59917600441114470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c34c528fa44e518eab0e0cf54d0e90 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfc83789561a4941b81a4415203beaad : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a40dfc8681d41fba443c82f32c4e506 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16956b4ae4bd43aa806c36241bb07add : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/403935fa023e4a858311625a2d4b9c26 : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6efcf81f56de49c5b7e71968bb481d67 : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e72693db7e0c4031ad54bcaf1aaabad0 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aadce958e56e42dc8a3d8369fead069b : SUCCESS in 20m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/717b9861720c4c9abbbdee92be2f3b4c : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb5d73ebdadb4f83a12a4b81df68fa9c : SUCCESS in 45m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f704e91c344d458293bd088d73e17c1f : SUCCESS in 1h 15m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3ff887389ba478987082e6419ffed6e : SUCCESS in 22m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0314302858344d7981c7d64db3a2ff98 : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6115e6d5c3c544c3a6d510a8414d2c49 : SUCCESS in 1h 49m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/02ffa5bc7d7744fe982a82b900f25cbb : SUCCESS in 59m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41fd68dbf2f54c5792916ca919185384 : SUCCESS in 33m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3513942bad874a65a742d9b7c10144b1 : SUCCESS in 29m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/27b17359b4b54fd3a0dafc280d053eb8 : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b3ce1c62424e7b9ab24e9f3e69d53a : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"730ad15e75eb8bb783b458a12ba8c17331a2e1d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:23:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"896524a751f89fe34b204c6ad370d1af4d444177","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:25:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072205f041094ce7b311de007e272c18 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d653f66548b74e458e7e6d1f702260b1 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9dd7e0dbb7f4b5e2247649ff2bd86fa838ea5fb","revisions":{"e9dd7e0dbb7f4b5e2247649ff2bd86fa838ea5fb":{"kind":"REWORK","_number":1,"created":"2022-01-07 13:34:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/63/823763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/823763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/823763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/823763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/823763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/823763/1"}}},"commit":{"parents":[{"commit":"32db78c932a5769d90ea7fca80de6aba67d44918","subject":"Merge \"Document that extra kernel parameters are important for inspection\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32db78c932a5769d90ea7fca80de6aba67d44918"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 13:34:48.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n(cherry picked from commit 4321471a13eb949c2f5d57784557ddebb9b682f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9dd7e0dbb7f4b5e2247649ff2bd86fa838ea5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9dd7e0dbb7f4b5e2247649ff2bd86fa838ea5fb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
