)]}'
{"id":"openstack%2Fkayobe~823764","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I6946b578f23843dd3cbafcb185b8d8cfbb00c080","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-10 09:19:29.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I6946b578f23843dd3cbafcb185b8d8cfbb00c080","subject":"Fix seed VM configdrive when admin network is a VLAN","status":"MERGED","created":"2022-01-07 13:35:02.000000000","updated":"2022-01-10 12:36:20.000000000","submitted":"2022-01-10 12:34:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823764","cherry_pick_of_change":752955,"cherry_pick_of_patch_set":6,"meta_rev_id":"1470a601d57be9bda29ade20d185e44928f183b4","_number":823764,"virtual_id_number":823764,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 12:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-10 09:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-10 10:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-10 10:17:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 14:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 10:17:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0fac3b6523faefa606e7cbafc4dd78c14d5809c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 13:35:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* requirements.yml","accounts_in_message":[],"_revision_number":1},{"id":"266a702355ce1b63dbcca69524cdbe916e31ae0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:44:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d96a4ff33e24e139ba35cc0f64d03fb : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a41fbd78f84c5f89595abe0a50bb6d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12479263abe245fbb4cb59f7b246eadc : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/542ee4cf7de54d43863a2bbbf141b794 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da6966580b6473dbb8bbe2cf087a377 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b100a91308f4e8cab4e6baf63978942 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4769a38cfc6d44e596118facc533c15b : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5a07885fd424112a8cbf17847afd6f4 : FAILURE in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6dea686e7e14b3a83d52b3efa24f277 : FAILURE in 6m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb36b7b8669f42d7a3edae362928c62b : FAILURE in 7m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/be1124ca4f214c9c8b646a20336d77f0 : FAILURE in 9m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8796663575c847b789ff54f9dbe5c695 : FAILURE in 8m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7499fe7561ce493ba6846423774588a6 : FAILURE in 8m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ce1c4725aa6e4785b039edc663f43b32 : FAILURE in 9m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a5cd554fa24e2da43ee3f8bd2749b2 : FAILURE in 8m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76ed9705adfa41d7807111072c66aa22 : FAILURE in 1h 05m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1b61882fbad44f779015154136b371b4 : FAILURE in 9m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d55ba9606fb474b98aab79603365eb7 : FAILURE in 8m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f41c8b2a306646acb96054e7b1789ee8 : FAILURE in 19m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/740bd086fb364fd9a567282a1cee03ac : FAILURE in 9m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d83ff730123401490aa215d2aba7c76 : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a64b2dabc49bac43654824b2fc08f111cc7b8cf5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:50:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"00ce4b70f72a641134fae596e968656d36c80488","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:51:08.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f025803d2adca4ba984f300cc6f88d920928f3e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 17:40:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/324ab0df8be64c1b9c577b998a4e5197 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eabc8ed013354e868d1ca57bb4f24790 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2c363bbce1411d9e1799b95dfcaa88 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a73aa82ca064fec881c8ad81b999ae3 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c862dddca70541afaccfba28776c7b7c : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1364ac284bfe4e06b13da73afec294bf : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a379a85a200e49418971684cab59874b : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca6e26309a3d4d508c0f55dc936e6fd4 : SUCCESS in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/007c4b6b47704d98b58d9367a8a76eef : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6899651d86544f1096e121cf5ba2a944 : SUCCESS in 21m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b4538e884f044745bcdda8c78868a861 : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b872453fe0c24546a668404a4d576716 : SUCCESS in 45m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e7ef7c47cf3741ed86075b2ec6aa8ac7 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f1a3906ccdc246e4ba8a380bd7bd2e63 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e68c4f3013f49c2bcc49392f67736e1 : SUCCESS in 17m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6725fd94d94547f7837366a2b9f82af7 : SUCCESS in 1h 47m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ffe08220c4dd40458025135f1ac0ea2e : SUCCESS in 35m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33869ff50da04ae79a1e2a63f15bc2c8 : SUCCESS in 32m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66d473d723b541428835bc3879eb47fc : SUCCESS in 26m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fd8ec3de6af4ab9bb49a63a286f0a0e : SUCCESS in 15m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f523fe5e13c40638789195af8cf98cd : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c57d240e4b5dd036facd251988f6229e0e714f0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:19:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"72b80bcc4ace88fe8bd2883cb45eff03cd0e5827","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-10 10:17:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4fd2b3852f1828c4aa608e8f6e55fb21d930b30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:18:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5db24860976d4e3b2c020fc7e2d7100ca753137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:34:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3758f74627a24adf9a4ab073dd33f9b7 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c380b0b01c446b9a4fb7b30fec3ab3a : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75ced949c2ea46e0aaf3df6f9fd89e1b : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3c8d724f4a4a90b4407f859477cd6f : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77dca827f3e84b26b59fd60fa36d68a5 : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/490439b5459143c1a59e8ced0d46051f : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea13a3272e024ceebb9917a8860783f2 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08211b5c82294d19a8cadbc716efe62e : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4c3739f60ec049efa02f13c574fac21a : SUCCESS in 57m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7f19a577ae64d9484ca5affb2dcd770 : SUCCESS in 49m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/137d01aa32ad4828844b7c88ff4fa6d1 : SUCCESS in 1h 20m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e327f41abbff4f05a0835662062b0cbe : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6730fd5b7c7549f3a7fa255d806dc7d4 : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/590f5de626af467a921efc53a220dea8 : SUCCESS in 2h 00m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18aa35f1165a47cfa921bfdb97b618e8 : SUCCESS in 36m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f6e133da884dfeabba8d987d954e0a : SUCCESS in 32m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a9c52e5153d54d4a89e64e72e6299aba : SUCCESS in 28m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff7bee16367f44f189ae36a4d2f5552e : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/632573ec13e7407597ee3582f17292bf : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4352ea389074882fd7b4af9845af23e7f001d3a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:34:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1470a601d57be9bda29ade20d185e44928f183b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d215d9120f49a4a5893f33954b0fe5 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f13e6c6cc2de44c8b43d0ebccbaac3ae : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c65af8edca3ac3bc169d497b11f2572da096141","revisions":{"e66fdbea9134959003f005515bb73341db1275db":{"kind":"REWORK","_number":1,"created":"2022-01-07 13:35:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/823764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/823764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/823764/1"}}},"commit":{"parents":[{"commit":"32db78c932a5769d90ea7fca80de6aba67d44918","subject":"Merge \"Document that extra kernel parameters are important for inspection\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32db78c932a5769d90ea7fca80de6aba67d44918"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 13:35:02.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n(cherry picked from commit bc0885f5397af9ee69f8c873e57ad6f3d7a360e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e66fdbea9134959003f005515bb73341db1275db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e66fdbea9134959003f005515bb73341db1275db"}]},"branch":"refs/heads/stable/wallaby"},"1c65af8edca3ac3bc169d497b11f2572da096141":{"kind":"REWORK","_number":2,"created":"2022-01-07 15:50:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/823764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/823764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/823764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/823764/2"}}},"commit":{"parents":[{"commit":"32db78c932a5769d90ea7fca80de6aba67d44918","subject":"Merge \"Document that extra kernel parameters are important for inspection\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32db78c932a5769d90ea7fca80de6aba67d44918"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 15:50:48.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n(cherry picked from commit bc0885f5397af9ee69f8c873e57ad6f3d7a360e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c65af8edca3ac3bc169d497b11f2572da096141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c65af8edca3ac3bc169d497b11f2572da096141"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
