)]}'
{"id":"openstack%2Fkayobe~823765","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ief9adfcff28d269fc80a073cdde536e8e803ca37","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-01-25 14:55:24.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-25 08:50:08.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"Ief9adfcff28d269fc80a073cdde536e8e803ca37","subject":"Generate kolla config when running database commands","status":"MERGED","created":"2022-01-07 13:35:16.000000000","updated":"2022-01-25 14:58:02.000000000","submitted":"2022-01-25 14:55:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823765","cherry_pick_of_change":815895,"cherry_pick_of_patch_set":7,"meta_rev_id":"49b7df8ea38b26738e2a95f9b508450a9b4ba4c8","_number":823765,"virtual_id_number":823765,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 14:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-10 09:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-10 10:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-10 10:17:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 13:35:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-01-07 18:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 10:17:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"308377e168bef6e4930e27684a66ff9e599ee5bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 13:35:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kayobe/tests/unit/cli/test_commands.py","accounts_in_message":[],"_revision_number":1},{"id":"0a4f01d2c2ac107f96d4332a348ca0c14ee015f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 13:37:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f5693bb7f831f385d3667420e85ccafaa6c264e0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:50:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a2b918fdb3376f16daa3bbe435241981e58e0fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 18:23:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2db29531aed645a3b859756adfa917d3 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f12e072c4d754ba791474d0703e6fd3e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceeeb145d4df494c9891543665ec8bcd : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a8677b6f8704c3fa30df4f07755a73e : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8aab043e8e4f9fb5dcc3c6c3b84f2e : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d63b3cf1dd3e49dd9849d109841c18a3 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d40a6b3b5a2c457c972adf01dcf7c85d : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b6db006d46c438abeeedd26420338f0 : SUCCESS in 7m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17960ea8f33441dd8e90a21af0411579 : SUCCESS in 17m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7914f2764c7b49d690b52372fba0e2b8 : SUCCESS in 57m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/66c6c5c0cb7549e0ab16f8290813f33c : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8fdb7f059af14bef9cbea85eee977d9e : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe6c24ebea8c4cc6883832fcb29794e1 : SUCCESS in 2h 11m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b91d60d081a149fb9460be56a4ef7d81 : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eaee1f230a5d4bfc8cd19a8fd5f5e6b2 : SUCCESS in 29m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/87803b4f3547490eb6a9eb6498eafdb0 : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"18d8623c605467324d08bc6563ca55d4fceb047d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:21:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e01785b5a2da46189c024a75353f75a5e3f1b78","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-10 10:17:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9f0320c2dba8f8ab6bace0ea80b618044521ee16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:18:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"de678544acbb7882aacdd2c9ccdc82a19fe1e9c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 12:43:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef06bf584fd4b80a115a43a726cc941 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea1de61bbb194641a6e3a4b1d41e66fb : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b05b0198f7043dfbb7a3ba1db32093a : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed8018c2740417eb3171a349c572a95 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6efec5c30fb48e39a18ad2e6f4dd425 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b92e5d8806f348298616bd970b0a1ed0 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85a74654836c436eaecffa4afb66b8c9 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dddbc605f6604a92977de5963fc61868 : FAILURE in 8m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4de124a0d5db44a098c418b575b5adec : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fd513406e10540e0863999e2d81a8d66 : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6602f883970f48eca201cb855df3998c : SUCCESS in 18m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44bc2044ba8c479ab7c83f264ccebefc : SUCCESS in 2h 21m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/36b93e539b624800a347167e1226f38e : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/35acca7468a54d21b3a6bacc7eaa5496 : SUCCESS in 28m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/df5b71361db246e284b34121e2c6351c : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4ae742f24aef9c05542f0eb97f17a80d3ec8d8ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-10 13:54:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85860aa23db202da57157c071b53851fa596ce56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:34:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90b5d40cfa0b480db130bd20157b53de : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a269dcce144f67a27e3b531c9ecf91 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da3b5a907706442a9943b292a0bd2110 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e1d1759d05c4abbbb81b12ae78a273c : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585f72c4c0514405a4ba18a0af5aacdb : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7aa5518be44cbd99101d3c7d288103 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f259031f9594441b90b6514f793eebb : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fcf43618d19e4a418aa0db6902203f60 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88cdf4c2189044f48e65634084786c0c : FAILURE in 6m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6490b0f00b38424b803dcb44a92d4b35 : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ca0a3c6a63694ed0a7b6b1dd9edd9108 : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8d4788d76ccb4c12b40f40eed3b68b6b : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/317faefbd2774a48b93ea8645d592924 : SUCCESS in 2h 13m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0fabb537de914d7a8a6612138ced6366 : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6eb0f9f6ccee4f9ca2bda4d4a1d3c018 : SUCCESS in 26m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/79ad693ce860405999028fa641c57cd9 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":2},{"id":"eed85e721e7e8e0c6ea37a17e6f01aba1735a6cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-11 15:58:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a78bdb721c2b977abc58952e7481e7153d87e538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 20:04:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ec19c71702456fbbec22324b15cdfe : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2218270780d0491391d1fea9f3098373 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae09ac1c39b400a93cb91079cb0c70a : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8180716987914a82b36f98927a02ff42 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba746c12439349a4825c08295f9841bc : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13735148057641e7ba27ca7fc52b48bd : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/636d77a6cb2b469882110a7d33515272 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7105be94597f47e8ac59bbc9d2203f6f : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff4c7029647e4aff98f63c04ad887a27 : FAILURE in 8m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/11c1c7fe71fe4dd1801b085439b492cc : SUCCESS in 57m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/73f90873ed90494a981fbb2282cf677d : SUCCESS in 59m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9c9e3b1bc4d54930b23299b1cfc74917 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fac39d8b06cb405b860562150d7b201b : SUCCESS in 2h 17m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a93c026bafd242ea92b38a908da97b78 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e78825a66a9472d82793714cb704d01 : SUCCESS in 32m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b12b356f2b4d425ab0e7c7073f467e3a : FAILURE in 8m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c05e7df81b04f398064d59da42ea47feb68b9faa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 20:24:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bcf993bf84f704135c2c107cc87482d57221272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 21:05:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/230cca451e294dcaa59dba85f1962e73 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572dd32e60b749118fe3a549667d5279 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aaaa32acd1a4d8aa4a4a0c5d865b68e : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5fec521ca8c4b34940f857747f65505 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebe772c73624d15be8e32482c016775 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec07f652f924ee4982a8f33c75d3406 : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c2a98136264482a894731d8632218dc : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e76201a5dfe948cda7f2a0850cf6e1f0 : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e8b9d210e994ca1992b296756cf1fbc : SUCCESS in 18m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/264af12308a14b89b58f077a85b0e840 : FAILURE in 27m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eed3a1e7cb134b85a2e857876d2bfc15 : FAILURE in 30m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/95f5ae7609df44eba4d24c64dd63c2ff : SUCCESS in 17m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d5f64bdda36443d8bc83ca8766daf67 : FAILURE in 34m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bcd7852541f54110bb4f240ff0182f24 : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/80a8611b716c455186d4bf7be8c81ef5 : SUCCESS in 31m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5a07158bbe8244d1b8cca4a97a2b14b4 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":2},{"id":"36efb0a08b7c224f09091553b31ec5c9e6e563e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 16:51:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a2571766f22537f05b01a7575cd53537b49f504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:19:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4f08d6d738446da9de61ef7bcce1547 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0569490c0e48f180de42e079f766ae : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db190663a81c43e58b7cc177c068fe23 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33d755cc214f44dba12b98ecb11afa08 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d711a0e9c704f24bc2538695fa466fe : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c34e09982eb4ac787898e73391692a9 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/109c0e35d75b45dca7fb249a83e66d48 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b05a0940483642ba8269656c5ece5b07 : SUCCESS in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25f7ed2b4d864d73b7c1e900633975c3 : SUCCESS in 17m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a4b351f51d6a45a1b193c2e705735f3c : SUCCESS in 55m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e708f55655f04107bfc26c87a37c193a : SUCCESS in 57m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4d1b445e9a3b41209543b7a41898d359 : SUCCESS in 18m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e40c7090f9d4f84aa834a1fc3917050 : POST_FAILURE in 2h 25m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8ded42c0c9374bc6bd1f29dfb6534c0e : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b40b715721749e986a6186c58290a29 : SUCCESS in 26m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cfc67ea85305456e8935bd6f00fa9367 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6191444f058b50130e7969712eead1022b128f6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-25 08:50:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"205423f5c61c933062da73d703c7f3c21b1e2b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 12:42:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9874a1e002844932bae8fda43bf00c67 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b761c62cf046b49e6b5fac78fba048 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee196e018dae4dac88ce3d1f7763f377 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5879d3a342d499a928083f85eac9f97 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73bfc94b64c4ff7b53ba8cad5230150 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87e1309f7e374b97b30490fc11f5b8c0 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8726bbdb61d4954ab39bd9ec9115375 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/160d7368013841b587fe5e67441ced88 : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b17b105cd1243948ba2deac8f8ab506 : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3fa17172308640fc9611c3a2c4ee2321 : SUCCESS in 58m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3ffdab2d946d4a24ac2c6a6da5be5e57 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b2f2e4719d99470e8fa1a1c1bb72280f : SUCCESS in 18m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed9b2c6775ac4a9595e90b910e3b54f0 : SUCCESS in 2h 11m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2fdbd6e117f047c58c79d067043d0525 : SUCCESS in 35m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/807a82d6169e475f9b976e8c45d902d9 : SUCCESS in 25m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3a86c3a125594fc489200e9013359e3b : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c3de554f8c66e43cd3d4cced69e81bb590f43c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 12:43:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45bc5b9d9b1f0cccc6f05eeb1532a8ef6193b7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:55:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b7a6c9e62f43eb967db7ffc00016c7 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c064af7813e43819109ab51797fea1f : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a1ada81f514bcaa37e72a9e068ac97 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291ff1b0bbf7468a844d2cf84f704042 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/019a88923e5c4267bee7b9bcdad9f2bd : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51012d8777ec490ca7101fbdf48494fb : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/802f177252fb4aa8a0919fd72ac6e880 : SUCCESS in 7m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f73bc0f677a24f36a1492ab8f86e0f4d : SUCCESS in 14m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3e59d334c7e54742bdd13bc1b80fab45 : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/009ab5cd4ef64a3b98faace0796f39fb : SUCCESS in 49m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/18ae050265ea46099c3eb69957691e70 : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/15cd99e77aa6497bb26c413109558687 : SUCCESS in 2h 10m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/55383e7f35264e34ab512e44d7f29a73 : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2567e56870134fd5a2a8082366a668c9 : SUCCESS in 22m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a436e4e15624f338be6758699c31c60 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8f064672e5614ba7fadd2ace352c5e00c47c9e74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:55:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"49b7df8ea38b26738e2a95f9b508450a9b4ba4c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 14:58:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2be24a766143a5be3571d91d87e0c1 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ea1d57d0ae4822a5c257f4ee0793b8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd3b16d27cbe30ce453e5473b64bde66ba2197bb","revisions":{"18825f0896c05402685e53860ef6c5a07f7f30a6":{"kind":"REWORK","_number":1,"created":"2022-01-07 13:35:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/823765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/823765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/823765/1"}}},"commit":{"parents":[{"commit":"f235bc8661d4da6d8e345e79327598c8f7df3e1f","subject":"Merge \"Build neutron-sriov-agent image when enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f235bc8661d4da6d8e345e79327598c8f7df3e1f"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 13:35:16.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n(cherry picked from commit 4321471a13eb949c2f5d57784557ddebb9b682f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18825f0896c05402685e53860ef6c5a07f7f30a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18825f0896c05402685e53860ef6c5a07f7f30a6"}]},"branch":"refs/heads/stable/victoria"},"bd3b16d27cbe30ce453e5473b64bde66ba2197bb":{"kind":"REWORK","_number":2,"created":"2022-01-07 13:37:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/823765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/823765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/823765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/823765/2"}}},"commit":{"parents":[{"commit":"f235bc8661d4da6d8e345e79327598c8f7df3e1f","subject":"Merge \"Build neutron-sriov-agent image when enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f235bc8661d4da6d8e345e79327598c8f7df3e1f"}]}],"author":{"name":"Kayobe deployment user","email":"will@stackhpc.com","date":"2021-10-28 13:42:24.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 13:37:29.000000000","tz":0},"subject":"Generate kolla config when running database commands","message":"Generate kolla config when running database commands\n\nChange-Id: Ief9adfcff28d269fc80a073cdde536e8e803ca37\nStory: 2009648\nTask: 43807\n(cherry picked from commit 4321471a13eb949c2f5d57784557ddebb9b682f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd3b16d27cbe30ce453e5473b64bde66ba2197bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd3b16d27cbe30ce453e5473b64bde66ba2197bb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
