)]}'
{"id":"openstack%2Fkayobe~823766","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I6946b578f23843dd3cbafcb185b8d8cfbb00c080","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-24 16:51:13.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I6946b578f23843dd3cbafcb185b8d8cfbb00c080","subject":"Fix seed VM configdrive when admin network is a VLAN","status":"MERGED","created":"2022-01-07 15:51:17.000000000","updated":"2022-01-24 22:10:19.000000000","submitted":"2022-01-24 22:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823766","cherry_pick_of_change":823764,"cherry_pick_of_patch_set":2,"meta_rev_id":"c9ab247400a3309d1dd269d4453aa7e1af944c0a","_number":823766,"virtual_id_number":823766,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 22:08:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-10 09:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-10 10:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-10 10:17:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 19:28:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 10:17:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-10 15:32:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"5f9b323be81b5d8f49d427978360e5bf07cb40e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:51:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* kayobe/plugins/filter/networks.py","accounts_in_message":[],"_revision_number":1},{"id":"594a37d17037848c24268db91137032123775560","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:53:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"87d052647e2a158c44cbc054e38af2dd2f6dd5d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:53:57.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"bbbfe07badd4ea8dfeb2e5ca3ea112560dc6ed99","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-07 15:54:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"af6b33a51cbb3162168988a4ffdf084ee782ad69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:28:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b8f4d0eef944bf88b39cb036ebf5389 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df732ed44e3d45ecab063da470f12555 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/760c3009b19044da98170e7e811a25aa : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe80169c9794c4fbe5bfe2e316961d8 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbda989975154260b17b7347d7e7dd65 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009acee0f96646b08455b8f31357a263 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe01babc33e143129b50efefa5c7c89a : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8316e220b0e4d53a50d3052363d803e : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e7079bd71df4c99ba3ff496d9c94c37 : SUCCESS in 16m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9f53f292ddb44ce8aff0c43e8c453d6c : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/92cd88d151384d6e802f32675867d2a5 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f47ca3e310c34bb5a53d006a0baf2b08 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc459e4551bd45d7851dee8e64c73f56 : SUCCESS in 2h 19m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d26daf4ce31b4cee8e4551733d8f2df9 : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4f0337790b4b43fbafe8d6ef9e777428 : SUCCESS in 30m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/34cd3fe6ba9b413abef42667c789ccfa : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b40180b49445104795e285957a9a444dcf537328","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:19:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e7ddf4b40768262e25d4b20c797b69c470342c87","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-10 10:17:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f2db2c0f4e8d4ebdfd411d081a6ed8941e1c98e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:19:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e9bcd21ec273f0a451009162d98687cbdf01e6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 14:35:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8de33e1cda54c6eaf60c2cb7d00d532 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17a156d7397243c5b912f72539ff7c9a : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ec070e219764967aaaec569a8737a10 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf142cc244d46429b3124caf1d4d447 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/064590050f3b482290532da3431e4f53 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb48f764dfed42d0960e8d672ef9ef84 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c4b88807c6547c288b9da03864213d3 : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a931c68aecf44fa94639c875a55087c : FAILURE in 5m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/192d418136374da9ac93b55c8ca9d37f : SUCCESS in 1h 02m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d1971897de5f47cbaf8d41715967b0e8 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/042dd38768b14d81a915ddd4940eaf31 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf750037d78f4e0a8f89704025f3881a : RETRY_LIMIT in 1h 23m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/209db1df5cdf4a1c88bd11aca0f0185d : SUCCESS in 32m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e2d240fb3a04354a2214918e8564fe3 : SUCCESS in 25m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/86523364196f4fa598c1e94c04df1f0d : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":3},{"id":"97de524f5c973b106f7eecac7bfdf0314d5bcee2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-10 15:32:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"366069f0a1411a4209c5c5c9927c32333848d77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:53:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c33b9a4a255142319082c0dfcc026bb6 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd81f83747e444069f30963f639e1949 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ca67dc2eae84aca810e9641d0413161 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8ed639975246ecbcf34a7f0f8a56f0 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4fc82fad7942e9b2a7a1daa4e919c1 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b1236087bc74f43ad6b7172fa73df9b : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3337c0830e264ce398bf7346a7a09870 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5017bff1994b4a3b9e1d88bc2ec380a9 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30bda1b5b46d4cc392ea652b826dc39c : FAILURE in 7m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c28b3d29e2474affa1c9b911b44a5752 : SUCCESS in 55m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/65e80803aaf54d848d22ad7e224821d8 : SUCCESS in 58m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3e15ffd8b5e54402b37acb1a2b712f1e : SUCCESS in 17m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bdc15c59a1574a0ba9932262cd933bcc : SUCCESS in 2h 07m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/effd43b651e848acb92772d04dc64d3f : SUCCESS in 33m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc85185926e84839ace8e19085b8c1e0 : SUCCESS in 33m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/710ed16e0e2d4621ab8de5bdcf4593f0 : FAILURE in 8m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3c2b3c4d941cb089a562cbf8f81b1522480ef18b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 20:24:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5078a3d46015b9a9bae258f135250cedd947aff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 21:04:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f2d303fd374938a4060eadc0cfd550 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333d02ee63a74361a72c93279e0cd020 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d95aa86b0eb42378fd9fcc13be9e9b7 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb961ddc72734d9fb804ea92e59191e2 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2098751fe622410883f61b9e64f0c31d : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9da08067de549568a1bc33aa50d1517 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28963b57fb994620be2f44e86ba2db53 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98b3dbfc4b64409a89c5138a9de7937b : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85a4f506efe941328ffa5d1f963b53e1 : SUCCESS in 16m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/de1505b61a5e4dbc95a0dc97db8254fe : FAILURE in 23m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d8c8f3d3661d40b3966923ae690fa2a2 : FAILURE in 23m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6074cd10d7e74075a84c8ed9f0284238 : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2754cf0c2ce4cb684ad5751c9826807 : FAILURE in 33m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8881d391e2514e41937e7982bb531a58 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fded4e83b32b4defbc2298990ee8f2bd : SUCCESS in 27m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a367234597148e6800e4ddb33779137 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2b63ab6760a2bc4355e29cf17df5729510ead884","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 16:51:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7e451f0814af2ca15c28457f4d8f3573482a554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:59:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/352b8600d9f74e68abbac5a62405d5ca : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1583953ed64f98afb28aeb75855c9f : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd395cde3efa456f8e20d7e486a70ac8 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e50503ce9142d7ada093aa942f6b67 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/023c9e5ae7cc42cea50fa3da26606395 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3176b6b3687a48d89f1d7e81c6b57b9a : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd1de4d16cf34134acd1b85dbc962315 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d99c6b25a90e4503b9a8fb3cf77cc4fe : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/145e457b64194f1baa842097b6e79507 : SUCCESS in 17m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/42e1be8be62f4ef6b13c1039d5afea9d : SUCCESS in 51m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ee2ea0d5ae0f487990cc4ad6cb8c335b : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/97c32860de994db19c8bba9028e32b8c : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53da3fafe60b4b4ea21a90548cb45b7a : SUCCESS in 2h 20m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eaaa1dc4dba643458f5d06d15729ea91 : SUCCESS in 32m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d4dee05ad5e54f51be2f799906b794ea : SUCCESS in 28m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b61eeaeef7974adf9d39c9090a456fa4 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e6395c70ad094c3241d8e1e3438cf459d7241f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 20:00:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b5d18e2f059fe4ee5116304c5d0738691ded80e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:08:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f1f130eaa54e5fba12328d613f21d4 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4926df5f0a6e4aeca7de3e32d10ab1a1 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8678af1121534b1496ef559f09b9bbbe : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2131c35af734bb0aaa4153c4a8d206d : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c55a20ef9184490fa769b05abbb0a9a5 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b86f1ebc66d40f08ccca6d3f4342fea : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b0c10cb0e0f4d86b9d1cf515f80b72b : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/beed739538d44c9da1441f710bd90d97 : SUCCESS in 19m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/efa62b1ec1014768a5903bf8a2bdcfa1 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/88bb9bbe20104c5ebe84b7642c63dbef : SUCCESS in 55m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/31f8d2d79a9840f49a218290c91b273f : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1adfcb6815044329726797768548bf2 : SUCCESS in 2h 06m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d5fc6d6e5b8d4ab8b05ec9c0e61aef31 : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f964811206b14b1196188894e3f72637 : SUCCESS in 31m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dcdc7dd42b4441479b2a2e266b9f1bda : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":3},{"id":"86c79af868c146b5c61ff52a4fa3650fcafc1c33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:08:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c9ab247400a3309d1dd269d4453aa7e1af944c0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:10:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13eb2de332fc49eba47a3b7beb2582b3 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08effd0f77dc4b2f8e534748cc9bd824 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"091d191ba654b174bc84bf92920f22227f2f9758","revisions":{"9296193a214e2745d727fa88073f8c4a3ce8ddd3":{"kind":"REWORK","_number":1,"created":"2022-01-07 15:51:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/823766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/823766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/823766/1"}}},"commit":{"parents":[{"commit":"f235bc8661d4da6d8e345e79327598c8f7df3e1f","subject":"Merge \"Build neutron-sriov-agent image when enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f235bc8661d4da6d8e345e79327598c8f7df3e1f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 15:51:17.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n(cherry picked from commit bc0885f5397af9ee69f8c873e57ad6f3d7a360e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9296193a214e2745d727fa88073f8c4a3ce8ddd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9296193a214e2745d727fa88073f8c4a3ce8ddd3"}]},"branch":"refs/heads/stable/victoria"},"5e224ab6dd063d01590245d0fc5b7249f67717eb":{"kind":"REWORK","_number":2,"created":"2022-01-07 15:53:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/823766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/823766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/823766/2"}}},"commit":{"parents":[{"commit":"f235bc8661d4da6d8e345e79327598c8f7df3e1f","subject":"Merge \"Build neutron-sriov-agent image when enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f235bc8661d4da6d8e345e79327598c8f7df3e1f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 15:53:08.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n(cherry picked from commit bc0885f5397af9ee69f8c873e57ad6f3d7a360e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e224ab6dd063d01590245d0fc5b7249f67717eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e224ab6dd063d01590245d0fc5b7249f67717eb"}]},"branch":"refs/heads/stable/victoria"},"091d191ba654b174bc84bf92920f22227f2f9758":{"kind":"REWORK","_number":3,"created":"2022-01-07 15:53:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/823766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/823766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/823766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/823766/3"}}},"commit":{"parents":[{"commit":"f235bc8661d4da6d8e345e79327598c8f7df3e1f","subject":"Merge \"Build neutron-sriov-agent image when enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f235bc8661d4da6d8e345e79327598c8f7df3e1f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-09-21 13:40:18.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-07 15:53:53.000000000","tz":0},"subject":"Fix seed VM configdrive when admin network is a VLAN","message":"Fix seed VM configdrive when admin network is a VLAN\n\nThe jriguera.configdrive role does not add VLAN\u003dyes to the ifcfg script\nof a VLAN interface on the seed VM, unless type\u003dvlan is passed in the\nnetwork device list. This change passes this field for VLAN interfaces.\n\nDepends on https://github.com/jriguera/ansible-role-configdrive/pull/11\n\nChange-Id: I6946b578f23843dd3cbafcb185b8d8cfbb00c080\nStory: 2008089\nTask: 40790\n(cherry picked from commit bc0885f5397af9ee69f8c873e57ad6f3d7a360e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091d191ba654b174bc84bf92920f22227f2f9758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091d191ba654b174bc84bf92920f22227f2f9758"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
