)]}'
{"id":"openstack%2Fkayobe~823894","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I16d721539e5956f0d86503dd5b01ea608b129873","project":"openstack/kayobe","branch":"stable/xena","topic":"bifrost-netmask-filter","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-16 16:15:52.000000000","reason":"Michal Nasiadka replied on the change"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-16 18:39:13.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-11 09:25:15.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I16d721539e5956f0d86503dd5b01ea608b129873","subject":"Use net_mask filter instead of ansible\u0027s ipaddr","status":"MERGED","created":"2022-01-08 13:50:37.000000000","updated":"2022-02-16 18:43:37.000000000","submitted":"2022-02-16 18:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823894-bifrost-netmask-filter","meta_rev_id":"d77673be8139676b848c6e37873a1aadd42cdca6","_number":823894,"virtual_id_number":823894,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 18:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-16 16:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-01-12 09:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-02-16 16:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-08 16:06:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 09:17:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-29 16:47:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7558b239f87b897ed085666311f257602bef22ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-08 13:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cfee64578b623c20584e3ae913b53c0f01b6600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 16:06:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7f269a6d8844c297d76153f4548bf5 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71c7c914fe0441b81db07336dc3bc23 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1c0c072f174c8ba8605173ee38b3d3 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d2e36a43bf4b06a7f022d995f95233 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614350bada154ecdb2509c0dcad591c2 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c72bce0d6834f3aaedfc4873342cfbf : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1f72d2bc90a409f9f8f13c905efd6b3 : SUCCESS in 8m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60de251903fc49b18b63a9bc8eb4d0b7 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd1592567fe645a09919218642700398 : SUCCESS in 33m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/df65932a46544a12a53a7306b7a0e147 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a0aae29619b47079ab32ed195b7db51 : SUCCESS in 46m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6623bfc3b476479aaff80c255591d61d : SUCCESS in 1h 22m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f7535e17904e4192954d0db367ffd63f : SUCCESS in 25m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d8c4d67e0a44882ab0e94a4a61708c3 : SUCCESS in 15m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a8bfb111ce14c91b47a0ff932079d26 : SUCCESS in 2h 09m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d11bafe88d934ded8dc1884986894059 : SUCCESS in 56m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fee2326f9ec940839079b52803081897 : SUCCESS in 43m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b39e645709943678094e20f4beb68bb : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16edb2b5d5f24104bf03f787433bf592 : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/057640d2cce84e6884f16b6f5c8b66d5 : SUCCESS in 25m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6ac12f5a51f84cda9fcc1fcac061a21b : SUCCESS in 17m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e58d530546b40848ea6022091becfa7 : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b8bbcbda6cb34119bdcf7ab2f7770c0e : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b0ccf9bb5641b0ac26d62a283a925d : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4f3ca504c1730e1bc416a45aac178a349f6368b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:17:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8d356333b7756e80d08ae588583e71ba8601a83","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-10 09:48:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39c5a6cabf16db627c85e8ad658a1cb1df429058","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-11 09:25:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f859a1e5b4a9c0e13e37261f9086f7d900da0b35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-11 11:09:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eefd366434064641a4240fa12345158eff5cdee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-11 11:09:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a847f2d5730aba14d1faa73e5fd948bbe6982399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:19:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24e78007ee0a4acea12b24154299b5c3 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1592bae3a3a4f6baec7524e7f409cb2 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b217fda4be449b3ae26cb616e111634 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a602350c8c34a8d83576c55905c3ea3 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99e4735eb1d54faa8509ad3d28f9e86d : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6564bec2137544e5b017866a96c0bce7 : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99e1e17d110042b3aebf634519f9ec65 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9acc6fd706c8438bac25671e190d7a90 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d393f595965411084c20e6551ebad7c : SUCCESS in 31m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d52d42b636ca44c3b57e115505aae779 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b89c14a6b0e45ebabad61fc4555f35a : SUCCESS in 46m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bacee49df080453e8510d95dc534fd09 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/72cc5a3c862344beb8c398293d622091 : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69b544ec21b34eebb25b644289a08a0f : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ae3ec220cd04efeb5041f8688b063ea : SUCCESS in 2h 03m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a718c4c2e8a40189b82c3d2db484004 : SUCCESS in 1h 17m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6fe53ba2306d45a1ae447b6c407e8313 : SUCCESS in 41m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/210b7a73fe6b461c935bd95296018ea0 : SUCCESS in 30m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bfafd8a35e274a919adbf5b8dca37e86 : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbc5778a58db448eaa817390912fe527 : SUCCESS in 23m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d4c842dbc2a84007a16da2b756b0cc64 : SUCCESS in 17m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/741ec81bcb4d4a4bb62097f94ea4d98e : SUCCESS in 15m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4289a3de7fd64ca5951f75d54f1721ed : SUCCESS in 18m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d5d01ae95c4966ab582d4ce3c31c42 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b73619f868a1dc1842c1d91c69af36b0123db52b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-12 09:28:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13e8ec180c16b8cb8b67c52321fd2dad2e82cf12","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-16 16:15:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"553e7a2899f1df07549a50d57ff3ec1763eee845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 16:16:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bc4c96434ad2b764bf8c210ca78cd51ae3c3d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:39:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1fc4c929ff41ba80c1811214f748cb : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3efc05fdc74550815f6d5e134af8af : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12cb14c9ccec4d08a7414aad58670a66 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e271f2cbea54a5a9ecd4352c189d221 : SUCCESS in 9m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/764cc36887f14740964e28c506a2e66e : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c0d65fb283a4ed2b877852824e3be75 : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70fdb6a219154279a5235da15786cdc9 : SUCCESS in 32m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0573bc920526449b807eba4454a344ac : SUCCESS in 58m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e09c2b5355541e6a9cab92989946012 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c61c6bb61239429e948cfc182bf3513c : SUCCESS in 59m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69597d29190b4b47baa959a823cf56f3 : SUCCESS in 25m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68eeb7d0dd4f466a841e62a05e7c4fa5 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e060144786c34ba0b6ebe1ee4c36f796 : SUCCESS in 1h 55m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4d09716727a4ac4a233c2648c57d7d4 : SUCCESS in 1h 04m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e0cf1d7f1b9e4af0a08dd35b4c25f353 : SUCCESS in 39m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51afb0e814494a82bc158da5b7a3f4b2 : SUCCESS in 23m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a46111ce2d7b4a9297daee8227e6f015 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee8481ec8776489c9b1195e09ddda671 : SUCCESS in 21m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/051ed01765df4ed7916011bb965f23e2 : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/319cebad59b846218d870c50a81d65b1 : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/00467d0fc53a4f91ae8f864b970c6fee : SUCCESS in 20m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62ad75f0b03147ad84f97c8e25affaca : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f4046018377a19316ee9444ea978c0815faceadf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:39:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d77673be8139676b848c6e37873a1aadd42cdca6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:43:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465f4cf77a2b4cb2a8b6848ed12b9a4e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"717c6321e598ffd129b701a2708cc0a9828109a2","revisions":{"3eff739b7d622e39d3a8b6f07201bddbe795c492":{"kind":"REWORK","_number":1,"created":"2022-01-08 13:50:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/94/823894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/823894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/823894/1"}}},"commit":{"parents":[{"commit":"a2c7237b746a67a0b136cc30e822849cf1475898","subject":"Merge \"Document that extra kernel parameters are important for inspection\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2c7237b746a67a0b136cc30e822849cf1475898"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 22:39:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-08 13:50:04.000000000","tz":180},"subject":"Use net_mask filter instead of ansible\u0027s ipaddr","message":"Use net_mask filter instead of ansible\u0027s ipaddr\n\nFor a long time [1] we have net_mask filter, so no need to use ugly\ncombination of the net_cidr filter together with ansible\u0027s ipaddr\nfilter.\n\n1. https://opendev.org/openstack/kayobe/commit/dc45036c92cb28fd4e9f2d9c84a03438bf7ad26f\n\nTrivialFix\n\nChange-Id: I16d721539e5956f0d86503dd5b01ea608b129873\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 0751a316f4ef093221ae3bab2d6514689cd980de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3eff739b7d622e39d3a8b6f07201bddbe795c492"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3eff739b7d622e39d3a8b6f07201bddbe795c492"}]},"branch":"refs/heads/stable/xena"},"717c6321e598ffd129b701a2708cc0a9828109a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-11 11:09:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/94/823894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/823894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/823894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/823894/2"}}},"commit":{"parents":[{"commit":"a2c7237b746a67a0b136cc30e822849cf1475898","subject":"Merge \"Document that extra kernel parameters are important for inspection\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2c7237b746a67a0b136cc30e822849cf1475898"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-08 22:39:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-11 11:09:20.000000000","tz":0},"subject":"Use net_mask filter instead of ansible\u0027s ipaddr","message":"Use net_mask filter instead of ansible\u0027s ipaddr\n\nFor a long time [1] we have net_mask filter, so no need to use ugly\ncombination of the net_cidr filter together with ansible\u0027s ipaddr\nfilter. This saves minutes on CI during the day.\n\n1. https://opendev.org/openstack/kayobe/commit/dc45036c92cb28fd4e9f2d9c84a03438bf7ad26f\n\nTrivialFix\n\nChange-Id: I16d721539e5956f0d86503dd5b01ea608b129873\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 0751a316f4ef093221ae3bab2d6514689cd980de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717c6321e598ffd129b701a2708cc0a9828109a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717c6321e598ffd129b701a2708cc0a9828109a2"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
