)]}'
{"id":"openstack%2Fkayobe~824103","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I65631db4c4fd2a3248cfb182d78f4c71683dc150","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-13 09:14:28.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I65631db4c4fd2a3248cfb182d78f4c71683dc150","subject":"Pin rich to \u003c11 due to breaking ansible-lint in kayobe-tox-molecule","status":"MERGED","created":"2022-01-11 09:17:47.000000000","updated":"2022-01-13 18:17:22.000000000","submitted":"2022-01-13 18:15:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824103","meta_rev_id":"5f9999641b4395b2f4ef14b8b4aa680abf92e679","_number":824103,"virtual_id_number":824103,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 18:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-12 08:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-11 17:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-01-12 08:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-01-12 22:40:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-11 17:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 13:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 14:04:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-01-11 15:59:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-01-12 08:59:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"02f15d8c17e8e2f3b4cd13f0023341fe8b33d027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-11 09:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7cd0894c50c13ace826a53adb4b7b1af0a698b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 13:25:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d1dc8256e742499d486df7644ba2a6 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fc2dbdc334443ea76b7e5124617098 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a869e1ec42574a2da8264f56db5e22a6 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dce79370b40481cb37b6d2ab97dcee0 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ebb021c4874dba873a66b8e2330d70 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dbfb6c2e3994c49bdc23ddf54570e57 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35505ab2054441c19ab381b2a0dc1ed4 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbbd1233a28a42418be97b3820567f78 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b706cb24d6ab4acabd52192f17acb5ea : SUCCESS in 18m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/653ba93206324ef0895313d8164ed753 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fc48816c47734a3cbc56b8dc48091405 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a750d2fbb6f843b7a956426728f102ad : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2e2c0ae5b5a642a88453a769b19bdb2f : RETRY_LIMIT in 1h 14m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8d2763cc774d46d09b0570b73737f739 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0010e8dd7704777b712549dca6364e5 : SUCCESS in 26m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/689908a314de4e00856db9a0a2fbb649 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"195bc22c513dddaa077772c8c710f4f5e60e9a3c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-11 14:04:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd57a56892bb433e709018bcdca90fdfae1acc2c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-11 15:59:45.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ff1c151a63afe6a98d148c5e6c2d43f6d8e835d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-11 16:07:22.000000000","message":"Patch Set 1: -Workflow\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e5d2786ef163c3ec017906e658f01ac3d285ed8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 16:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1942df113194c4f80af5b52eeb208b7 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8469ea7b7243cd8ee3bb831943fa6c : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9386c05b18fc4fd2b6383d6ca7083aad : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23245ec22fba48aab22a5888a989f50f : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a4fc6f1df3440b8573888325ee8111 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9708677ce85e4500b696027fee2e0f0e : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00886e9867864ea798ebdf97700b4b08 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2bd582754d894928965f5891f3aca2e8 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b55a79589074baa9b8aa1ae172bbb59 : SUCCESS in 20m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/68dd331e8d244cbd81f2d6ba88cae931 : SUCCESS in 57m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a7b78883dca14839908d6673cb731c26 : SUCCESS in 57m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/de1870796a994af6b481b9344d3ddd74 : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/feb7e25166824241a964b93920cf094a : SUCCESS in 2h 16m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7b0205c3f4a542c08d6c01688586594b : SUCCESS in 27m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4405efd736694d18b37877435d7ce01c : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/881eea6d142c46528e0382ecd09b23b3 : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5c8249f07179aa56ae6db2083082c8d87ea97ce5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-11 17:02:03.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e4f305bf5617482ffc5d2bf866aa9929daa01326","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 08:59:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d4d875c290c3cdf079f655fc1f61703fbf459fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 09:00:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41cdf856185a667293c50669877ea9c2888d6d3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 13:11:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf9c3b1d80d74e93b44116a0d005ff72 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b3efc624dea4f3cbaf223bf00db9929 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4af1670a3148e3ac0a21fe4582eab4 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8880e93a22a460f81a35ba4b4ca1689 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43035fc32bb4436fbae40afdce49f83b : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74576b2dc5a5432f8f4003a54e646ada : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab617ae8f1894e9b98ea3d9ed65caf10 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f02f134e6f2450aba2fd03be12b8a9b : SUCCESS in 18m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4207837bf56c4718a697794cfd67d021 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/16705b5ae8c94980a4f77667685ee94a : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/675bf2b6075a4798a7dd24f73e966124 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/744d1ca299114945adcb97c690fa4567 : RETRY_LIMIT in 1h 17m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3658d7f23a414bd5b78e28b96d3f4c0b : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8d9ce5673eca4d87bde8a50d2308c188 : SUCCESS in 30m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/999066d1c99d43b398d94dcdfee46757 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5863c7f44a424a6e62b346248378c2b7335468bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-12 13:20:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d7eace9f79afcac8fe6752b042fda11bbcf1569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:29:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6968c180254842388ae96ea2c091f1d9 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a547d43c862415eb3d7ff49859fcad9 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5882af62c0b74b89894b847ed2e1b300 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b316934ff82a45789cd405443eb9318b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63d3e6e74e7448ba1270a489b321792 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cc081a32f0e4e65a8fe10ed59745579 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/56ecb2e6568b4dd6bd01748fdeff5719 : SUCCESS in 7m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01ca8cb2473647dbb3172d7d0446ad2c : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd267b21754c4260abd802f184bdfdba : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f20d50d1de7e4baeaaa8ef197420643a : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e3b25c6b297045d9bcf100261fb6ab62 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ddf7300437a34047ba9762069bc7fc04 : SUCCESS in 19m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/805c9c83a0a747829ed54498b8dd7938 : SUCCESS in 2h 29m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8286eeb663a0449cb4af8cf9fd7650e2 : SUCCESS in 34m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eddf2582aa424d009a8ae8057979b67f : SUCCESS in 32m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9312e24f758e45ef87d11af425ab767e : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"19832deb673ea0a81aaec3f97c4407a9cb3e997d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfb79179dc7fe2a0c20b5021301d82b7ac9d22cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 21:16:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3786aa950ce6458ea3b2b14fef897b3b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a7d94d6f7424e3fb18af310f3e1edeb : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2190c8ee298d4912b8443dcf9ce2b60f : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060f52708bd94385a21f335497be6fcf : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09bbca4171eb4722b3bb2c89c61826b7 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0a30177145e4084b2eaac5b1eb9c9b7 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f715e86469dc4b6c9aeb836795d589ea : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eaa74c33547c40ae8d7be9a467aaa935 : SUCCESS in 21m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4858eed4dad640a0953ce70b7ebde287 : FAILURE in 37m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f541ffa2b3e04d54a1ce17875d1f0ae9 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ed1d6fa604cf4ac5be56bad0921b5a5a : SUCCESS in 16m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/03ae6d0f3ceb4ed79fbef8ecf72da0b5 : SUCCESS in 2h 27m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/30e1ecc98d35445a8ad907827c4f379a : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/784cf9bf72104f7ba3153c0f5ebac0f2 : SUCCESS in 32m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4138206776584bcdac4bd262d2e21407 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"eae366145165889e9cb39186e45da0f2edc67654","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-12 22:40:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f6f05712fb081ea3b9a29bfc7254fc990c645c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 02:09:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b89bb82b051f4487bb7a175b01eb1409 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f363188bc82b48b5bf8d55e672a2473b : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef919fe33a274f3090f55f69bedebd93 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2e7e7a67d3d4fec88ecede1f229478c : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81baafbba9fa42e1925d20778b76e7f7 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a347d6dc3d5d4122b8c637f9531c60a3 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/590e7f8314e34bafa9cae75f26cbf672 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2009bc6ff6348d69eb5da0b2d1c51b8 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fd98358e4674d438e777a3d15997e96 : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b3d751a18d174ac9a2d9dab800e04a15 : SUCCESS in 59m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b01e24828aba46b780b84fe61b6099f0 : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/383f9727534041bdb6f4b37bc2d8063a : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cfc704de6836423c8da244f1f34c0eee : SUCCESS in 2h 05m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/17d86766bf2840dea1756d712791042f : TIMED_OUT in 1h 32m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b76dd3f33c14bcab07f2b678ce0edc2 : SUCCESS in 27m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7e947773816045ffbeb59dc8e21caf43 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4dee774cd9065379a051af0d091874af03f74ac7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-13 09:14:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2209dbe78603f8301ceecc470671b4f8b63eb678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d76633d39048378a85c2340b2ff3f4 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ecbe7d6b8c46f98068ab7729d0773b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bdba35907b94844b9747ecb4ebd4e49 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac84a535a33c4a51976e8998646a836e : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf266ba896340d9ad7fd0c1167cd3d7 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ae7a8a722f7481684668777e4e510c0 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ad6d4b16883404d882d24d799a926cc : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aeff86adc54e46dda95cf993e83b0afb : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b405a43e3acb401bbbfb21b1a3577ed1 : SUCCESS in 19m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/53e6861dbf124d079e5648c0136fdde3 : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ef82e32c1434f2aba4613acdd023936 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6101eb52600f4f609aa50a75bc64de16 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e32e397d8c1c4d7a8977019b17d4d3e1 : SUCCESS in 2h 08m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0dc3c3794f0148b1b32a3661389fe236 : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6c434d28d79d4434b5cd98d59be09ca5 : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40d29c2de73142708698a5cb58acc39a : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9dd0354e7fae86f4ce913ae1ba63325f626b202c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:47:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe41f3cc7b093cabbc6f2e4755904577dac32041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 18:15:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd3eda5a4dd4312bba1a96570b8bb4c : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae9b3f2002a94fefaf7fc3ee4172f4ab : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9625cb4bb2d14be89250b6190f5eeb02 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d65d719972d4e1aa2f8dad65b0fc6b0 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762cf035d8d94613b5f0156f5a0c6f11 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e51041f40064088923315808c5d292b : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f7932d6af0d42999e4d0853e6047f86 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d19f0ab00c1f47c987330afcb0447bf4 : SUCCESS in 20m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/54dde1cddb784b0e84c1863cf238499d : SUCCESS in 57m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c44085745b7347b2874311ae57635597 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/274c595b591e421db3a6e30527b2e4cf : SUCCESS in 18m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cfd643a85c574c2d83b00aa9f16f2e0f : SUCCESS in 2h 37m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8ddab306dad44bc3bc5a48bdfb153007 : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22b5557981604c9e8b44a476ee0fc624 : SUCCESS in 31m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc9e79344ff54941862d7f9ce5d3dd54 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c46fbfe023893b673dc5a54b3c9df7231df96259","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 18:15:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5f9999641b4395b2f4ef14b8b4aa680abf92e679","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 18:17:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41c2029e67a4e08ad36e4046dc27cf5 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5735143b5a3d4694bf40ecaa0d831fce : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2605a2f0bce4b64702cda7332aa356af18148e4","revisions":{"e2605a2f0bce4b64702cda7332aa356af18148e4":{"kind":"REWORK","_number":1,"created":"2022-01-11 09:17:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/824103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/824103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/824103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/824103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/824103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/824103/1"}}},"commit":{"parents":[{"commit":"6180c4d87c921ecb3a5e24bbc1604bf368dfd391","subject":"docs: Improve all-in-one scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6180c4d87c921ecb3a5e24bbc1604bf368dfd391"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-11 09:17:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-11 09:17:25.000000000","tz":0},"subject":"Pin rich to \u003c11 due to breaking ansible-lint in kayobe-tox-molecule","message":"Pin rich to \u003c11 due to breaking ansible-lint in kayobe-tox-molecule\n\nWith rich 11+ we see the following issue when linting in the molecule job:\n\nImportError: cannot import name \u0027render_group\u0027 from \u0027rich.console\u0027\n\nChange-Id: I65631db4c4fd2a3248cfb182d78f4c71683dc150\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2605a2f0bce4b64702cda7332aa356af18148e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2605a2f0bce4b64702cda7332aa356af18148e4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
