)]}'
{"id":"openstack%2Fkayobe~824309","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I762c845e9f3ed858f36a6da32c3faef25c6de4ca","project":"openstack/kayobe","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-28 20:19:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I762c845e9f3ed858f36a6da32c3faef25c6de4ca","subject":"CI: Add CentOS Stream 8","status":"MERGED","created":"2022-01-12 11:55:16.000000000","updated":"2022-01-28 20:25:58.000000000","submitted":"2022-01-28 20:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824309","cherry_pick_of_change":807829,"cherry_pick_of_patch_set":1,"meta_rev_id":"196e80318607a657fc2435c2f8f4edf18dd85e2c","_number":824309,"virtual_id_number":824309,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 20:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-01-28 14:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 14:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-28 14:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2022-01-28 14:52:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 11:57:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 09:12:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-13 09:49:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"82fd2a824ea52ab5be0eb72f7a154165c27e335a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 11:55:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* zuul.d/jobs.yaml\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"99db5d17491c3ad6f44bcbf681bd7ece5a598716","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 11:57:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2a4232f34494dc6756eac9f1048b1072afcf28ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 11:57:02.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch stable/ussuri.  The error was:\n\n  while scanning a simple key\n  in \"openstack/kayobe/zuul.d/jobs.yaml@stable/ussuri\", line 104, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/kayobe/zuul.d/jobs.yaml@stable/ussuri\", line 105, column 1","accounts_in_message":[],"_revision_number":1},{"id":"fa55ce0168f909c6b14a6cc1c94b35ad5ef04bbc","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 11:57:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c50fb75ebb0c248fb149d8a599f7fe97bb03e0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 13:31:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"32b6174cfe58905c956ae381cd6c29448fff664e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 18:29:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62aff39a8ad84407bad68f939d304a3a : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182a2cbf027a4aa3bdd2b3fb4e231dc5 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76b7f83688244712afd461795f95bc1f : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac687d5ceb7d40b6bec377a2a397d273 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a228752fa94669bba67b74672a4b72 : SUCCESS in 9m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f7302cdb259b4747945b9534faa15688 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a59f596abfaa4075a4b5dfad94db4110 : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/599ad7e6d65443869f4707633d6d2a12 : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e0a9cff460c4a98b9798eff8218a535 : SUCCESS in 1h 12m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b4b54bc103bb419a91faa656004edef1 : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cec67c1216364eb88f3e77202341815e : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5ddea8eb37fe45c29ca37839902209a1 : SUCCESS in 19m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f41e6240eac41928b3bf225cec7b1d1 : SUCCESS in 2h 48m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/adae3e33d00e4f1c89ac4ff1441eecf3 : SUCCESS in 2h 18m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba4f43636d954a3aa7108061b66d61e1 : SUCCESS in 35m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f35eb670c982422ab9d7372a55cd4f5f : SUCCESS in 37m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c2f9592f2dfa4a11b0d935be92f793ee : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/184e5e1a24624790b1e997f626dfe60c : SUCCESS in 29m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2d6b4a2d9ce34721918038c82e2a49cd : FAILURE in 9m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ac2bda013c8341a68643f7cbe955f5ab : SUCCESS in 16m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"eee19b9feb06f0e7b664e0dfcbc8167d8035f3be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-12 19:37:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f29c689c38d5f316558e83b6c90bde2f165ff9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 22:16:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c296f4322b7547629f6329b481909bb8 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c9ae22671b45efa8ea657beeefd414 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f86d1a1b2849495a8a75947e4363d6e5 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/672506f6216a4fde814ee6b1005b9cc4 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfc87222028427e8ead3a14b3d2435d : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f8a7fb88d6a4c949aebd717cc83ba4b : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9991360a0d594982b75c5351e412d106 : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d140e1d5022d45d2a4a6be753f73a2a5 : SUCCESS in 19m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9a09390b95c4478d8a40603451c26887 : SUCCESS in 1h 16m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fed8d62331344a65888c99fddf296460 : SUCCESS in 1h 17m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e62346b90a4643848c6a02af4ffbff26 : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/172a1267cba8436bbb450a2d07cca766 : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c4ae5eb645d0421b85d2a95fabddbdb8 : SUCCESS in 2h 25m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4aeab6641b44963b3f3f4e45f317e0d : SUCCESS in 2h 29m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81decdcf34f14d68b16dbc91ff58c255 : SUCCESS in 33m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3396fa84ef4b435d99b2b8301d71318b : SUCCESS in 37m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3a8649ed1e2342b8be5fe2ca773873d9 : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/845e85d036f64036ab57afabfd0496c3 : SUCCESS in 31m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f02f32841874a77aa9097d2a9e33e02 : SUCCESS in 15m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/00a753ab2b874eff87a24d9ab67ef0a6 : SUCCESS in 13m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"a0686df2bdd01b558ac29ca016052645dc52c095","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-13 09:12:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6131b92a937b48d4cd4835771bea1b5f4a7e80f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 09:49:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c3f1bb15a2c6bf2516ce686dd2e53af5a1b099c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:50:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"528009a62df4a32d9bb6156ae534fc837aa5b2db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:59:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea98f3fc60694468a030452cb97656c7 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a656f3507ac6415f99459fbbac3a9104 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64d1893ae11a4aeb9e4c4da8dd4541e1 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35891670876d4f11bb5e0fd7938b607d : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/567b73099d954c2495c154259bb878cd : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/884b50cb96c047f28b2d8de384f8ad51 : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7001dfcb8a3247d58e31d84df7b36291 : SUCCESS in 19m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f15dc917bf3e414abbcf731bc6f75ebd : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b1ed4ad526a443b09e39c28a74fcd895 : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/069634423f844595973f5b418ab15558 : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/39375eae676142a2b073e88fdf0ee5d8 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2180673405934dfea8ddb200e2c91506 : SUCCESS in 2h 07m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a62c2825b3724138b245f5baa9f7e1de : FAILURE in 2h 04m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/31046e96de7b471d920dd9b97bf31e6f : SUCCESS in 27m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c5eeaeddfc264ba886f3a266f69fd4a9 : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/52776ded776a46dd926c427fc6863cdb : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c5ef6f2d7e94e8fbb18090ba7d6a27c : SUCCESS in 23m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/05f2685386b74ed6a4da2f7b32ac1e24 : SUCCESS in 10m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/002db9f415ed4ebbb1d8fff90bbf23d8 : SUCCESS in 13m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"f4dc1801e1cbfc12f7a51f03b072423ed9bc186e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 12:09:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"487f34802ad19231852cc801eea30920a56c5af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:12:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8939f7e3102a494bb49c8317211c4dc5 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb496addb38487cb4d2275ce05a5860 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/465b9179bbbc45da985580848f46a545 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24e1f89ef5be42c8a268f1917dc080ef : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898ad72887b849c3856b677f4e548e40 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/217ff861e5d24037998e5426cb375687 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e2af8c6f1684bdb918caa2fe8ae06b4 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e1b818f8224547b6877abdabe0bd6a3f : SUCCESS in 19m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/305f1313b8c2442db06835e2cc00d85e : SUCCESS in 1h 13m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e4c85a7c2e46422ab2a6a355842f5116 : SUCCESS in 1h 11m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/48fad70588e2478280dbc094044c1da0 : SUCCESS in 18m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/74c222fbae724e1e9efc0b45f013dcc8 : SUCCESS in 18m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/08833675437042ffb5d3496ce3cfa5b5 : RETRY_LIMIT in 1h 25m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d3b11696b8d643dba1ca161615e09300 : FAILURE in 1h 58m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3c813a98836a4b14bef3e55ca3a2ae1d : SUCCESS in 28m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/adac7fde3e724e34bd3d08903b5b823c : SUCCESS in 40m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2e83b7df94e54b66a1bb395800d221ff : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0b463da8b6149dba5d9dbb2b0af0f31 : SUCCESS in 32m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f4a3f864bb74b85a86c36ae217f135d : SUCCESS in 12m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7d20f1bce64a4f18bc0c60efb0fec078 : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"ea98c7f1159a59d450f9490c948bc5873178ad8c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-13 17:26:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce9100c2ca1767965c506d5315bc9bfdadb45212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 22:43:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ced72d7f21e41df88e6eeb3d19455da : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab0010f88d434b06b2bfb7109c6fba66 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a948bf697ac44cb4a9d1ca24702df174 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52709e67f72042eb8ef18e2ceaa1caf0 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9cf2da1c884462bd6a30020a2ebf7e : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9223316ca4b04fdf91819bd25dac3fe6 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96ef1ad9950e464fae219cdb57288312 : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a793b0a6b7c434c84217e390e469384 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/be45574029364821b2ab8979aa224836 : SUCCESS in 1h 23m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2e6e810777f14e60a40f47c4d8f5c477 : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f65d1c3772d44e86bbdece1c528bb3ed : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd3f16735894496c9c1aa2ffa7ac2b89 : SUCCESS in 17m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e704672a209f44aa9e03b803423a9146 : SUCCESS in 1h 54m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/21bcf359cafc4dc5b766aaf584305612 : FAILURE in 2h 13m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e615fee3909d40edbc301af87cdfc4dd : SUCCESS in 34m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/04900e7a0be6495c8342e548e1f9a70a : SUCCESS in 30m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2c3097c6bc024e3a8c29d27e331f42e7 : SUCCESS in 34m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f190c292249452b8d3b63cc11c02ded : SUCCESS in 35m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f3f8a142a89e4ed7a8665f43e2aac995 : SUCCESS in 14m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/561b9ca18831446582c4a50c00a2c592 : SUCCESS in 14m 37s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"ec3dde74659facd0ca34d43286b9c72692757d57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-28 11:23:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2448cbabc5abfb4da6e8628b70ab069c9322a130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:37:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa60f3c9c35346d6a065f8fc20ca126a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e01fac7e594e80ab112af7e643517e : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c54e566265cd4dfaacad4d5345f01708 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/517c0c60e9f040868dc33ffcf5cc7de3 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0899f0270c491787c0cc7b74dc8923 : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8909ecaa286c41dea100d958b5be206d : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bd8d3cfb2804c2cae25fb9328f5d284 : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87801c7b25344ae69e9bea57b345d569 : SUCCESS in 19m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e1f44aef3d4e427a8ba5a993902afcd8 : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f178dcf22ec340b9a8c390d6bd7d8c15 : SUCCESS in 1h 20m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a957bed0fcaf49aca838e065bb2ad487 : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7de2c42453234090896d8f9fb6e664d4 : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f34c64dd42374a1595662c42d5121bde : SUCCESS in 2h 04m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/06afa1af93e04d889ba1ef9e5caaccc8 : SUCCESS in 2h 10m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9bcd6633940747a89c1909545060c56e : SUCCESS in 33m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/60ccf6fa02e34f938bd03bd8efac1ca0 : SUCCESS in 32m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/019b14eb6bb6499d91a8acaad9b24ea2 : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/89debe77a55b4732adbaed94085c459b : SUCCESS in 31m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b9b60d57389c409cab3524464a63d908 : SUCCESS in 16m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3020d8b19c1145bea1e31a58f93e42b3 : SUCCESS in 13m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"cc68ad7889196571330c9c1fe7057c524da5d308","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-28 13:39:48.000000000","message":"Patch Set 4: Backport-Candidate+2","accounts_in_message":[],"_revision_number":4},{"id":"149b2a78bbd8b57974a2448ddf0f956d16b35c4e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-28 14:52:46.000000000","message":"Patch Set 4: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d3d54782449738eb063c8256d2d957e78a021e40","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-01-28 14:56:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9c19aa373fd789ee7b4d8af822323ab5073054dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 14:57:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"779bb0ad342cf5fa68fe58aa18819877a9c10cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:19:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1896ec2f3d40a2a89ef950d017ab7e : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf2ac16c6934a2e8e7638a903634197 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01d3a3eec2824809a13ee5f2bac8b671 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db6fa64a5574c8cbad532b276829999 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f377d3e73fe74b619d43c8a2dfbad4cf : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbc02f9974c64ce6ad68ec5ae0ef14e1 : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26cadbc87ae0473ebf2f2b3516a47b05 : SUCCESS in 16m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95ce26eb283c495b9e9bf81d3f89c9e4 : SUCCESS in 1h 15m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e380707c133c4478a420cd16be75d831 : SUCCESS in 1h 16m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1afbb4dd80b7474787e4e436e6254c4d : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a65cf6944ad4460b9f3bb3117eb06b5b : SUCCESS in 17m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2969c27c4b4a4e7bb391798465fb1033 : SUCCESS in 1h 44m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a6cb6c1fc6e14a34898ca8e1973df9ef : SUCCESS in 2h 13m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc5ca7e748034c15be760eb81485cfcb : SUCCESS in 28m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/72ce6d96b99d4ec0b599e82f782d4ecc : SUCCESS in 34m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9321d99caac04b9092bf157cf2e30041 : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/864a6027e1fb435f81369d6f64bf8861 : SUCCESS in 21m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c2f795123c3a4b01bc5305f606a4f1af : SUCCESS in 18m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4f969fbc972d4148b27668abe8a392c8 : SUCCESS in 13m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"1d9d2af4cf60a141b500b69e4a90800ea57c3120","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:19:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"196e80318607a657fc2435c2f8f4edf18dd85e2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:25:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e997167423584b47afbe92146fd93c3b : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"763c043fa09a1279e7ab4c8ab841b5422d91b5a4","revisions":{"bbd94c029de62bab475365ca6ac59e106839a0d6":{"kind":"REWORK","_number":1,"created":"2022-01-12 11:55:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/824309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/824309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/824309/1"}}},"commit":{"parents":[{"commit":"824dd7462475d551c8b5df37c205f4aef67f4e58","subject":"Merge \"Add support for Ironic inspection through DHCP-relay\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/824dd7462475d551c8b5df37c205f4aef67f4e58"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 07:12:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 11:55:16.000000000","tz":0},"subject":"CI: Add CentOS Stream 8","message":"CI: Add CentOS Stream 8\n\nThis change adds CI jobs using CentOS Stream 8 after introducing\nCentOS Stream 8 in Kolla stable/victoria.\n\nChange-Id: I762c845e9f3ed858f36a6da32c3faef25c6de4ca\n(cherry picked from commit 11554bd3d58df679fea2d517b70db1b3f490153d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbd94c029de62bab475365ca6ac59e106839a0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbd94c029de62bab475365ca6ac59e106839a0d6"}]},"branch":"refs/heads/stable/ussuri"},"3ead38f4a016cd1f0722caf56c471e5fbdbc3f3e":{"kind":"REWORK","_number":2,"created":"2022-01-12 11:57:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/824309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/824309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/824309/2"}}},"commit":{"parents":[{"commit":"824dd7462475d551c8b5df37c205f4aef67f4e58","subject":"Merge \"Add support for Ironic inspection through DHCP-relay\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/824dd7462475d551c8b5df37c205f4aef67f4e58"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 07:12:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 11:56:58.000000000","tz":0},"subject":"CI: Add CentOS Stream 8","message":"CI: Add CentOS Stream 8\n\nThis change adds CI jobs using CentOS Stream 8 after introducing\nCentOS Stream 8 in Kolla stable/victoria.\n\nChange-Id: I762c845e9f3ed858f36a6da32c3faef25c6de4ca\n(cherry picked from commit 11554bd3d58df679fea2d517b70db1b3f490153d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ead38f4a016cd1f0722caf56c471e5fbdbc3f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ead38f4a016cd1f0722caf56c471e5fbdbc3f3e"}]},"branch":"refs/heads/stable/ussuri"},"e756892b1cf6983fa18aae54b3e1eaa5044a4f39":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-12 13:31:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/824309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/824309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/824309/3"}}},"commit":{"parents":[{"commit":"824dd7462475d551c8b5df37c205f4aef67f4e58","subject":"Merge \"Add support for Ironic inspection through DHCP-relay\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/824dd7462475d551c8b5df37c205f4aef67f4e58"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 07:12:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-12 13:31:54.000000000","tz":0},"subject":"CI: Add CentOS Stream 8","message":"CI: Add CentOS Stream 8\n\nThis change adds CI jobs using CentOS Stream 8.\n\nChange-Id: I762c845e9f3ed858f36a6da32c3faef25c6de4ca\n(cherry picked from commit 11554bd3d58df679fea2d517b70db1b3f490153d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e756892b1cf6983fa18aae54b3e1eaa5044a4f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e756892b1cf6983fa18aae54b3e1eaa5044a4f39"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"},"763c043fa09a1279e7ab4c8ab841b5422d91b5a4":{"kind":"REWORK","_number":4,"created":"2022-01-28 11:23:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/824309/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/824309/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/824309/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/824309/4"}}},"commit":{"parents":[{"commit":"824dd7462475d551c8b5df37c205f4aef67f4e58","subject":"Merge \"Add support for Ironic inspection through DHCP-relay\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/824dd7462475d551c8b5df37c205f4aef67f4e58"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 07:12:10.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-28 11:23:05.000000000","tz":0},"subject":"CI: Add CentOS Stream 8","message":"CI: Add CentOS Stream 8\n\nThis change adds CI jobs using CentOS Stream 8.\n\nChange-Id: I762c845e9f3ed858f36a6da32c3faef25c6de4ca\n(cherry picked from commit 11554bd3d58df679fea2d517b70db1b3f490153d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/763c043fa09a1279e7ab4c8ab841b5422d91b5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/763c043fa09a1279e7ab4c8ab841b5422d91b5a4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
