)]}'
{"id":"openstack%2Fkayobe~824565","triplet_id":"openstack%2Fkayobe~master~I350f58c8a82f0f31608b34054e804c5c198d6806","project":"openstack/kayobe","branch":"master","topic":"custom-themes-support","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-01-28 11:07:56.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-24 10:38:03.000000000","reason":"Mark Goddard replied on the change"},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2022-01-28 13:01:33.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-01-28 13:01:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I350f58c8a82f0f31608b34054e804c5c198d6806","subject":"Support for untemplated dirs in kolla config","status":"MERGED","created":"2022-01-13 13:32:29.000000000","updated":"2022-01-28 13:07:53.000000000","submitted":"2022-01-28 13:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824565-custom-themes-support","meta_rev_id":"60d34b0a2f79044528bbdc361a84c00323a01f69","_number":824565,"virtual_id_number":824565,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 13:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-01-28 11:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-01-28 13:07:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-01-28 11:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 13:34:27.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-13 13:34:27.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-13 14:00:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-01-13 14:46:57.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2022-01-13 17:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 08:39:06.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REMOVED"},{"updated":"2022-01-18 08:39:06.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REMOVED"},{"updated":"2022-01-18 08:39:06.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REMOVED"},{"updated":"2022-01-18 11:21:09.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-01-18 11:21:09.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-18 11:21:09.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b7f70d7244224b42b92ad2aa1be8aa17b44ef48d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-13 13:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0317712bf209d685c039d4ca9e36186f0d4d9937","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-13 13:34:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a63769bb693a0cb538bf0e550dc2ba76f1514a4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-13 15:09:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5efe3a1d6ee192718613766038a0f7cbd68a0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-13 15:20:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1084e365871217cb59ff3bc6cb895e133ff58f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:30:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3fd5fe1ba5436391487409079d2bb4 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ad75e0b181342cdb9d0298d699b3abd : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28eab08cf9ee4b2fbe4427f71e16f72d : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f667faa42f41559e6bd21fe34f8532 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4a9246c4404d5984cc57a8e689b0ac : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bc865ce089f45088a68cc0494b4a0c0 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed700db2617f4df8953e79bd5ae1bf29 : SUCCESS in 9m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ba4a3821a156494f8aa71576b713f554 : SUCCESS in 16m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/51a1624563f048eeac7edb5a181f4bb6 : SUCCESS in 35m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a9ba92cd41a34a159ff4d7add9b929ff : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ab997f9f1cf4475acf54fc0c7c7106d : SUCCESS in 54m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/056187b55bcf46389932cabbed3c5c23 : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/56a22688ab0b416c8c74993c992d51b2 : SUCCESS in 24m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9268adacaf3e478f9de704e234adef6f : SUCCESS in 19m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/998df69da580461387d497f11a3f9e40 : SUCCESS in 1h 59m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13f67fd73bdf450a8f1d4cd00b0843ce : SUCCESS in 1h 20m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b83a4d62837746c3a74f7ce8ba3ddadc : SUCCESS in 54m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8af705bf0834a3f879348b50ceefeff : SUCCESS in 40m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c2c862e20bb4942821d986aa8a073a1 : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8077121227a04531a41bf9521f08a19b : SUCCESS in 25m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c12611812a6490cb6318b7010dfedfc : SUCCESS in 18m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9390f7504ff74f5786a9d4c343256592 : SUCCESS in 15m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/95b148ffb8ad422880822fd55bf39dbb : SUCCESS in 21m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56706a4e98da458f957acaaaba6eade6 : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":2},{"id":"dfcc6b87b6c41dab32ee234835ae9fc6087851b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-14 13:25:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c8905dc07238b0b52b34267324c0793db6e93da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-14 13:43:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"436cd3e773417ea9b226374c9107e51188765580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 14:36:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ba63160c2de434694e7c2ab121736ec : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa938d4ee9ee4eee8a66e034883c9e66 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a22e83c373443db18be412169d7a3a : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b99a41627fdc496fb5e331d6d0ff8de5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47326693355458d9b35d35c1a53f8b8 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a99f332261774ae7a58685d3b8f124e6 : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06b1d72b3c304984ba5dd38ebded00d5 : FAILURE in 23m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/baeef4b953e946a19a5600c4f0e9e0f3 : SUCCESS in 14m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c53532515644cdbaa6f6704e09b8443 : SUCCESS in 27m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/06cef2861a9f4b9293c06b20de085693 : FAILURE in 30m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8531be74cb9f4c59b904b76fe8aa297d : SUCCESS in 50m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0ce07d7af56e4186a1bb558937d25a21 : FAILURE in 25m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5bf1092752c345dfaf810c97b08acd47 : SUCCESS in 26m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/151fe169358c447da1193c321f106825 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/014582d7caa14de28669bef43dfa939d : FAILURE in 26m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5906fd82583d492da80509cb65bf8fdd : FAILURE in 40m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/27919dab723f4899a99507267479ffb6 : SUCCESS in 51m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43cb973c72c347f881917517dd0f8871 : SUCCESS in 39m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca00136beacd441d9661595b8c54ca69 : SUCCESS in 29m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2675cc7eef6e43febf7a416c9b1a4710 : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/78db34857fb1456588d1fcdbb8ba5ed7 : SUCCESS in 18m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/436bc309ca54418987451491304074f8 : SUCCESS in 12m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/16381d9d5b364645a6c3dd8e69ecc966 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b1f6e212f44364b96deba6a280b398 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d4f8e7aea9ef1913735aad7278dffa3e68ce0e60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-14 17:04:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"896a343fe689e77d28207aad3fc068a138737b04","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-18 08:39:06.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9fb3470d0d3572bacf1c2d04475021eb9c513d17","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-18 08:39:06.000000000","message":"Removed reviewer Mark Goddard with the following votes:\n\n* Code-Review+1 by Mark Goddard \u003cmark@stackhpc.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"2a95edef671e67261fae821a6e08cc4800ff507a","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-18 08:39:06.000000000","message":"Removed reviewer Bartosz Bezak.","accounts_in_message":[],"_revision_number":4},{"id":"b7c46ec1520a8291f73b8b04d997e84e2c966a8f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-18 08:39:06.000000000","message":"Removed reviewer Michal Nasiadka.","accounts_in_message":[],"_revision_number":4},{"id":"af665ed47d8ec630bf57e51b54f59dffd74e28b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 09:53:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbc416f79fd1423eb6d2a3f7da79e77a : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e68cab49b24df6bc9abdfd39b8b212 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62fe410385db4244af26b7a83790df19 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2796dc7024884dbe9e0aeb7425a4db96 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50c4ab018624a49a249db84c5372d83 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8077fb2f0bf4e40bb00f545ede1cb04 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7a7694de6084e5bb8ccd3d1f0ccbf55 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a9fc91543ba346bb8aa7e42da612a06e : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d7787a866b346e2af38f89edcdcd9b2 : SUCCESS in 32m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ce5c1f95d0a2480f8d10fb507d83c799 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5911816db61f4f2d8daf1309c149dc61 : SUCCESS in 52m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/659d7783a6de40858ccbf89465364495 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a46c018e398a4b769ac3d7dd446dbc07 : SUCCESS in 24m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e6aa0ce374f42da97c2393b214957cf : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e42082446a9f48fb8401950314ff0d81 : FAILURE in 32m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37038c7ad24246f5b622a791d243980e : SUCCESS in 1h 09m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e415dbc4fee94efca13977d1df9cb8c2 : SUCCESS in 56m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ce29ae3784944c8863d15ac4c0611e4 : SUCCESS in 42m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf62d5872d2f44f59e107f9413e310ad : SUCCESS in 36m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d3d902cb1d248f495ebf4a58208bf3b : SUCCESS in 27m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/27c475160d544e3ca925da1e66ba06a2 : SUCCESS in 16m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46aa64ad2d834ff6af5b831d59475304 : SUCCESS in 10m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e909bba821c5444ea2c2bedeaf74304b : SUCCESS in 29m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1859d98e4f8447ac8bc8cae922e8fdea : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":4},{"id":"ee5672316b2d7c2fbcda9333bf6c210da55da811","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-18 11:21:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36210a883e72a2de8b15f56d6e8f8e564157f1bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 10:38:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1da6fc1a906c61510135f64d75b2327985b2531d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 10:38:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0a8b4caace8968aa8e20aa0d6a01c185a82a9aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:24:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0faa3db58e24b8e918b6506038555ee : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ddfec68c1e4921979116e867d4d54c : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf6c4990cb44f45a99eaa29f79aa34c : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00748bc6395c49ae90623f5c97ccdf2e : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac8fc670cd24cafb6de45a964e69798 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ce263644c342ca986ecd8b57730dca : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/039d1ec7b66a48d5b5a7da0f43f236d9 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55c85911258847ef817752133e7eb139 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ebdae32e3e4424da8f7375401d16142 : SUCCESS in 31m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d9830dffc83741c481b9d9c5ef4bf77a : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66717ddc93fd4d75b25c1f8fd4b14b72 : SUCCESS in 47m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/98c43c19a1c2408b9a411a7df3ccf0be : SUCCESS in 1h 17m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1771dfca7b344335ab786b1b648e5a23 : SUCCESS in 22m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66b45bc64d114363aa1d921c04ab629a : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/221b3ebdaf664534a0b6e97f35efb6b9 : SUCCESS in 1h 44m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f3af87307a1491cbdfe3836333889c2 : SUCCESS in 1h 15m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c88093abb7004eb697641ce077edcdaf : SUCCESS in 52m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fba3f4050a4444395a6eccc151f491e : SUCCESS in 39m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2666d784d7644513aa0713dde4fe384f : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa901dcc805941d5bd211dd07f61185a : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f8f1a7d65dea46ae87e00a70cfa030c6 : SUCCESS in 16m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01cb592ba34f41bca4a06ce504f871a4 : SUCCESS in 12m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/73843fe1fe3e43e780c92a6792efc89e : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/718e57a083624093b00c32273b15554b : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"7296ad7a8d41547c89101187190538e02278e116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2022-01-27 19:46:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"5f0c38c4457ebce24da76556a63900fedd04f7ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 22:08:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9803ee297a644c66b9579bfbd15043f9 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ab676f140946e5a682b99e977019a3 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d495c60a2a04315b7b8058be665425a : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb3af4db74cb4aa4825cc05fdfe893f4 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc22dcf81f4342859dd536b0152b61e5 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80f0207a3ce44da0aae1994a4180a1ac : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/264f1521e26342a6b8dfd948cf6cd082 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/539f13bf7d8a4fb29721e1c656b0a2ec : SUCCESS in 15m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e2f7e08693a4a21861689639951df39 : SUCCESS in 33m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70a847ecac854701950fd8ed94b10631 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a98c3b32ac4454bad801bbd7d91843 : SUCCESS in 41m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1b77d510a8d54107ac10db6ab554b0bd : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d42d5b0388f44a1784e4352219c665cb : SUCCESS in 25m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/963425c99a6d4671a7f966e076edf3d0 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57426e3659e246c4a49ad8b70867e014 : SUCCESS in 1h 55m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/192407c6e559437ebf31354093f490b6 : SUCCESS in 1h 20m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b44a5cf018b44e5996444d76e7bb107 : SUCCESS in 39m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c952b4ca637a4ae5a2223ccbfd4832cc : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ab03e62dc3324414a253fd00772bcbb9 : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0038fe2fba014007a5336287ee440a5b : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/668cde22c19a4106ae36b1622ce5e851 : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b60f8eb3e544524af853a0276793779 : SUCCESS in 12m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6f37041399204d74bc596f9289c4bbca : SUCCESS in 21m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/091da1a7d26b4421ae60ae6e91cfb11b : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":5},{"id":"a3d2acd4d7444c68d52ba0bc86a4dd020ecd45c3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-28 11:07:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f3eb368c692663450263622fda88f5db7a9605d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 11:09:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"eacb9a877852eff8f5cbb8ebf9127e1332d35ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:01:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c2d8bca60394a31acd4e6b33a2afa4f : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/322dfdd60d8948cb9bf74f709752275f : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/761efdabfc314c3abcd41dc4d599336e : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5db0ad31df2420d88d5414eee9b0878 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40098ddb09ef4876b16f1cef85e73ad3 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/732472ce03584efd81dc009836f3dc7b : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd3d1eee5eab4817a072a41d3e06c6ed : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4da680eccf794cc181fc1d7894f420f1 : SUCCESS in 33m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0dc807be5d65408d8e01d26b947e29ed : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/325ff282131e41c2a5c3b28c7d32058b : SUCCESS in 49m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/46a7a9e5634e4b73a1ca0b47a8168287 : SUCCESS in 1h 15m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a586a3a9eea405e805d794fbaa23b30 : SUCCESS in 27m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc5021b8472748879451812a082ab7ca : SUCCESS in 15m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f66a847029684706a0473931541bedd1 : SUCCESS in 1h 42m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a03fe57907a4258ab5467263d9326a5 : SUCCESS in 1h 19m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4527d22c3c54d57a086c4c001503f7a : SUCCESS in 39m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0452542e66c74666814150f200f2a61b : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fdbccb9f1a9248859a9774be3e5f2e92 : SUCCESS in 39m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/304426e8f5b54976ae2f0e67c4aa13fb : SUCCESS in 21m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e29aae1de75d49e08523e92a6a277744 : SUCCESS in 21m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/800e8075fb7f41059d96c036e18d5746 : SUCCESS in 17m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/07e37c2c906342778abd4c750b51fdbc : SUCCESS in 22m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46aa63e4d9c0493e98a693108abe668d : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":5},{"id":"3b16bd615c0a0bae3191083c32d4aaf1fb189556","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:01:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"56264b79f5c8969b0149d3e9571655e6d4d3073a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:07:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fbf4078e1d4f4cbbd4627a955cd523 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ba597043ee74ad9ae307e1b119d4352 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"35e0d76a96fbcbc08f600294ca7daac6288df1d5","revisions":{"8bf2d6bc75d9e07706e086d56e7a5d06fcc1f609":{"kind":"REWORK","_number":1,"created":"2022-01-13 13:32:29.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/65/824565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/824565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/824565/1"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"subject":"Support for untemplated dirs in kolla config","message":"Support for untemplated dirs in kolla config\n\nVariable untemplated_dirs has been added to allow for\ndefining directories under kayobe/kolla/config which should be copied\ninstead of templated by kolla-openstack role.\n\nThis is needed to support custom themes for horizon -\nchange If9982c8e18be31772cb031ef72b7eebd4d768be5\n\nChange-Id: I350f58c8a82f0f31608b34054e804c5c198d6806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bf2d6bc75d9e07706e086d56e7a5d06fcc1f609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bf2d6bc75d9e07706e086d56e7a5d06fcc1f609"}]},"branch":"refs/heads/master"},"41c489797a6f2b77f80f0885d5196b8d677b0ead":{"kind":"REWORK","_number":2,"created":"2022-01-13 15:20:56.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/65/824565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/824565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/824565/2"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 15:20:24.000000000","tz":60},"subject":"Support for untemplated dirs in kolla config","message":"Support for untemplated dirs in kolla config\n\nVariable untemplated_dirs has been added to allow for\ndefining directories under kayobe/kolla/config which should be copied\ninstead of templated by kolla-openstack role.\n\nThis is needed to support custom themes for horizon -\nchange If9982c8e18be31772cb031ef72b7eebd4d768be5\n\nChange-Id: I350f58c8a82f0f31608b34054e804c5c198d6806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c489797a6f2b77f80f0885d5196b8d677b0ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c489797a6f2b77f80f0885d5196b8d677b0ead"}]},"branch":"refs/heads/master"},"39fcd9a80488b3a5cd6f944197ac4f6ca35c2ce3":{"kind":"REWORK","_number":3,"created":"2022-01-14 13:25:58.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/65/824565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/824565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/824565/3"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-14 13:25:41.000000000","tz":60},"subject":"Support for untemplated dirs in kolla config","message":"Support for untemplated dirs in kolla config\n\nVariable untemplated_dirs has been added to allow for\ndefining directories under kayobe/kolla/config which should be copied\ninstead of templated by kolla-openstack role.\n\nThis is needed to support custom themes for horizon -\nchange If9982c8e18be31772cb031ef72b7eebd4d768be5\n\nChange-Id: I350f58c8a82f0f31608b34054e804c5c198d6806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39fcd9a80488b3a5cd6f944197ac4f6ca35c2ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39fcd9a80488b3a5cd6f944197ac4f6ca35c2ce3"}]},"branch":"refs/heads/master"},"454e6b783bce750cbd9405ab7cbcc5fd6d6f2971":{"kind":"REWORK","_number":4,"created":"2022-01-14 13:43:28.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/65/824565/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/824565/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/824565/4"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-14 13:43:20.000000000","tz":60},"subject":"Support for untemplated dirs in kolla config","message":"Support for untemplated dirs in kolla config\n\nVariable untemplated_dirs has been added to allow for\ndefining directories under kayobe/kolla/config which should be copied\ninstead of templated by kolla-openstack role.\n\nThis is needed to support custom themes for horizon -\nchange If9982c8e18be31772cb031ef72b7eebd4d768be5\n\nChange-Id: I350f58c8a82f0f31608b34054e804c5c198d6806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454e6b783bce750cbd9405ab7cbcc5fd6d6f2971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454e6b783bce750cbd9405ab7cbcc5fd6d6f2971"}]},"branch":"refs/heads/master"},"35e0d76a96fbcbc08f600294ca7daac6288df1d5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-27 19:46:45.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/65/824565/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/824565/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/824565/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/824565/5"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-13 13:24:52.000000000","tz":60},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2022-01-27 19:46:45.000000000","tz":0},"subject":"Support for untemplated dirs in kolla config","message":"Support for untemplated dirs in kolla config\n\nVariable untemplated_dirs has been added to allow for\ndefining directories under kayobe/kolla/config which should be copied\ninstead of templated by kolla-openstack role.\n\nThis is needed to support custom themes for horizon -\nchange If9982c8e18be31772cb031ef72b7eebd4d768be5\n\nChange-Id: I350f58c8a82f0f31608b34054e804c5c198d6806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35e0d76a96fbcbc08f600294ca7daac6288df1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35e0d76a96fbcbc08f600294ca7daac6288df1d5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
