)]}'
{"id":"openstack%2Fkayobe~825671","triplet_id":"openstack%2Fkayobe~master~I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-14 14:31:42.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-14 09:23:02.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-14 12:26:05.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2","subject":"Bump up manage-lvm role version to v0.2.6","status":"MERGED","created":"2022-01-20 20:20:04.000000000","updated":"2022-02-21 11:57:41.000000000","submitted":"2022-02-14 18:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825671","meta_rev_id":"b60d156ed7462d51618c612a77682605c6a466a2","_number":825671,"virtual_id_number":825671,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 18:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-14 09:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-14 12:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 12:26:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 22:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 08:57:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-14 08:57:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c6a2dca3de98813562844118095c277f557dae4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-20 20:20:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aa3232869706defc5cae8159bd3677fb7575209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 22:16:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d4c20ffaa84837aa1fae63c5fdb358 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca68c358afd04c59ac8aa837a9b83610 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557501e4a18e4d5887316aa11265ae3f : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/908a7c2213c14549a1c39edd0570a9ec : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a2e7e49c584f52a180f2bc1c99675a : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2ff981788d94a4fa1f961ddb8a69580 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/136c032ca3e94e398b26a0dcc10f8764 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71248db4996948e38abc9bd245739781 : SUCCESS in 36m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/89230df88f114928857e5431bf13a545 : SUCCESS in 1h 11m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97bec510943242e39940c626fc3f894a : SUCCESS in 51m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a48ae88f67c4d509fadba97ab41f036 : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/32fdc1ed5b9e4e3e88d784e106cb3720 : FAILURE in 17m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/325e07002f2848b78ab091dec178fe1f : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f30fbf997c84bceb794186f10f23bd2 : SUCCESS in 1h 52m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62e445497a7144e287d0b5cd4318e227 : SUCCESS in 1h 19m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7a4f1a62000349f7af21cd59ba126bbc : SUCCESS in 52m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2082b78fae5140afb510c68585e60375 : SUCCESS in 40m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b76d8410f3d41a985d3a21879754218 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea4058a053174e2fa706c47b6382013e : SUCCESS in 24m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed4c8171cf824e078ea37fde50471074 : SUCCESS in 18m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57f0e4ba4fb64524a7d867ba8d2bcb19 : SUCCESS in 13m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/de3ccf3cf5a1455bbb2dbd6f33c8ad7c : SUCCESS in 18m 48s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f32c3054979041b9b16f9add83a4232b : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":1},{"id":"873c474ccb11d7a077c0680611e7a8df4f174361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-21 06:18:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"097cff9ae97df340e84223d72cc1f2a6b6972439","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-21 06:18:06.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"2e04b4bdb8be666b86777990f4978f378a790030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 07:19:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da20af1df7314599b8dac4686eac727b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b38bda9542a4ae6bdb232f4259950de : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89009861babc4702b7573b0c1fe39c53 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abea1a3a2df84dc1b01f7d181471f003 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3eb5ae6d1447929ef284fa7e226745 : SUCCESS in 8m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a90b5d3d66e244ecbe94d7ee58df56a3 : FAILURE in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/134506e891874cd79e8d12396aa28c37 : FAILURE in 6m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cbf8674e2d84625a02348ba1e8bce00 : FAILURE in 7m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7829f650c154e8593beee70b04a747f : RETRY_LIMIT in 3m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2548cbbf9204ebcb6589c36a6c7c2b7 : FAILURE in 8m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6710ccb1728a4b13b5de0affdf9f43dc : RETRY_LIMIT in 2m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f6d3f80bc1e642218a6f5a5aae455da7 : RETRY_LIMIT in 2m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffb4b0de1d26449bb6b486012ead8188 : FAILURE in 8m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/59a3a1eb9b04486197837469df48c758 : RETRY_LIMIT in 2m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebea20450ff4ccba0c097d22b8d4e2f : FAILURE in 56m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/70fdd99f95eb424b8e7dd36f35e4c26f : RETRY_LIMIT in 2m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df06f39c1c3342debda64af9bf316766 : FAILURE in 9m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ded716998f704f18af5ac9b5a44ef9a2 : RETRY_LIMIT in 3m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6e5c27d4f6d428da217485f9d46f99f : FAILURE in 19m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b807442c6dab43a69287441fbef500a4 : RETRY_LIMIT in 3m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36a0978a1fcc4b63accfb535bf3a5a45 : FAILURE in 7m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a2a5d6f67474ff5a628d709bc063b68 : RETRY_LIMIT in 2m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/014ddf120bc24d5ca5327ba09d5d99d3 : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b434ede74d0412c9fdb6227a0e8b0c38b2cc752c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-08 13:34:40.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"335bcd4bd1fc4b40f845aa46f88e3cab5e5a93b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:41:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0ac987ae316477fad614ab0fa398d37 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43023565c80e42e6bd1425200f4fecc2 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f36bb7a91d4f538e865493d56890b0 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/718b870c536348c8bf7393276dd5125c : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457850b43876487d94b0098c84a399db : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81cbc2a283414a1e8cd1fa2b70fe81ce : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25a0068ab6004f908c40f4762a880ce8 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a04bfb1aedef47d5abc5235741b8785a : SUCCESS in 28m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/37ccc65957954618b52968a8b37e0f96 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816730dc08d74315b17268ee76f71b73 : SUCCESS in 58m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2d1452c946974cdd82ac590708c2d34b : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/01dc1c7e574e45b4839a8f75f926233a : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e51cdaac004a447cbb00616984c2c82b : SUCCESS in 16m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a7e4170c713c469bb3e5fe34bfe5343a : SUCCESS in 1h 54m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a46b21c239ae4ffd970cc9d5cd72c71e : SUCCESS in 1h 20m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a1f7a0113199428d971cbbb16cc6483a : SUCCESS in 46m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d6a933635cd4575ae716626ad0e8cb6 : SUCCESS in 36m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f789cf245c4a41e6bf1ec6e6c0eb85b5 : SUCCESS in 31m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a29b7d17ec6492689641fcc35682625 : SUCCESS in 22m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f03882b7afb4793be5a91ba0c6cf86e : SUCCESS in 15m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5238febced5842ac8d3753c157e6e2a2 : SUCCESS in 11m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb9e839e841a424d9ce15d8136770044 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d0f95709ea545a09ee5e55b091ad0c5 : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":3},{"id":"d6b540a354e87584b496e827cd70544922980a78","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-09 06:32:47.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fb9375ff3d232adc71b00f3f9481aea7377e7c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 08:24:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070a1d45defe41069b280af57883a789 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a813401c886f4fb6a7805c129907bb91 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aeddd5b26e74aeb833a688bdd4f6524 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c95aa3406f1f4eae91de4e8fdfaf2978 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29862a830c4041f2a0ba4f9dfcbca9a2 : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5066fd9a2eb744f5a628ddfd29b71ffa : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0abf0fdd0e37405f9f9e431d6bba160d : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6eee4a6b937344859a059ea4c755e359 : SUCCESS in 32m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c97b8e17e60a41c0af9140b157aa8c1f : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/282dd89a8d574de8978f1115bf5beb7d : SUCCESS in 48m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1656fe673f7a4f7a8d39096e7a337725 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f37cd00a7f1b427fa3101bbceb71f8e8 : SUCCESS in 23m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6e145c6639045399dcd50be40f972bb : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/07d1e539ba1644c7bf01eede5cf4848e : SUCCESS in 1h 48m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c582b51ed1c84dc2bd68ee8696f9ab6f : SUCCESS in 1h 13m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c35a5316b108492390e3d23ee278ca60 : SUCCESS in 44m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d08a94449bb4a7ebca0e1e7d32540a3 : SUCCESS in 41m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/29b57d6815dc409f90a673e5fce11b49 : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/302e4bbc8f944229b3a16203115743df : SUCCESS in 24m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/023d143b00f240289fa821b20f5ff3c4 : SUCCESS in 16m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/282b87d069cf4c9bba0ec49954957689 : SUCCESS in 14m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1dcc0af15e6c465c95c0112041b7274d : SUCCESS in 17m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f7f97e77754474a7c0e20d6a1eae19 : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":4},{"id":"20e65d8ca544569aaf6b600c1898a3f8d07d1e43","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 08:57:03.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"0a1056154ed2523d5f355a273f927464e6b497d9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 09:23:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"be1dc045e3b080e52ece244e1aab47464f2e508d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-14 12:26:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"617638559d28c6dd5cca26a66bdc9a4e3c1f0ec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 12:26:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9a58760078136dfb877dba37c96c63fc458fd0c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 14:31:27.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa59d6ae33c4ee6a52e0f276cdca0e6 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d4635fb12846d9b6f6d851592f3f42 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae9c29081cea46bfb6fa6fa759260d55 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc4d3c390834268ba0e012782eb34f4 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68b31badfa2b4db18518d0abea38ddcd : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb96305381e443e7bf9a86de803d60d3 : SUCCESS in 12m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d59f2a5dd3974167a6fc2d27aaa7c53f : SUCCESS in 31m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b13e42adc810482cb0ed5f2339e8b4de : SUCCESS in 1h 16m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d389635b8c9437b95ce324e1858cfa3 : SUCCESS in 48m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/03053d713c4d4b188f951c08b3448657 : SUCCESS in 1h 16m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/61adc656655646058bb49904900a6b93 : SUCCESS in 23m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9201328ffa641fcb38ee3097c478528 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b9dedeaab004ab899abc5b21ff4f556 : SUCCESS in 2h 00m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b89c7642034d41a8b101c92f378a8530 : SUCCESS in 1h 09m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ff69b118db6a43639220bfa48a09e589 : SUCCESS in 54m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f71b97d967048acafb4657922b92a34 : TIMED_OUT in 1h 33m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/408d402f7d2042ce9162ea6b170780f1 : SUCCESS in 30m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eced8331d2394fe58bfeef55b99bf35e : SUCCESS in 22m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a95a5025dd904ed7a5fe0fbe686ae8d4 : SUCCESS in 17m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0a8fa8554e44c98584825fa848c928 : SUCCESS in 12m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f87781ad581f49dda190317316813116 : SUCCESS in 20m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d87ab03290654a5d9b963cb018daecda : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b5fb7fb21314704642343be2abceafec127bf672","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-14 14:31:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"09a74532cbef16a825ef3a9e752a05dda8484a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:27:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f3d631041ee46c4a83c1d4e1f89e8d7 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b242f291a442d89f9dfd24ecc40279 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0308ce8084e8480e9cc75195e9944f5d : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0c96e1a0484f98bc83e99c07e7a64c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9992ab50ec4e1b9a1347323babc1e8 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8c335a918604694b247114a3e4ac490 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce3eaaa4b05d493589b17eb0ad2d432c : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3205997f286499293bf5d288c1a040c : SUCCESS in 35m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/49fa1a7f03b54ab3bdd8c84479494d02 : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab90941a024a4f6a8868b55dc7e2f782 : SUCCESS in 49m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/409fdfff2ea94fa19600358e54ef22e9 : SUCCESS in 1h 14m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/771dc22bead6497b8637b87f7d298efc : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5152609c14d449c896dddb35e8477958 : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2492c4e122e04525bc36c08a9c4395f0 : SUCCESS in 1h 53m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deff5b4093e44fc78310b870053af3c1 : SUCCESS in 1h 17m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c3082cfc0354375a08281ceb4162c03 : SUCCESS in 51m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59f0b0ba78a74e7f9321f1236448226b : SUCCESS in 39m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e64fd1b619ac46d786bce735917e0372 : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3965ee61ac41c587da096e3ec9a3ec : SUCCESS in 26m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3172e475d5354335918f9ae1a843c616 : SUCCESS in 16m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5800386072274e34b81473e9fadb9968 : SUCCESS in 12m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0ffe8e6fbf2c47e183b8094dd9fbcb12 : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e8f0924cd51462992727c86da31ce1b : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1705010e6029845abf521888c166264a80404f15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:28:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"79c5f3855ca7ac307202c194659f58cf81c6d78a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:06:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee2f877f50a4141a11a3a8d1c618a7b : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e51f67f0b7a6467594eae7a1b4803556 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde2f8adcbbe453891a26f9985107b39 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a184dad4c8146e8bb17cdce978dfe5b : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0df620255d9541879e6f8d144a56f0d2 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d31d62e5f16846eeb3fa7948b9490074 : SUCCESS in 14m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87dcb496d46540eca7294e5bbc6568d6 : SUCCESS in 26m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/17ca64d3d7bd4c14985690274f6f4507 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69c56271f3a24f62871e283c4488798a : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28fafa414ea8483a88c2d7c075d4e0d0 : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5cda69b154f84eb19328a9e6d36ef083 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9ed44841d2a463a99b9d076922e78e0 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf252cf438ce43859a1afbb7603f0d4c : SUCCESS in 1h 37m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f3fa8c9ca84500a0b221af01ad1d89 : SUCCESS in 1h 16m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0ce6aa9e0ddf4f559aabe4b26d63bda7 : SUCCESS in 49m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb064fb9b10e478f8be3bc8c0bad3850 : SUCCESS in 46m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61a9a6b847bf4dbd8ecddb9f3d398ddc : SUCCESS in 25m 55s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbe7675e68d94686982f53085ec1bcf1 : SUCCESS in 22m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/79bccab6abc74b3eb146ec1fa9ee9232 : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7684bb1cfe74570aa3c28ef0c0ce974 : SUCCESS in 11m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/acc20a5730e54f04806159f962a0016d : SUCCESS in 16m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5314782bac1e4d73bdd1354e5c0a9e78 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":4},{"id":"0eeea3a2b07550f91184f72f5e452dae81056bbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:06:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b388c665ec2a45069170797a901508e89c28ebd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:09:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6635f5fdafc41a4aec24e33a16e555a : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":4},{"id":"407b7ce314bbb09515f3ab81a8c3106cd7081057","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 11:52:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b60d156ed7462d51618c612a77682605c6a466a2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-21 11:57:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0ce066ca3652688cb3486f29390f9451ed9437e7","revisions":{"9f34cece88db339bf0c0041335b763b8fdaee80d":{"kind":"REWORK","_number":1,"created":"2022-01-20 20:20:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/825671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/825671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/825671/1"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:18:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:20:02.000000000","tz":60},"subject":"Bump up manage-lvm role version to v0.2.5","message":"Bump up manage-lvm role version to v0.2.5\n\nThis fixes a bug that caused a role to fail on NVMe only hosts.\n\nChange-Id: I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f34cece88db339bf0c0041335b763b8fdaee80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f34cece88db339bf0c0041335b763b8fdaee80d"}]},"branch":"refs/heads/master"},"52dcd2e76e069ecec7fffd1ee41da672cc7510d4":{"kind":"REWORK","_number":2,"created":"2022-01-21 06:18:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/825671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/825671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/825671/2"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:18:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-21 06:18:00.000000000","tz":0},"subject":"Bump up manage-lvm role version to v0.2.5","message":"Bump up manage-lvm role version to v0.2.5\n\nThis fixes a bug that caused a role to fail on NVMe only hosts.\n\nChange-Id: I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52dcd2e76e069ecec7fffd1ee41da672cc7510d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52dcd2e76e069ecec7fffd1ee41da672cc7510d4"}]},"branch":"refs/heads/master"},"8a219dbdc1a3707a17a8027ba49b54bc71b6774a":{"kind":"REWORK","_number":3,"created":"2022-02-08 13:34:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/825671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/825671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/825671/3"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:18:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-08 13:34:40.000000000","tz":0},"subject":"Bump up manage-lvm role version to v0.2.6","message":"Bump up manage-lvm role version to v0.2.6\n\nThis fixes a bug that caused a role to fail on NVMe only hosts.\n\nChange-Id: I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a219dbdc1a3707a17a8027ba49b54bc71b6774a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a219dbdc1a3707a17a8027ba49b54bc71b6774a"}]},"branch":"refs/heads/master"},"0ce066ca3652688cb3486f29390f9451ed9437e7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-09 06:32:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/825671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/825671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/825671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/825671/4"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:18:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-09 06:32:46.000000000","tz":0},"subject":"Bump up manage-lvm role version to v0.2.6","message":"Bump up manage-lvm role version to v0.2.6\n\nThis fixes a bug that caused the role to fail on NVMe only hosts.\n\nChange-Id: I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ce066ca3652688cb3486f29390f9451ed9437e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ce066ca3652688cb3486f29390f9451ed9437e7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
