)]}'
{"id":"openstack%2Fkayobe~825831","triplet_id":"openstack%2Fkayobe~master~Idbf181c795ee98ad653f11ae483f9dab4ef1b599","project":"openstack/kayobe","branch":"master","topic":"network-mapping","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-30 11:20:16.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-08 15:50:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-30 11:20:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbf181c795ee98ad653f11ae483f9dab4ef1b599","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","status":"MERGED","created":"2022-01-21 16:02:45.000000000","updated":"2023-11-30 11:21:32.000000000","submitted":"2023-11-30 11:20:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825831-network-mapping","meta_rev_id":"43e4527be2fe23b541c54f6c83d43e92ca9a6994","_number":825831,"virtual_id_number":825831,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 11:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-08 14:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-08 15:50:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-30 08:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-30 08:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-08 14:30:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-21 18:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 07:16:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-08-27 00:07:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-08 14:30:06.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-30 08:40:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"5a4c8b1a870e78febe0beff4b7a7d2f5980714fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-21 16:02:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4f6a8ce23367a72faa4972c718675de2c2c02e5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-21 16:07:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"16b712313f21ffa3f1a2b54be0e747cc11353e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 18:01:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eacce498f35b4239ac2cd3100f51141f : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0d550c60ab4aacb78c07e45b6e795b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c93834809eaa4268bf0a5dffaa4614aa : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea55384f9c744538912a675e9020f482 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dd7a2635dc42388fb81c7da5d1916f : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60ad733649364149a1e6c89fbe498f77 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a4054b970354da0a89373ed5fcdcafd : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61f5edfbce6d44c2b2337fc6fc0dca0c : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e55c8ce235b44dad86f13790477cfb61 : SUCCESS in 30m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d02b308bd42d4683ac3bfeb427539238 : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cacdf13f79b24447b096c7909fc0809c : SUCCESS in 48m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6e8766a4fbbf49b087a8746b2c95a4f2 : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/608ec9117e7a4f1a8ff56df678d2e596 : SUCCESS in 23m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a43437cdc9c44c286614c5ccacbef25 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12f6b45d26e3414b86342e123c71b920 : SUCCESS in 1h 55m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5f1dd059e0f49de9d3d8f8db27b50e7 : SUCCESS in 1h 01m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5b708b5a69e14d169b3b38e729ccca79 : SUCCESS in 51m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27bf6db2b9db41ba87a78fd6552c9f41 : SUCCESS in 43m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/21c3c9536822449a8eff6cb5b2a694be : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43b7b00599644f48881f32143693c4e9 : SUCCESS in 25m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/31a97eb285914674b36a8b64aff4ee5d : SUCCESS in 14m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdc4497a90cb4d89b89b96b698859744 : SUCCESS in 12m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/aed4f9be2a094357885d58bed230d3cf : SUCCESS in 18m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a43dfcc0a6644c33bf684596c7705c99 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ee70ac3ef0ca39be2c4834c23cbbaf64239507fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 14:18:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"feccf159a724210437ff465378da49c579d1a1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:14:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d18045eebcd4ab5a64ccd3bbefbabac : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35c601afcc4414a8bf583e5269771e7 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6935f0221f584151bb9c4c5484c1253e : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0364c5d3937420b93a7596c76e88c54 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc97db3bdf4848e080e11439fa1f09ab : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0924947f4e7f455fa660b7e98a31bac7 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a04ec783dea4a2ba66c5445dec3b2b7 : SUCCESS in 8m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/844bb774799543518a706d6d8a8720ec : SUCCESS in 14m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c884538e3a5f4bc0a6f2245d72a31d18 : SUCCESS in 36m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4cd7617f5ad243bc9c8fbd556db19564 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b7e470d3fdca4ada8ac04cdd2205648d : SUCCESS in 1h 11m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ee1d4b385f4699a6213a58214dcd6e : FAILURE in 11m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/97d28c6d6f3744228962c179f05ca39f : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cfa131b58ed542b3971ece24edf2cab1 : SUCCESS in 25m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4dffe7a304854c22b1e982e7a0309471 : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f437fe7b7846fea09a10a0fa8c144f : FAILURE in 10m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8452ac788ae24218bbf118081185339f : SUCCESS in 1h 51m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce14314a7f4d4f80951aaee64db2f89e : FAILURE in 11m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3aa751d27df0449891bdc8a53e81d9c4 : SUCCESS in 52m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9a37802c8367448a905815ca201ae83d : SUCCESS in 36m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/001a664c8d51479e99c4adeffd35b080 : FAILURE in 11m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/123f147a80e44bc7b28eea4fbc5aa184 : SUCCESS in 37m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69e22c8361b04cd79a34b0138e72cbb3 : FAILURE in 9m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab13aa51642945c4ac15d9d7226780fe : SUCCESS in 18m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/68bca4a1373a4c7d9e91210efc66c966 : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad19b44151d468c81862b9d87ea3b71 : FAILURE in 10m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2b37501b11324a7b84be7f72c026be8d : SUCCESS in 20m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ae2461c86738444a8d821acf5b246493 : SUCCESS in 18m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a689716274f64cbd9cc426716f24b606 : FAILURE in 10m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b9bf0b915a4af00ff4f5e15174ab2e03dda53ce6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 16:32:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"50b2ae9db83ff2d64a0b6c287a2693e1e43a6c4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 09:24:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d8dd24765e8372490e2f7227a1beabf55288fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 11:16:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d38bb0a6cf643dab0570d2a1aead9f9 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4da3966c36d40faa5ee1a68590224a2 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b231f34a70174b5ea54a12676cd83d4d : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/270ec326c0174ac9812e2865d02fc6ae : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ee7d994eeb49d3b778a00aaf375b1d : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c67f7fde3c419797205b36211fc514 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c63899e87d1044248cb540307918b0a2 : SUCCESS in 7m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c07b536454034ea9a872ff68bd356261 : SUCCESS in 14m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7692784ac605493b94ce090aa1fd7595 : SUCCESS in 34m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/394f07da3697485da3f91f46602c568d : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/35bcbd73171741a1b14c346f7397e47b : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fce2318214384422932db8a71f6bfd78 : SUCCESS in 55m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5441900ba004419ca6b6b9783b39e315 : SUCCESS in 1h 19m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3512e562f63463bb2d5dba52860bfe9 : SUCCESS in 25m 40s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9e84c94f093944a799134f8538e83b9f : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a94206618e0744b2ae9e55a2aa8bc533 : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54d1773770444416aa664aae15cfa033 : SUCCESS in 1h 47m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7766af0b7454f9699e5e30582dbf1de : SUCCESS in 1h 14m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/59c0cc63a12b4dc08a243a6818a53412 : SUCCESS in 25m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8f662f792d7f451db91971378c3d5b7e : SUCCESS in 26m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9ee4f52146448279b7d0c26bc04f3ee : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4aeb7e8ff0cf42d08ccb6878003e700c : SUCCESS in 41m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cda34d2313ff456fb386ff6037762ede : SUCCESS in 32m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1b2095db64334b9db7bd0225efbe9713 : SUCCESS in 18m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/adddb35a48a54e03b6d72216c78ffc0e : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/364f84cd3cfc47c38e5f60ca75fa9d14 : SUCCESS in 13m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f5a679ac8a844d49933e75fa37795580 : SUCCESS in 20m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/880574246b4b414da26035f33ba9e247 : SUCCESS in 18m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7220c1be5a2641ef8358a1f314392519 : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7c89632ad6f09eda07563b0f9247a083b787c53c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:58:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"648068058513cb6b68e6d2d1c1cfee642d8f6e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:08:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edb7362dd60f45f785652f029e891e4d : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c198b36f2114c9f8937328e9d461e36 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2f1f1045dc244a1bb6266c6cc0b39c3 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ee7930a65b74ee9b64dafc0d4111f4f : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90282df1a994020900fab046a2af1ba : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25d5e01d8cf04eceaa45a8fd89b5b542 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e94584f6763b4ccd9e08c734c693f6eb : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e1d7cd2f6814690958f0c26156a87a9 : SUCCESS in 13m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b45ae1dd37e4356a5a52db785d17861 : SUCCESS in 17m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/09dafc2dddf449a49612d7ff4cdac106 : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/eec72929dfec4469b8d734b07be19aed : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25b7354acead4bb786326880b707f273 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fdd8473bfab5456b971d3d2b5a85b081 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3c7ee51e5bbf4d7892c09022ecc2ccb5 : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/af4294b0d54f4fa1a486293d49c0872f : SUCCESS in 23m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7bd86161e8343348068528f9266625b : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1e2d3ed0fde24d3daf7e3c87da1ae8a2 : SUCCESS in 1h 49m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/320245fa09ce4aa892f6581461f6d070 : SUCCESS in 1h 24m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b6de371494764e92800363d9dee76aa2 : SUCCESS in 22m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/83e25984eb224828926ab575abd9a355 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cea13b12b4b45559891ed270cf3a314 : SUCCESS in 16m 34s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/ae5c06309b724e42b06aa6f213d8b1c0 : SUCCESS in 53m 35s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ddf856bb0c7048c88979fd7a91c4ec04 : SUCCESS in 36m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5554c7d0e514793995eb2cfc7291bb3 : SUCCESS in 30m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff855d604d904a9c96558f957b703893 : SUCCESS in 18m 27s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/363ba2ae194e44bca7853e6831cd1d93 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7036b27fd97a462e907b67fac64d6f3d : SUCCESS in 22m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/aa34f13dce194a068d335b803beb6598 : SUCCESS in 18m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/45efc7049233432b85c7edcdcf881cf2 : SUCCESS in 19m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef9b20512bfe4509bdfa6590fe5a379d : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":4},{"id":"4c103759938c7e7b4185d8666dc60f22b8637260","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-29 07:16:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5e7803e71b279479da50cce42d689d02e1dfef14","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-27 00:07:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"005009fcac2a6888a8ef49c8516a257e2d95be2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 16:16:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9f00f4cb48698461751a93803e81f069f4444769","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 16:18:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b6907afda442c0cb1db9d258700b2e25d76d043c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 18:43:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d8104ea6fda44f189f4fba9870f0baf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae31ca5918f441a093b9251a28bb193a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d7c4a28011d4107a035d908c80d078e : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a8a1864183e4b338213ac007a04e524 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/170b62984f294cecba740939264e4f16 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18eb38271c46440e80e1d595fc68c04f : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1536765a0b7a470eb4421d769d1cb5f7 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/251b17e0914f4176bf699af3d7ba51af : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11772887058b4228a1ac8ec62f2a2ba2 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cc6aee4a4ef49bb863b67d3b13b1e2a : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/128865dedbee4032abbe611b6d4cf7ce : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/756b2f16731245e69f1dd385f7670dc5 : SUCCESS in 1h 04m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eb8d2e6eac24338ad91bfad6f166ef4 : SUCCESS in 52m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e8f0fe98bbed49a48e95f5b46f8bb7e6 : SUCCESS in 1h 13m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f2b3340181974c5f950ac784a2e2eca0 : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1b9582a4eb564333aaec94b92a90421e : SUCCESS in 17m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff1ce379ffd742cab4e751f7c7c84c59 : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/90355d930f1c4260b34cd751170048ec : FAILURE in 1h 58m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e73afd42207488d874297928f1b7fc5 : SUCCESS in 2h 18m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f0cb22535b3344dfa3a640f63df55010 : SUCCESS in 16m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4470290a219b452192afafb3a4b3da3a : SUCCESS in 13m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/834f4e2b753c4586bf98c0785e7df5c7 : SUCCESS in 33m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a0799f8208d46deacddd0e53331f4be : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21ea02b7a9054520a42f252484781742 : SUCCESS in 26m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/beaca5b623654166955b12d306f3e548 : SUCCESS in 11m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bca057834d9498b981de933359f9d1c : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/162e7baf112f467f8de3ac4edf61a843 : SUCCESS in 15m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132573274d914ee8a1ea8e8cd09d38e3 : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":5},{"id":"7cf203f0a6104312fd72f94e771f7c0e4659fa5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-02 10:05:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43dbf9ae8388e00fe8c53b92f9caf2c452895d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 12:50:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c14eb3899d433695b7d48c40201e6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8d2958bfcb44579a75791738714a563 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce4afc7cb6e46a887ae6ae41dc36151 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61107ef53df4725bd2012aca5666612 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59db596a29be4509acf8bc677e01a332 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044b46c6d7c94398bbb9537541005b15 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6feb73e37384e39a70c85ad069e8f40 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a7da444ebe749628d9f2233167dfb73 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/837c2b31c14a4a629df7b32091ded1cb : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fabc2c00cb8446daee4c373a1fc2d7c : SUCCESS in 9m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1589838f158146cd8fe58289ce9cb5e2 : SUCCESS in 54m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e1c6310f1cf4a2ab8450033eb2a46be : SUCCESS in 55m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac7faaeeb2a4a50b851de0ca2eb25a7 : SUCCESS in 45m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fac2bb361d724c89818d0b75bdad2f08 : SUCCESS in 58m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e6acd65f4734447d99a596c15a114b02 : SUCCESS in 15m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d04ff76bbe92413ba245bcd3eaff4f09 : SUCCESS in 17m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/919375b5280145f385ed9fae7fbc80eb : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ddc85087bc6849528e716b681048c0e2 : SUCCESS in 2h 20m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c14f627a20b949b3a44932951e297141 : SUCCESS in 2h 36m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/833d9d71ab704981ba3fbc6481ee2b88 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b900d8073d1c48eaa2cbcd4b9d501e88 : SUCCESS in 14m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/877427027c724fbaa823feafca1464d9 : SUCCESS in 34m 01s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b60782bc344b4981b8de77e8ff83ecb1 : SUCCESS in 30m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a37c71b84ea417ca2529c4035dd5ab4 : SUCCESS in 23m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed7ade3430484afab283a40461deefa6 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da00a1bc8ce54b29bd41ad0e0542f1d5 : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c65bebbcec5456f8902c18bfc856e4f : SUCCESS in 16m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bd9c99758df4faf98161d6331a9302b : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c26c54089906c863ddca2b865073ed7ff869ba6a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 14:30:06.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"b69fb1590ef2895ee4554a64e404be28c95ed009","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-08 15:50:19.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8e5cf9f08d15b26c089d503727516c37b75201b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 08:40:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f7fe17d19e5df48176e4a8a7e2ad8120aeee7e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 08:40:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6e619dbb64dccc7747fee5bbab753e2f017f956a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:20:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fc42507a224e42a0e8c90eaf88c980\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f8198a077d4a0bae44cf3bcf469790 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f429d98871040f9aae56a0cf99a936a : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1cb2963e9c7a4a66bbddb1fceda69b6a : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/281d5cb0374141e6a1cb4d390f33d8fd : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26f14d00920a456fa45f2375d16555b8 : SUCCESS in 10m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2d20250e379e44f68fe379d4117480c5 : SUCCESS in 50m 44s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e239df33c51c47bdb6aceb9eff06874d : SUCCESS in 42m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e41c1768aa064486b29dcda2209d8fd7 : SUCCESS in 1h 17m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/33b5d5b391f448d7ac96d13a22cb5197 : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/985d1b11ec6e47fa8566528386189c18 : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1240d3d1f9644904adcf13dc463c222b : SUCCESS in 2h 34m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b1750c942594fd7bc7e7a3345017d61 : SUCCESS in 2h 20m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0156f1ab54ee40ca9463a7e054c5678b : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef87ab27174f4de2826bebabacc67bf3 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/63ab8c5296324884bf1f96c25db1f05f : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfe1d9c24a474b56ac85c7bd718910e0 : SUCCESS in 26m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/668b7d8935c345868fc1eceec2410ce3 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/448518572205434a963cbfb9129cccd6 : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da57cbbf669249fca18bd0ec268f3f39 : SUCCESS in 16m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8bbf13fbd77402dbf626a9bd04d0ce9 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":5},{"id":"175336814b28862392d6bf4e1d84fe02f72f7af3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:20:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"43e4527be2fe23b541c54f6c83d43e92ca9a6994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:21:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746d3e11164c48139ff1fb8d648600da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb77bdd94ff46c087191174361696bc : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b352c345a6c4ab2b7aa3249494216c6 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2071dc9a3889a5a1e6cba546acdfc20170bf7610","revisions":{"1b49f86e86e971e326fb4dd6f493f5a9be1cab83":{"kind":"REWORK","_number":1,"created":"2022-01-21 16:02:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/825831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/825831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/825831/1"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 13:48:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 16:01:30.000000000","tz":0},"subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","message":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network\n\nKayobe has fairly coarse-grained default groups - controller, compute,\netc, which work well in the majority of cases. Kolla Ansible allows much\nmore fine-grained placement on a per-service basis, e.g.\nironic-conductor. If the operator has taken advantage of this\nfine-grained placement, then it is possible that some of the assumptions\nin Kayobe may be incorrect. This is one downside of the split between\nKayobe and Kolla Ansible.\n\nFor example, Ironic services may have been moved to a subset of the top\nlevel \u0027controllers\u0027 group. In this case, we would not want the Ironic\nnetworks to be mapped to all hosts in the controllers group - only those\nrunning Ironic conductor services. The same argument can be made if the\nloadbalancer services (HAProxy \u0026 keepalived) or Neutron dataplane\nservices (e.g. L3 \u0026 DHCP agents) have been separated from the top level\n\u0027network\u0027 group.\n\nThis change abstracts the placement of Ironic, loadbalancer and network\nservices into separate variables, rather than referencing the top level\n\u0027controllers\u0027 and \u0027network\u0027 groups directly. These variables may be\nupdated by the operator to match the service placement.\n\nChange-Id: Idbf181c795ee98ad653f11ae483f9dab4ef1b599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b49f86e86e971e326fb4dd6f493f5a9be1cab83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b49f86e86e971e326fb4dd6f493f5a9be1cab83"}]},"branch":"refs/heads/master"},"b3fed016f3fecc58cfb8d6643369bdaf114a703d":{"kind":"REWORK","_number":2,"created":"2022-03-29 14:18:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/825831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/825831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/825831/2"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 13:48:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 14:17:24.000000000","tz":60},"subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","message":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network\n\nKayobe has fairly coarse-grained default groups - controller, compute,\netc, which work well in the majority of cases. Kolla Ansible allows much\nmore fine-grained placement on a per-service basis, e.g.\nironic-conductor. If the operator has taken advantage of this\nfine-grained placement, then it is possible that some of the assumptions\nin Kayobe may be incorrect. This is one downside of the split between\nKayobe and Kolla Ansible.\n\nFor example, Ironic conductor services may have been moved to a subset\nof the top level \u0027controllers\u0027 group. In this case, we would not want\nthe Ironic networks to be mapped to all hosts in the controllers group -\nonly those running Ironic conductor services. The same argument can be\nmade if the loadbalancer services (HAProxy \u0026 keepalived) or Neutron\ndataplane services (e.g. L3 \u0026 DHCP agents) have been separated from the\ntop level \u0027network\u0027 group.\n\nThis change abstracts the placement of Ironic conductor Ironic\ninspector, loadbalancer and network services into separate variables,\nrather than referencing the top level \u0027controllers\u0027 and \u0027network\u0027 groups\ndirectly. These variables may be updated by the operator to match the\nservice placement.\n\nChange-Id: Idbf181c795ee98ad653f11ae483f9dab4ef1b599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3fed016f3fecc58cfb8d6643369bdaf114a703d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3fed016f3fecc58cfb8d6643369bdaf114a703d"}]},"branch":"refs/heads/master"},"cc9181f5c52b2716c290ff75e593f6913a646b6b":{"kind":"REWORK","_number":3,"created":"2022-03-30 09:24:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/825831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/825831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/825831/3"}}},"commit":{"parents":[{"commit":"0cd0f05781cdc1915792fe1aecbde18b0aee4063","subject":"Merge \"libvirt: support SASL authentication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cd0f05781cdc1915792fe1aecbde18b0aee4063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 13:48:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 09:24:09.000000000","tz":60},"subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","message":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network\n\nKayobe has fairly coarse-grained default groups - controller, compute,\netc, which work well in the majority of cases. Kolla Ansible allows much\nmore fine-grained placement on a per-service basis, e.g.\nironic-conductor. If the operator has taken advantage of this\nfine-grained placement, then it is possible that some of the assumptions\nin Kayobe may be incorrect. This is one downside of the split between\nKayobe and Kolla Ansible.\n\nFor example, Ironic conductor services may have been moved to a subset\nof the top level \u0027controllers\u0027 group. In this case, we would not want\nthe Ironic networks to be mapped to all hosts in the controllers group -\nonly those running Ironic conductor services. The same argument can be\nmade if the loadbalancer services (HAProxy \u0026 keepalived) or Neutron\ndataplane services (e.g. L3 \u0026 DHCP agents) have been separated from the\ntop level \u0027network\u0027 group.\n\nThis change abstracts the placement of Ironic conductor Ironic\ninspector, loadbalancer and network services into separate variables,\nrather than referencing the top level \u0027controllers\u0027 and \u0027network\u0027 groups\ndirectly. These variables may be updated by the operator to match the\nservice placement.\n\nChange-Id: Idbf181c795ee98ad653f11ae483f9dab4ef1b599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9181f5c52b2716c290ff75e593f6913a646b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9181f5c52b2716c290ff75e593f6913a646b6b"}]},"branch":"refs/heads/master"},"715366b527932b09190f9b61d90759ea4d033ced":{"kind":"REWORK","_number":4,"created":"2022-04-12 10:58:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/825831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/825831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/825831/4"}}},"commit":{"parents":[{"commit":"5ede87656c05a95e34369417a660fd1b6d06bbfc","subject":"proxy: add ansible issue 8743 workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ede87656c05a95e34369417a660fd1b6d06bbfc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 13:48:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:58:16.000000000","tz":60},"subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","message":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network\n\nKayobe has fairly coarse-grained default groups - controller, compute,\netc, which work well in the majority of cases. Kolla Ansible allows much\nmore fine-grained placement on a per-service basis, e.g.\nironic-conductor. If the operator has taken advantage of this\nfine-grained placement, then it is possible that some of the assumptions\nin Kayobe may be incorrect. This is one downside of the split between\nKayobe and Kolla Ansible.\n\nFor example, Ironic conductor services may have been moved to a subset\nof the top level \u0027controllers\u0027 group. In this case, we would not want\nthe Ironic networks to be mapped to all hosts in the controllers group -\nonly those running Ironic conductor services. The same argument can be\nmade if the loadbalancer services (HAProxy \u0026 keepalived) or Neutron\ndataplane services (e.g. L3 \u0026 DHCP agents) have been separated from the\ntop level \u0027network\u0027 group.\n\nThis change abstracts the placement of Ironic conductor Ironic\ninspector, loadbalancer and network services into separate variables,\nrather than referencing the top level \u0027controllers\u0027 and \u0027network\u0027 groups\ndirectly. These variables may be updated by the operator to match the\nservice placement.\n\nChange-Id: Idbf181c795ee98ad653f11ae483f9dab4ef1b599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/715366b527932b09190f9b61d90759ea4d033ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/715366b527932b09190f9b61d90759ea4d033ced"}]},"branch":"refs/heads/master"},"2071dc9a3889a5a1e6cba546acdfc20170bf7610":{"kind":"REWORK","_number":5,"created":"2023-11-01 16:16:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/825831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/825831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/825831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/825831/5"}}},"commit":{"parents":[{"commit":"bd9043dca0af94bd6ed83268ace613de91a3ee0e","subject":"Merge \"Remove unsupported kolla_docker_registry_insecure variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd9043dca0af94bd6ed83268ace613de91a3ee0e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 13:48:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-01 16:16:45.000000000","tz":0},"subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","message":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network\n\nKayobe has fairly coarse-grained default groups - controller, compute,\netc, which work well in the majority of cases. Kolla Ansible allows much\nmore fine-grained placement on a per-service basis, e.g.\nironic-conductor. If the operator has taken advantage of this\nfine-grained placement, then it is possible that some of the assumptions\nin Kayobe may be incorrect. This is one downside of the split between\nKayobe and Kolla Ansible.\n\nFor example, Ironic conductor services may have been moved to a subset\nof the top level \u0027controllers\u0027 group. In this case, we would not want\nthe Ironic networks to be mapped to all hosts in the controllers group -\nonly those running Ironic conductor services. The same argument can be\nmade if the loadbalancer services (HAProxy \u0026 keepalived) or Neutron\ndataplane services (e.g. L3 \u0026 DHCP agents) have been separated from the\ntop level \u0027network\u0027 group.\n\nThis change abstracts the placement of Ironic conductor Ironic\ninspector, loadbalancer and network services into separate variables,\nrather than referencing the top level \u0027controllers\u0027 and \u0027network\u0027 groups\ndirectly. These variables may be updated by the operator to match the\nservice placement.\n\nChange-Id: Idbf181c795ee98ad653f11ae483f9dab4ef1b599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2071dc9a3889a5a1e6cba546acdfc20170bf7610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2071dc9a3889a5a1e6cba546acdfc20170bf7610"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
